)]}'
{"id":"openstack%2Fpuppet-neutron~756527","triplet_id":"openstack%2Fpuppet-neutron~master~I2a321fb780e789b7556d557e2170a3267049c432","project":"openstack/puppet-neutron","branch":"master","topic":"min-bw","hashtags":[],"change_id":"I2a321fb780e789b7556d557e2170a3267049c432","subject":"Add new classes ovs_driver and sriov_driver","status":"MERGED","created":"2020-10-07 14:44:57.000000000","updated":"2020-10-13 09:54:07.000000000","submitted":"2020-10-12 19:40:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756527-1602531624580-eefd7343","meta_rev_id":"3b7101e5db6e201344d72ab08188105043b74ab1","_number":756527,"virtual_id_number":756527,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-10-12 07:31:04.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-12 19:40:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-10-10 04:07:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-10-12 13:52:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-10-12 13:52:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-07 16:45:04.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2020-10-09 14:08:11.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-10-10 04:07:10.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-10-12 13:52:39.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-10-12 19:40:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"10fef2f6a98bd395fbdf0caa7fd45bb1142f58f5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-07 14:44:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0ab4aae00d78f92868262acd8b8e297640f9715","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2020-10-07 16:42:14.000000000","message":"Patch Set 1:\n\nIf this is related to a lp bug, you might want to reference it here for backporting.","accounts_in_message":[],"_revision_number":1},{"id":"9c0ead0b16a1a6410bd21dfe148cc7c8b8cc9771","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2020-10-07 16:45:04.000000000","message":"Patch Set 1:\n\nMight also want to add some unit tests to match these files.","accounts_in_message":[],"_revision_number":1},{"id":"e2ce1c30bac0df54c74b2e5f235742b9878b19fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 17:14:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a44f2f2953a4146a46b5202d1f06afb : SUCCESS in 3m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1d13d4ad0044a5bb4a864a1816daf2d : SUCCESS in 3m 15s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/dc151999c58a4714b7cb30d5aecc7fee : SUCCESS in 18m 20s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11407a2c8e9c42c0a73f423159bf87ad : SUCCESS in 16m 04s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f64982752ad34f62a9d0e918c3462249 : FAILURE in 18m 08s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6573747782ba458a8cd205dba5c6b447 : SUCCESS in 40m 43s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80bd7b51bf614c49b9a4e1ff3016370d : SUCCESS in 49m 14s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afb58faa2b6f4a7abbb170e2ae717478 : SUCCESS in 52m 35s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2e05c1c082a341108345680ee43ccc49 : SUCCESS in 36m 57s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/850cc92dca044242be02ef39375e9e21 : TIMED_OUT in 2h 00m 52s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9020bf9db4bb463abeca00b06903340e : SUCCESS in 1h 39m 11s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/543be7a97cfa4dfe997449f474da0c7e : SUCCESS in 1h 31m 20s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f1e727d004e9415593c4aee15e8cc7a5 : SUCCESS in 56m 04s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/aeb61f6bd1ec46d7a6c45e40772231d6 : SUCCESS in 22m 27s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5870aac4c74468aa71ebffa90383d40 : SUCCESS in 18m 08s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7190483c3e1f46ec80ccece1c39e146e : SUCCESS in 2h 15m 21s","accounts_in_message":[],"_revision_number":1},{"id":"23d1c1630a646f59dc807df3edf2e1e055285e64","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-09 10:53:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2eec50df1a032cf0c6fdec6211fb1ea2741dd3c2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-09 11:05:09.000000000","message":"Patch Set 2:\n\n\u003e Might also want to add some unit tests to match these files.\n\nDone","accounts_in_message":[],"_revision_number":2},{"id":"27ec6e1e238ab6fa5874335d9203bf2acd2c26a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 13:13:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45bf406af53845c582a8224502d76a68 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83bc8e846d9644438ef7ca7f3aff4df8 : SUCCESS in 4m 44s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/419b78ecddcb4a6e9c95d8634e039ea1 : SUCCESS in 20m 41s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a8915cfd67b49468a22ae05fbe35ea0 : SUCCESS in 15m 36s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c4d1a1a779647e382dcc561c76d37ea : FAILURE in 21m 12s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/452f0c9c361e467aa5c9bab2403cfe69 : FAILURE in 25m 58s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a61a0a90e9514122914b773c2d5b031d : FAILURE in 26m 48s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afc1e5ff1a8b4cf3b5668d647dfd36d9 : FAILURE in 23m 05s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a3701c765ee5489185a6d2b49d9abd83 : FAILURE in 17m 59s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9a557633046f40efb08a5e04e01260d9 : SUCCESS in 1h 15m 40s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/08f2df1ba2b1461f94a0c0a3b971d601 : SUCCESS in 1h 24m 46s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3e7b5c52ed974d739e66400357918835 : SUCCESS in 1h 29m 00s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a0cf0d70fe574cb48b29460d6c5f182d : SUCCESS in 59m 04s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/22f0f32d479a46ddbbf55d1f1eba1708 : SUCCESS in 23m 24s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8cef835c839d4eb380cdb42580c09cc1 : SUCCESS in 17m 56s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8bdc90419dd648c9a368c6aa442d87b3 : SUCCESS in 2h 09m 13s","accounts_in_message":[],"_revision_number":2},{"id":"4ae27f49f9ab0f0cc7daf08e4be7d9612674c23c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-10-09 14:08:11.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)\n\nsome minor comments inline if you want to update but not blocker for me.","accounts_in_message":[],"_revision_number":2},{"id":"61734f55c5cd0a70d359cc91cd68e1690b1f2250","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-10 04:03:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"440f86f6cb11db5acd04b7616655af6f222dd432","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-10 04:05:52.000000000","message":"Patch Set 3: Code-Review+2\n\nthe patch is updated to include a release note about the new classes and fix several minor stuffs like ones Emilien pointed.\n\nAlso, please find the subsequent commit to deprecate parameters in the mech_driver class.","accounts_in_message":[],"_revision_number":3},{"id":"e4aec73e15474bf681a245eaba1e58bb5fa7974f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-10 04:06:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d823af5de2a298371b93e3265ab6706c51465d1c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-10 04:07:10.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0ae8d180c6383999dc7eaf458fa5955795cd3c5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 06:19:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61387af7393f4009a79942ec4660f7a1 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc2cb53e691d475bad674675d74dc516 : SUCCESS in 3m 50s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/6f80491c92ff4e608e2849c5aa34ecf1 : SUCCESS in 18m 02s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5369e81a89d94e2bbe167fdd170274c0 : SUCCESS in 13m 53s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0578753bcbe545839b3a192ca5e2f76a : FAILURE in 19m 59s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/11592f4a2cf447c1990f9e75d76b49d9 : FAILURE in 26m 32s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba67194d620f4aadbc9ec33308fefc6b : FAILURE in 27m 07s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4514f981d754e02af9dd0a57c9f86bf : FAILURE in 23m 49s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2386e14babce4332b3f66b0b942aca4f : FAILURE in 18m 53s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1d4f9885f0014fbdbdeb5d44d6985852 : SUCCESS in 1h 22m 30s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/736fcd1fdeac4e4d8a5cc7306e2a7d30 : SUCCESS in 1h 29m 37s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a5e931933f5a4d579f92dce15aa80cc8 : SUCCESS in 1h 22m 56s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b18c170a3b1449bb8f466978357c4ad9 : SUCCESS in 59m 58s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/c2146d26ba7d442788f881775821b71d : SUCCESS in 21m 19s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b97f1d655570404ea34349ca136d1f9a : SUCCESS in 16m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1314507299e4479a23c0567bd7747e2 : SUCCESS in 4m 33s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ef33f0e161974c90b1be398ff693853f : SUCCESS in 2h 09m 41s","accounts_in_message":[],"_revision_number":4},{"id":"f8b4795a54ce82a32c22f2c7e492ec59326fb08c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-12 07:31:04.000000000","message":"Patch Set 4:\n\n\u003e the patch is updated to include a release note about the new\n \u003e classes and fix several minor stuffs like ones Emilien pointed.\n \u003e \n \u003e Also, please find the subsequent commit to deprecate parameters in\n \u003e the mech_driver class.\n\nThank You for addressing those comments and adding release note :)","accounts_in_message":[],"_revision_number":4},{"id":"8e482fc0b238d246e6c763aed59008787c887e34","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-10-12 13:52:39.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4c310df6c5c519830412db067990c220442339b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 13:52:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bbd02e8c5294058333f3f19c101b5599b8e449ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 19:40:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"523a992eb7da8799330854946c15a298321be64e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 19:40:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e0e12f0c4954a56ad6cb3528de3592a : SUCCESS in 3m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f3573420dc94ca7bea9016d3763f048 : SUCCESS in 3m 46s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/8e58d43314b84d26a8d3d960c6c72ea2 : SUCCESS in 19m 58s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff3351fe6723400991542a0241eb8a92 : SUCCESS in 17m 31s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4f4a57913d904adea969a01faa1ebf59 : SUCCESS in 1h 31m 02s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ebbdb52650414942ad9c9cde8dc49acc : SUCCESS in 1h 40m 23s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9c33b96713ce4d659203b7039061f66d : SUCCESS in 1h 45m 38s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0e9dc70790b74ff8810c097209a8643e : SUCCESS in 54m 23s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/67e02c4dc9f64f3484cc382bbcaf887b : SUCCESS in 25m 05s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6f82ebec5ac4aad85b0ee21b4ee30b3 : SUCCESS in 16m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6986d8032974e15bd7c416b315d67f5 : SUCCESS in 4m 04s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4f26c435da60472c94a8a2bcbced6970 : SUCCESS in 2h 13m 19s","accounts_in_message":[],"_revision_number":4},{"id":"6fc6d6d73a6d1f7e7aa82e0bb7ab7285b1270ac0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 19:48:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b241893dda1441009d949878162a00fb : SUCCESS in 58s","accounts_in_message":[],"_revision_number":4},{"id":"24813ff4b18efd4892e42169c18047e7b18bfe1c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-13 08:20:08.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 957b18a636d6887a31e3ca9237ae16f8414e34cc","accounts_in_message":[],"_revision_number":4},{"id":"472030df1d5269f82070df021e2381ed9b37727c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-13 09:54:07.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit d2ce5cd0052f2fcb27b3690208cc43221f077f5d","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0b8a90ec5c4cdac145672150ee4d58f7959cf6bb","revisions":{"3d07ae75ea21dc2a41177d8c0ef3eadfa116622e":{"kind":"REWORK","_number":1,"created":"2020-10-07 14:44:57.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/27/756527/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/27/756527/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/27/756527/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/27/756527/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/27/756527/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/27/756527/1"}}},"commit":{"parents":[{"commit":"59d60747b1b904c8442f7e0ef57a7e625508b42e","subject":"Prepare a new release for victoria cut off","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/59d60747b1b904c8442f7e0ef57a7e625508b42e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-10-07 14:41:40.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-10-07 14:44:51.000000000","tz":0},"subject":"Add new classes ovs_driver and sriov_driver","message":"Add new classes ovs_driver and sriov_driver\n\nThose classes handles ML2 plugin\u0027s configuration related to the\nOVS and SR-IOV mechanism drivers.\nPreviously there was neutron::plugins::ml2::mech_driver function which\nwas used by neutron::plugins::ml2 class but it isn\u0027t used at all\ncurrently and probably it will need to be deprecated and removed in\nfollow up patch(es).\n\nChange-Id: I2a321fb780e789b7556d557e2170a3267049c432\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3d07ae75ea21dc2a41177d8c0ef3eadfa116622e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3d07ae75ea21dc2a41177d8c0ef3eadfa116622e"}]},"branch":"refs/heads/master"},"cadaeb4a05cf5f4be6846d0a40bf7454bbe49402":{"kind":"REWORK","_number":2,"created":"2020-10-09 10:53:32.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/27/756527/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/27/756527/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/27/756527/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/27/756527/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/27/756527/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/27/756527/2"}}},"commit":{"parents":[{"commit":"59d60747b1b904c8442f7e0ef57a7e625508b42e","subject":"Prepare a new release for victoria cut off","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/59d60747b1b904c8442f7e0ef57a7e625508b42e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-10-07 14:41:40.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-10-09 10:53:19.000000000","tz":0},"subject":"Add new classes ovs_driver and sriov_driver","message":"Add new classes ovs_driver and sriov_driver\n\nThose classes handles ML2 plugin\u0027s configuration related to the\nOVS and SR-IOV mechanism drivers.\nPreviously there was neutron::plugins::ml2::mech_driver function which\nwas used by neutron::plugins::ml2 class but it isn\u0027t used at all\ncurrently and probably it will need to be deprecated and removed in\nfollow up patch(es).\n\nChange-Id: I2a321fb780e789b7556d557e2170a3267049c432\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/cadaeb4a05cf5f4be6846d0a40bf7454bbe49402"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/cadaeb4a05cf5f4be6846d0a40bf7454bbe49402"}]},"branch":"refs/heads/master"},"336c32a58c07917844fe40b93df8e96d3d535c3e":{"kind":"REWORK","_number":3,"created":"2020-10-10 04:03:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/756527/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/27/756527/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/27/756527/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/27/756527/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/27/756527/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/27/756527/3"}}},"commit":{"parents":[{"commit":"59d60747b1b904c8442f7e0ef57a7e625508b42e","subject":"Prepare a new release for victoria cut off","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/59d60747b1b904c8442f7e0ef57a7e625508b42e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-10-07 14:41:40.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-10 04:01:32.000000000","tz":540},"subject":"Add new classes ovs_driver and sriov_driver","message":"Add new classes ovs_driver and sriov_driver\n\nThose classes handles ML2 plugin\u0027s configuration related to the\nOVS and SR-IOV mechanism drivers.\nPreviously there was neutron::plugins::ml2::mech_driver function which\nwas used by neutron::plugins::ml2 class but it isn\u0027t used at all\ncurrently and probably it will need to be deprecated and removed in\nfollow up patch(es).\n\nChange-Id: I2a321fb780e789b7556d557e2170a3267049c432\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/336c32a58c07917844fe40b93df8e96d3d535c3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/336c32a58c07917844fe40b93df8e96d3d535c3e"}]},"branch":"refs/heads/master"},"0b8a90ec5c4cdac145672150ee4d58f7959cf6bb":{"kind":"REWORK","_number":4,"created":"2020-10-10 04:06:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/756527/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/27/756527/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/27/756527/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/27/756527/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/27/756527/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/27/756527/4"}}},"commit":{"parents":[{"commit":"59d60747b1b904c8442f7e0ef57a7e625508b42e","subject":"Prepare a new release for victoria cut off","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/59d60747b1b904c8442f7e0ef57a7e625508b42e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-10-07 14:41:40.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-10 04:06:45.000000000","tz":540},"subject":"Add new classes ovs_driver and sriov_driver","message":"Add new classes ovs_driver and sriov_driver\n\nThose classes handles ML2 plugin\u0027s configuration related to the\nOVS and SR-IOV mechanism drivers.\nPreviously there was neutron::plugins::ml2::mech_driver function which\nwas used by neutron::plugins::ml2 class but it isn\u0027t used at all\ncurrently and probably it will need to be deprecated and removed in\nfollow up patch(es).\n\nChange-Id: I2a321fb780e789b7556d557e2170a3267049c432\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0b8a90ec5c4cdac145672150ee4d58f7959cf6bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0b8a90ec5c4cdac145672150ee4d58f7959cf6bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
