)]}'
{"id":"openstack%2Fpuppet-neutron~757245","triplet_id":"openstack%2Fpuppet-neutron~master~Iaa52568d7a1c8bde97a24e87ead863c46a28b408","project":"openstack/puppet-neutron","branch":"master","topic":"min-bw","hashtags":[],"change_id":"Iaa52568d7a1c8bde97a24e87ead863c46a28b408","subject":"Deprecate duplicated parameters for vnic_type_blacklist","status":"MERGED","created":"2020-10-10 04:03:29.000000000","updated":"2020-10-12 19:49:24.000000000","submitted":"2020-10-12 19:40:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"757245-1602531625738-0a98ea85","meta_rev_id":"78f0b89d3d5f96d0f34aa1054b09944dde6c0edf","_number":757245,"virtual_id_number":757245,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-12 19:40:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"recommended":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2020-10-12 07:31:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-10-12 07:37:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-10-12 13:52:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-10-12 13:52:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-12 07:31:28.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-10-12 07:37:23.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-10-12 13:52:43.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-10-12 19:40:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c42c4ba34435eac5edd3675d2ac643aa5c4d4db6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-10 04:03:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b3ff349f926a2a60cfb93762c6e4be3aded06e0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-10 04:06:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"42f3d08a1d11ed74446092db6f31d29b8d19b3ae","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-10 04:08:04.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f89801b49bd40af2c591b4e4458f9aa1bb946403","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 06:27:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2319aa05f0b341e38437019cf1ebe836 : SUCCESS in 3m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80922b53c14a4223862a56e9e77b9053 : SUCCESS in 3m 01s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/85cc7c27ea6c4104b1dc7f7bec019921 : SUCCESS in 19m 51s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf92d299e6aa4716871d6be7f18b1873 : SUCCESS in 17m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64b0873369dd45379e9ec624463f3071 : FAILURE in 21m 16s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a2fa3cb5bec74850abc36708b3ea8007 : FAILURE in 30m 38s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0fd871fef90f421e94d9b9a3a28f334b : FAILURE in 26m 01s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a9d37cfe2444d68bede5e8dedc055e1 : FAILURE in 28m 00s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a7abbc822c62488c90920797a64a7d25 : FAILURE in 18m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f910c35e875c4ea0b38ce61cbd89f595 : SUCCESS in 1h 20m 41s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b368947c69ea4e04a3f3a7c4b4c9e0d9 : SUCCESS in 1h 53m 47s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e4eb2b019d6a44c084b2497a91b5f7d4 : SUCCESS in 1h 39m 54s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/aea83501ba594f3c90b20f6cb3854f6d : SUCCESS in 1h 02m 17s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/38e9043d09d04414856766d5c4030c8c : SUCCESS in 34m 48s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc02757931a745b7bf33964497414f95 : SUCCESS in 13m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50b0fdbf8ac34baf8e2683d4fecc6316 : SUCCESS in 3m 30s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0a6e9f08c81642948e38b3341eea310f : SUCCESS in 2h 12m 13s","accounts_in_message":[],"_revision_number":3},{"id":"cec3139661135b551d05d5aec9891e943fa43a9b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-12 07:31:28.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0747af0ac73c2a4df63d33ccd98397d1d9a03387","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-10-12 07:37:23.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ebfc6008b08e3b9f2fe05a004fe20f1532e1b4e1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-10-12 13:52:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c98cce77e610d222e3c971f9b5f5510d4559d1d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 13:52:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"079f9eac23d97aa37538e38a00049dcc01d6a4c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 19:40:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"be5d8fd82a6ddd54ae47b60015fee009dd116b3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 19:40:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10fa8cbb8ddf4cd5892afaee8380c8ed : SUCCESS in 3m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4130b55ed3c34f00bb3d961945f2a67a : SUCCESS in 3m 15s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/59260788acd8479db42f549b083646b0 : SUCCESS in 17m 35s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72432515c5ae4a36bf9b539550d85f90 : SUCCESS in 16m 30s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1389d83796a3434189ca03772cc6397c : SUCCESS in 1h 23m 58s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/088ef673b9824379ad1b4ec341a9926d : SUCCESS in 1h 30m 15s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6c175f80707c40dab4046e42c0bae104 : SUCCESS in 1h 27m 44s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/832d3adbb78c403f8a8bc70d3a0ef2e4 : SUCCESS in 1h 02m 06s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/16a83b3e5f484bf981f83fff6a89160e : SUCCESS in 23m 25s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/723d21715ba54d878801b2a081bc5f6d : SUCCESS in 16m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e337405d2724b7f90f506885ed8a990 : SUCCESS in 3m 39s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/393514e840024a20ac5cf49336624131 : SUCCESS in 2h 14m 24s","accounts_in_message":[],"_revision_number":3},{"id":"79446a41ee5c7d2ce195eadc06ff23915ed752b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 19:49:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b2631f0bca04f9088c886ab1a868bb3 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"28ca6097fcdb92998d0e14d8f2b110e8263beb13","revisions":{"d918a956a7696a39a68514b54f14da183806f0d1":{"kind":"REWORK","_number":1,"created":"2020-10-10 04:03:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/757245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/45/757245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/45/757245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/45/757245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/45/757245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/45/757245/1"}}},"commit":{"parents":[{"commit":"336c32a58c07917844fe40b93df8e96d3d535c3e","subject":"Add new classes ovs_driver and sriov_driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/336c32a58c07917844fe40b93df8e96d3d535c3e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-10 04:02:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-10 04:02:13.000000000","tz":540},"subject":"Deprecate duplicated parameters for vnic_type_blacklist","message":"Deprecate duplicated parameters for vnic_type_blacklist\n\n... because now the vnic_type_blacklist parameters are implemeted in\novs_driver and sriov_driver class.\n\nChange-Id: Iaa52568d7a1c8bde97a24e87ead863c46a28b408\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d918a956a7696a39a68514b54f14da183806f0d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d918a956a7696a39a68514b54f14da183806f0d1"}]},"branch":"refs/heads/master"},"3c37bf604072842bce692e8261a82f775eacc451":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-10-10 04:06:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/757245/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/45/757245/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/45/757245/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/45/757245/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/45/757245/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/45/757245/2"}}},"commit":{"parents":[{"commit":"0b8a90ec5c4cdac145672150ee4d58f7959cf6bb","subject":"Add new classes ovs_driver and sriov_driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0b8a90ec5c4cdac145672150ee4d58f7959cf6bb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-10 04:02:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-10 04:06:45.000000000","tz":540},"subject":"Deprecate duplicated parameters for vnic_type_blacklist","message":"Deprecate duplicated parameters for vnic_type_blacklist\n\n... because now the vnic_type_blacklist parameters are implemeted in\novs_driver and sriov_driver class.\n\nChange-Id: Iaa52568d7a1c8bde97a24e87ead863c46a28b408\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3c37bf604072842bce692e8261a82f775eacc451"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3c37bf604072842bce692e8261a82f775eacc451"}]},"branch":"refs/heads/master"},"28ca6097fcdb92998d0e14d8f2b110e8263beb13":{"kind":"REWORK","_number":3,"created":"2020-10-10 04:08:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/757245/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/45/757245/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/45/757245/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/45/757245/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/45/757245/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/45/757245/3"}}},"commit":{"parents":[{"commit":"0b8a90ec5c4cdac145672150ee4d58f7959cf6bb","subject":"Add new classes ovs_driver and sriov_driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0b8a90ec5c4cdac145672150ee4d58f7959cf6bb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-10 04:02:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-10 04:08:01.000000000","tz":0},"subject":"Deprecate duplicated parameters for vnic_type_blacklist","message":"Deprecate duplicated parameters for vnic_type_blacklist\n\n... because now the vnic_type_blacklist parameters are implemeted in\novs_driver and sriov_driver class.\n\nChange-Id: Iaa52568d7a1c8bde97a24e87ead863c46a28b408\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/28ca6097fcdb92998d0e14d8f2b110e8263beb13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/28ca6097fcdb92998d0e14d8f2b110e8263beb13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
