)]}'
{"id":"openstack%2Fpuppet-neutron~757747","triplet_id":"openstack%2Fpuppet-neutron~stable%2Ftrain~I2a321fb780e789b7556d557e2170a3267049c432","project":"openstack/puppet-neutron","branch":"stable/train","topic":"min-bw-stable/ussuri-stable/train","hashtags":[],"change_id":"I2a321fb780e789b7556d557e2170a3267049c432","subject":"Add new classes ovs_driver and sriov_driver","status":"MERGED","created":"2020-10-13 08:20:39.000000000","updated":"2020-10-18 15:56:35.000000000","submitted":"2020-10-18 15:55:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"757747-1603036530730-eaa21e4c","meta_rev_id":"b0baf3514a9703df2e63853f03a8ed22a9340c24","_number":757747,"virtual_id_number":757747,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-10-15 07:17:16.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-18 15:55:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-10-15 20:19:48.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-10-18 12:32:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-10-15 20:19:48.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-15 20:19:48.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-10-18 12:32:55.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-10-18 15:55:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce5a27cb4795e8002aa86e93ef64906862ed4327","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-13 08:20:39.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"80edae85bfece2394810cae3b03d6694269a2e75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 09:47:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a79b9b183a0468fb2768fa7c729fd2c : SUCCESS in 4m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19765dfbf26e455ebd585a2c7b29ef0c : SUCCESS in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6256094608f4c478e862f32e532a443 : SUCCESS in 5m 21s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/5281b6c107d54dea9c7f6f6c9a81c105 : SUCCESS in 25m 15s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/495066dd11594b0bab4712b3d7ce52aa : FAILURE in 19m 29s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/6b3384b72d974ce1b4ebdb6d4b54ef8f : SUCCESS in 16m 27s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/caf560c006db439d93bb3e3bc430ead8 : SUCCESS in 16m 10s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a80f75a7b8945d5830f45ee22abd9af : FAILURE in 28m 51s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/04d566e857cc4b06aa25138e15e5af82 : SUCCESS in 59m 54s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f38ab86d90ed4bf0b821c306e9178526 : SUCCESS in 1h 01m 22s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a29b9a3bced04827964233b4f3b52ad8 : SUCCESS in 1h 04m 29s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/7070d821e0574655b474319090458c89 : SUCCESS in 1h 08m 35s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8d989ed3553141de88049cc00cf59c70 : SUCCESS in 50m 27s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9e8985a9f6c4be0b8d62f2b4a96a58b : SUCCESS in 45m 49s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38a159f9f67b4f00b9ebe8642183678e : SUCCESS in 43m 10s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d1d97b286a2347ceb85e0851a07961eb : SUCCESS in 44m 04s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/436daf5a24fd427d938e0637b76d05ab : FAILURE in 9m 27s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1735de448d2f46d292a75b0b68dbb8a2 : FAILURE in 9m 19s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e741fa6fd31e4418ac46178fd69437e8 : FAILURE in 9m 55s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7ef116bdb3184b3d8442feff517301af : FAILURE in 14m 50s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2e64d7e6b52b4b6ab4da70353f09f097 : SUCCESS in 19m 29s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66725f2399ef414ca7b2ad000fc0796c : FAILURE in 6m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af935dd79bbc4d58b50eddf42e365c41 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"43b36bbde198eb5b4fe76bc6fd84e2cad97e3cec","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-15 07:17:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a2d6a0cc87a38a3db1196ccff6886e2272ff5527","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 08:29:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c165022231c4874b38f3dd0ac8936df : SUCCESS in 6m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac5eb79bfc0e4403ba3e6b2a5fc75b8f : SUCCESS in 5m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6e1510e2e3a4bfb9d0604330c71f15e : SUCCESS in 5m 41s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/aedf0a50d43246bc924893c9305a8cf7 : SUCCESS in 34m 24s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/586d6f6842db488c89869bcf1de27f65 : SUCCESS in 14m 28s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/91f0c6e657ad4df580a57e9195a6ede0 : SUCCESS in 21m 39s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9208bd043fad44638ad9d43b2a7e63c4 : SUCCESS in 18m 11s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ede37af717142828d054713169eba46 : FAILURE in 35m 22s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/4e55e59ff72a4513ab454a0cc5dd3566 : SUCCESS in 55m 13s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/64656873efdd469a937ae27ca94e378c : SUCCESS in 55m 58s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/6fed2285161f45588f7a88c7cf7a1b79 : SUCCESS in 1h 04m 57s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/6bd415e0ab6e4ae9b98fda80eab6495c : SUCCESS in 47m 59s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/941d68525f284ca6b6adf644945a03aa : SUCCESS in 48m 51s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e0ba84c8e854c0f85887bd76f1064d8 : SUCCESS in 45m 13s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3206bfee22114012a98d313519623214 : SUCCESS in 59m 10s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/acc4192219e84f4aae264a834f747195 : SUCCESS in 41m 36s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7866a19ac99845ebb19206cca69aec67 : FAILURE in 10m 32s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0b878276bc4448a8afba701962a32875 : FAILURE in 9m 16s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/41244a1d19de4477bbb64166af2dbce3 : FAILURE in 7m 40s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3218f680c9d547d8ac45fba0c33ec6de : SUCCESS in 41m 20s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7c87c6bf16aa44dfb888e92350cf8c21 : SUCCESS in 18m 53s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da5d70100fe24cd0a21bc9f9d8539162 : FAILURE in 6m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3e02b41ba9a4b689598f5ce7eeb34ec : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d0744c4ef014bb074046a8db8dfa1c5f28a2c87e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-10-15 20:19:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"05bafbf6072aa6d8ef771cd3dfcab8942fc7119d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 20:20:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b295e2272d85efcb8427bbd07430b21d4ebebc80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 21:34:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3776e656f9c4f9f8e9682eebd44df63 : SUCCESS in 7m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e60268b904b4e158a9f82d24991ea3c : SUCCESS in 7m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c15843d419c745bbb6562185a9e1f0b9 : SUCCESS in 7m 17s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/41427fa8452746f48a3ad36fd48d2f83 : SUCCESS in 32m 26s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a7a82c7da754ebeb6985f4f5afcb1c7 : SUCCESS in 17m 52s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/b31da278e34d4e8587783b3351e104fb : SUCCESS in 53m 02s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/40cd15fd95864f3485a6e464b7d3e9fa : SUCCESS in 1h 00m 00s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/dad4013fa3404f0cbb5d924f3c9b03f9 : FAILURE in 1h 02m 55s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/27528a6aeb2246f59d3613ad51b6e844 : SUCCESS in 49m 29s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/27f54586b1114f3f90fc1fdabfc2ad26 : SUCCESS in 26m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75b8874a78da4359895a1abd51072b34 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":1},{"id":"8b62017817acbcdf0d15d57f1203222975879196","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-18 12:32:55.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aecf118b23cc162bd4ad0d71f675f8400e85838e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-18 14:19:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4986c365246c45baa388b38736dcc5b4 : SUCCESS in 6m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5eb10a7f1051480ea5eb5bc6d41ea613 : SUCCESS in 5m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c20a4000a901470b94ee5ff22957537c : SUCCESS in 5m 20s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f3292632fbfc4a7f831cc45151caab9e : SUCCESS in 32m 35s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4290888068574981b12d2ac146c36ff8 : SUCCESS in 19m 53s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/0e5d481cc26d4031a71633f40e342e2a : SUCCESS in 21m 41s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60761584e7164f93b8413919f118b5ad : SUCCESS in 15m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37955a6ed4b547f7821b983ec2d276fd : FAILURE in 25m 28s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/11578ab6a3ee43c38639a773327da595 : SUCCESS in 53m 34s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4e1eea5ecb82471ca70a338e8d5dacdf : SUCCESS in 54m 12s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c9f50127555b481585bb3ee8ab058b87 : SUCCESS in 1h 01m 00s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/e0589e36f9eb42feb522284c0fd54649 : SUCCESS in 41m 17s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2ef6647d91f24af1b1163be95b37f648 : SUCCESS in 44m 01s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10d8669fcfdf4a5e93c55c92d3ca5bd8 : SUCCESS in 49m 02s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8cc1fe4c1adc45bf84a712572e34c7b3 : SUCCESS in 37m 38s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/85d198e519fa489ea274d4e6e2aaf1c2 : SUCCESS in 38m 34s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/704604d193d2489999bd1032afd47452 : FAILURE in 8m 08s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9c8bc5f713584a12a84bb61da283c895 : FAILURE in 9m 20s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cf69e28713814c8995b7860bec029cea : FAILURE in 11m 13s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d216b6ec9a114b238ad4128e2908b049 : SUCCESS in 41m 51s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d591c901dce24937b640e47c7e8796ef : SUCCESS in 19m 03s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e678c05db0904b17a7189b1ff58381b8 : FAILURE in 9m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a3f2e6dc79043e4bbee494d9d128fa2 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"4047d399b3d762ccef965a79cea9ab12224c43ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-18 14:19:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5ef27a2545eaae6253a276d29276f3076a015289","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-18 15:55:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21b9c6d8bd3044929d4dcf2d0940dcc3 : SUCCESS in 8m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3328e1cc6b384be8af208f1177de6dea : SUCCESS in 7m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/800089954e6d440b93b383806487dd36 : SUCCESS in 7m 55s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/0353f741ad984015b39a6039c06a03e0 : SUCCESS in 38m 00s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13dcea255aee4d69bedf13c8f8647c3a : SUCCESS in 16m 34s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/542afdeaf70046dcbea316119cff3e09 : SUCCESS in 54m 47s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/eb1ae8b75c8c4723b949db7459fc6ada : SUCCESS in 1h 08m 28s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c203287f343041c5b7d64c30010aec1f : SUCCESS in 58m 54s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/d95ddd30d35f4682bc0afa806f347bf5 : SUCCESS in 49m 10s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ac2cc70a1976494a824beabff7d61c8f : SUCCESS in 25m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc6e99d90288483a8f9a014993cf8105 : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c6880c153226570b7b3252c6562262b54f326aa4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-18 15:55:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b3d80676dc97b24900f8b5f31b7750bfffdb0c58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-18 15:56:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8de90f6a6a6547bc8bc66940763a4346 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a382db46e9efcaccce6b6ffbc2ef6da00b32936f","revisions":{"a382db46e9efcaccce6b6ffbc2ef6da00b32936f":{"kind":"REWORK","_number":1,"created":"2020-10-13 08:20:39.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/47/757747/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/47/757747/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/47/757747/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/47/757747/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/47/757747/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/47/757747/1"}}},"commit":{"parents":[{"commit":"7d0406b1a2bb054f42b198e9494ddc54372e7285","subject":"Nuage: remove check on mechanism driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7d0406b1a2bb054f42b198e9494ddc54372e7285"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-10-07 14:41:40.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-10-13 08:20:39.000000000","tz":0},"subject":"Add new classes ovs_driver and sriov_driver","message":"Add new classes ovs_driver and sriov_driver\n\nThose classes handles ML2 plugin\u0027s configuration related to the\nOVS and SR-IOV mechanism drivers.\nPreviously there was neutron::plugins::ml2::mech_driver function which\nwas used by neutron::plugins::ml2 class but it isn\u0027t used at all\ncurrently and probably it will need to be deprecated and removed in\nfollow up patch(es).\n\nChange-Id: I2a321fb780e789b7556d557e2170a3267049c432\n(cherry picked from commit 0b8a90ec5c4cdac145672150ee4d58f7959cf6bb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a382db46e9efcaccce6b6ffbc2ef6da00b32936f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a382db46e9efcaccce6b6ffbc2ef6da00b32936f"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
