)]}'
{"id":"openstack%2Fpuppet-neutron~758878","triplet_id":"openstack%2Fpuppet-neutron~master~I940575abfc2e21595143cf0fe12044b40d9de383","project":"openstack/puppet-neutron","branch":"master","topic":"at-params","hashtags":[],"change_id":"I940575abfc2e21595143cf0fe12044b40d9de383","subject":"Add support for the keystone_authtoken/service_type parameter","status":"MERGED","created":"2020-10-19 23:02:35.000000000","updated":"2020-10-22 16:53:32.000000000","submitted":"2020-10-22 16:51:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758878-1603385516912-e016dbdb","meta_rev_id":"5cff75a5d2098b90e60709782771e78cf6273d1a","_number":758878,"virtual_id_number":758878,"owner":{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-22 16:51:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"all":[{"value":1,"date":"2020-10-22 02:00:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2020-10-22 00:48:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-10-22 14:15:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-10-22 14:15:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-22 00:48:51.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-10-22 02:00:42.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2020-10-22 14:15:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-10-22 16:51:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"512dfa665d31774abe3a921875a4ff7ffdc12bef","author":{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"},"date":"2020-10-19 23:02:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"faf206a2b3187176f6a22a30fdba0df2ddc39e91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-20 01:31:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/daab5b2a951849fc92408ab2c021392a : SUCCESS in 4m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0a0b72ada8a4a73ad7dd0a431dd0499 : SUCCESS in 4m 31s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1cbe1518380845548237a9672449e698 : SUCCESS in 19m 27s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e8d1589a9aa4e81a2ce05603ee90239 : SUCCESS in 18m 29s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48de388c73554cdbb2e80bfbcb396412 : FAILURE in 23m 19s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/88039277f10e41aba3cf680a7e4c4ace : FAILURE in 30m 12s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81b9d15f98394893a59cfae3d274e2b8 : SUCCESS in 55m 20s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee308a099b114e58a9f53fe7d74ed65b : SUCCESS in 45m 46s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ff8f80c90d6e4732a6702c148a64c32a : SUCCESS in 41m 59s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a45c8d4ba5bc430d98048be9117cdb3f : SUCCESS in 1h 10m 25s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/90bc1f848a9844719de9f6cc7c623256 : SUCCESS in 1h 33m 21s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c717ae10d849422d9f5d0ba2bd3d1327 : SUCCESS in 1h 30m 22s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/949ce66ab7a74f169a35d95de72e6d67 : SUCCESS in 58m 01s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/ca306ebd295d4389a4ddc93b9691c619 : SUCCESS in 22m 59s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2520082bd455421fb8e5e7f575d1065f : SUCCESS in 16m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a26a7be5ed424173812d425cd63da109 : SUCCESS in 4m 29s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6db22b965518426ebdb943bb83100b90 : SUCCESS in 2h 23m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d5b736d043f509da26203216b5b6736bd3920bc8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-22 00:48:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7db19eeb5b5181aa9c8cac79be583f8aed9db980","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-10-22 02:00:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9fd26c64bffb34b8262b29add2ff88d0122e1781","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-10-22 14:15:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4150437c80b0b9df42b3fd88e2795fe080cabb92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 14:15:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d3f2ea21a647d524b69ef2854f647eb043cda95e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 16:51:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b1f5824db614437a62db2a2284e975e : SUCCESS in 6m 28s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3b38f52046147a89a1b4a6c7c78eb0d : SUCCESS in 6m 47s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0bcf82275d594cdf9488b587cd857fcc : SUCCESS in 23m 44s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76ed387e4f98455d8a6f19e61f509708 : SUCCESS in 19m 32s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f126871c589548749024dce19ce6c2b8 : SUCCESS in 1h 26m 46s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d968761b8dc0416fad297d668b6f32f4 : SUCCESS in 1h 43m 27s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2090930eefb7408b9fcbca8be200161e : SUCCESS in 1h 29m 16s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9e26d990ec574da1aca91137fae15a68 : SUCCESS in 57m 13s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/7cacdb51b8cd40a18f91695ad9fcc15e : SUCCESS in 24m 47s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cd6d5ae5c3946cd847eff455edce9fb : SUCCESS in 18m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/409fd5bdb45d4528abd133bab80b3ac4 : SUCCESS in 6m 40s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a8ce2ef25f3249169e4a910d34c333f6 : SUCCESS in 2h 28m 37s","accounts_in_message":[],"_revision_number":1},{"id":"7c3572d06e60e7b186002c5fa0c4ad5657c8f5ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 16:51:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dede18cae71c774b8316500a02485fad2cd47e19","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 16:53:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/710c64e587864ace81c47b8a10f8ce40 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"43a2ed4380a8ef0a2e6c64d4ad08193da6f75eca","revisions":{"43a2ed4380a8ef0a2e6c64d4ad08193da6f75eca":{"kind":"REWORK","_number":1,"created":"2020-10-19 23:02:35.000000000","uploader":{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"},"ref":"refs/changes/78/758878/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/78/758878/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/78/758878/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/78/758878/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/78/758878/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/78/758878/1"}}},"commit":{"parents":[{"commit":"2663cdf4735c7f71b037c8ba79c8f3f4323b27eb","subject":"Bump version for the start of Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2663cdf4735c7f71b037c8ba79c8f3f4323b27eb"}]}],"author":{"name":"Christopher Brown","email":"chrisbro@redhat.com","date":"2020-10-19 23:00:54.000000000","tz":600},"committer":{"name":"Christopher Brown","email":"chrisbro@redhat.com","date":"2020-10-19 23:02:32.000000000","tz":600},"subject":"Add support for the keystone_authtoken/service_type parameter","message":"Add support for the keystone_authtoken/service_type parameter\n\nChange-Id: I940575abfc2e21595143cf0fe12044b40d9de383\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/43a2ed4380a8ef0a2e6c64d4ad08193da6f75eca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/43a2ed4380a8ef0a2e6c64d4ad08193da6f75eca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
