)]}'
{"id":"openstack%2Fpuppet-neutron~760416","triplet_id":"openstack%2Fpuppet-neutron~master~I6ed8fe0cc60cd2775a8ae60ed22a4824f4bc7aad","project":"openstack/puppet-neutron","branch":"master","topic":"new-ci-job","hashtags":[],"change_id":"I6ed8fe0cc60cd2775a8ae60ed22a4824f4bc7aad","subject":"Build containers for single consumer job","status":"MERGED","created":"2020-10-29 17:03:59.000000000","updated":"2020-11-12 03:34:24.000000000","submitted":"2020-11-12 03:34:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760416-1605152064277-422172d5","meta_rev_id":"91dcfd25f067420a2888ba799abbf585293aee21","_number":760416,"virtual_id_number":760416,"owner":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"date":"2020-11-11 22:02:29.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-12 03:34:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2020-11-09 13:13:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":2,"date":"2020-11-06 22:17:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2020-11-09 13:13:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-29 17:04:27.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-10-29 19:06:44.000000000","updated_by":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"reviewer":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"state":"REVIEWER"},{"updated":"2020-11-06 22:17:50.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-11-09 13:13:53.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-11-12 03:34:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b74be46eb2194eea5851dfaa85aced7688ae2fe9","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2020-10-29 17:03:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"234d0e5027372fa2f5004eb703f51de3357c2f65","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-10-29 17:35:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b95ab35419f2b0336d43d9cb06d1cee010bee6bf","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2020-10-29 19:06:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"231540f2814390cb2f8a11aa040e363aec495e55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 21:16:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b850ded2b20d4e28b52c22bd9b4f3ed9 : SUCCESS in 4m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8fabe9c972134861b2b61ba8d4842865 : SUCCESS in 5m 13s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/04fde1f918444fc8828dbe074e093f10 : SUCCESS in 20m 15s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de6d9462dde049b1a6b49d3ed425f01f : SUCCESS in 15m 32s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80f13be9d78b44edac03fbffab62a2f2 : FAILURE in 17m 55s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/abe17edfd4d34ed7806711aea8d38c7f : FAILURE in 9m 11s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9bbcc7f20e3454a9f0066a1edfb7bce : FAILURE in 9m 51s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ef7f3fcfb7f4466972c813e591e5014 : FAILURE in 9m 40s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b20c83634fcc43c59d7f8f8546c6a796 : FAILURE in 9m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f21275e4c13d475a8e14c4d01ec4f565 : SUCCESS in 1h 19m 50s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ad25e9a4291041d0b779f3c685382c02 : SUCCESS in 1h 22m 51s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/42e8d47451f04dadb685f641078938fa : SUCCESS in 1h 18m 03s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a209399b9bb94df48354a7106c3835da : SUCCESS in 55m 28s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/e27fee741d8d4e9582183f023f5494e8 : SUCCESS in 21m 32s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d95e0b91a94435b993f8343a657526f : SUCCESS in 17m 50s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/163e1fc1123e4edaae07d33267e23dd5 : SUCCESS in 4h 01m 12s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/804047cbac604802b001bc0f6f2a88ee : FAILURE in 2h 54m 11s","accounts_in_message":[],"_revision_number":1},{"id":"560040180e547f96614c9a2013fb50b531f0ad8a","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2020-10-30 18:31:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f497dbc1f193857325d14a8096f2f315c853b71f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 22:48:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/148a6af27d4d4d9d861318cd2fd3af3d : SUCCESS in 2m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0219190daafe4832abac3851df74d6c4 : SUCCESS in 3m 19s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/b6255c5cfe494f7bb2a1dbad422568de : SUCCESS in 17m 48s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4d8cd423d9e4fe0a7f3cee078d9c993 : SUCCESS in 17m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8cbe789225eb4f739857e06543893886 : FAILURE in 17m 30s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/841d8f24bda64c92acea50654ec260d2 : FAILURE in 6m 53s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1d45c2684b54651a6115f1c3b25e78a : FAILURE in 6m 48s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66f46561f4324988bc8a4f0bc55489ca : FAILURE in 6m 35s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/9a3439bbd72d4d278fbcb4ec8ae9dac3 : FAILURE in 17m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4e8df130975a4f0c8a2be887d45281bd : FAILURE in 28m 39s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/54089d8571194a86bed5f2cb1f55d3d2 : SUCCESS in 1h 36m 15s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8aa455475601406ea32fa089799d2543 : FAILURE in 1h 24m 35s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b8a78bb3b2c34c46ac99f2abc57d684f : SUCCESS in 1h 02m 28s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/c04394f6d5584bf6a44f2ae1bde9a31b : SUCCESS in 21m 26s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1dbdbe2d3c154fbea6e5642d13066a6f : SUCCESS in 14m 31s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f445fb4340e247109e51a74c0213051b : SUCCESS in 4h 05m 52s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/957b6bc2892447c2b8f7151fd1165e1f : SUCCESS in 2h 36m 34s","accounts_in_message":[],"_revision_number":1},{"id":"8d8d54ecc90cd7e79902b60c0356ac9e898fd9ca","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-10-31 20:40:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"173575551b121f346a52853e75f6b5225328f397","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 00:10:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ff28055ef5345b4a032fae2ff74bf83 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4bbb3f95bb3d469f9b1f0e3a30769add : SUCCESS in 3m 15s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c1e2be8815244364bc389e0f747bcfd0 : SUCCESS in 17m 12s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35389b9348c54b7fa105bf84f5242c80 : SUCCESS in 14m 35s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae5c11af29f440ae86e4ab7dbdaa2eff : FAILURE in 16m 02s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/639a8df81b9544559a106f0616dd1d66 : FAILURE in 4m 57s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/339c8bf6c1764dabb30af04f1cf25749 : FAILURE in 6m 37s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a85a17469384d7f90f08857c928fee6 : FAILURE in 5m 43s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8ee27e998b604ba899a1f1be6ee26ffc : FAILURE in 6m 39s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/052877a4e0264f88a3315f1def42ab10 : SUCCESS in 1h 22m 02s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a80fcdecfe924b95a02c3264d9d9af54 : SUCCESS in 1h 32m 17s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/286656318afb4388ac4c4f6153380741 : SUCCESS in 1h 31m 55s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/83e22ee2c78f4dbca4bb59ce15794c16 : SUCCESS in 59m 13s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/1940f944641a488ebb019be459814b25 : FAILURE in 11m 25s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/46e63ca69e5544979ee23f35a493d114 : SUCCESS in 14m 59s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/46f36a67a32c4f68abc01a96a9b114e9 : SUCCESS in 3h 24m 47s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/53b03a7cd3d24cc6b9e73f5dfb598b8b : SUCCESS in 2h 09m 12s","accounts_in_message":[],"_revision_number":1},{"id":"646371e99150776585a1a3bc1403092fa1f4855b","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2020-11-01 23:16:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a4c7a2b83e86945d76d0a06766de4423d8c4e811","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 01:46:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/301e5e422acf4fd28af41f76f0a0c23b : SUCCESS in 2m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ccab075253e402e9b2b90e1853e0634 : SUCCESS in 3m 37s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/4e2f81b34f284b409ae5dde06c42a250 : SUCCESS in 16m 58s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9ec651326654039ab675dbdff39bbee : SUCCESS in 15m 42s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/554f0b6593144e0f96526f0615a653af : FAILURE in 11m 58s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5c77acf76d76414489182bc9c7e07cb9 : FAILURE in 9m 38s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00a9d1694e2a422e93b61cddad5333d0 : FAILURE in 8m 47s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24a7a02e1d794393a8f8b7c948fca57d : FAILURE in 7m 11s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5d2f7ace3ba849f9b0114b62e411c952 : FAILURE in 6m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8e675a80f9e2457b94eab9089de18ae7 : SUCCESS in 52m 36s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c981170344eb482f96368aa30658ae7d : SUCCESS in 1h 03m 28s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dc373fcf07144ec791725e56c26f68bb : SUCCESS in 1h 00m 03s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ba9429df7124413ea99f03eb907ae78c : SUCCESS in 37m 59s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/361f14b4ef794298aacadd621c806317 : FAILURE in 11m 08s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/054beac0a3244be9a399c0584757535f : SUCCESS in 18m 00s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/15855d9af1b94c45b1bf907a52fcc54b : SUCCESS in 2h 28m 43s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fec821e8e2504ac5b4f2ace5d63a7b32 : SUCCESS in 1h 26m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e72913fc50f7b40c06625ba8ede60a1c8dfc37ad","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2020-11-06 14:22:45.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"47fd1a7ba49716a6222fed27c38cfd01eca54ae4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 17:14:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85404a847c3c4820aa70c0f58d62c3a0 : SUCCESS in 6m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1aa84fcce0e6492996033db83673d95f : SUCCESS in 3m 15s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9f17fa13ad0947b49844e0aa5914dbd1 : SUCCESS in 16m 27s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6dac1d96328649ceacbcd08188dfa318 : SUCCESS in 14m 15s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36cfa19894834e378c718bab338b7740 : FAILURE in 16m 58s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/86fb3a945e2d4543b3eed2168773e38e : FAILURE in 6m 51s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/988bd9cdbf954415a6a7a11f99ab30a5 : FAILURE in 6m 29s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0fe49bcc6077492a89b11bb28aa33d4d : FAILURE in 5m 46s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/fe6b184119b64ed8a1da973c4ac723b5 : FAILURE in 6m 59s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9ac38f5e75ac478485d1078e7808e5d4 : SUCCESS in 1h 21m 15s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4bee20b8b4234fe4b1f140fa9e44fbea : SUCCESS in 1h 32m 40s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f6eefb28117d44c09199b16b67cc0fa0 : SUCCESS in 1h 35m 48s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/600aa8b933c846b580d9f7b976886d9d : SUCCESS in 1h 00m 40s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/9f077da54dce4c36b402de6810e0c99c : SUCCESS in 20m 36s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8550993d02814faeb6040012a7ea05e4 : SUCCESS in 14m 38s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/528b47e2f3984678b46f1c6a545a95da : FAILURE in 2h 48m 20s","accounts_in_message":[],"_revision_number":2},{"id":"6747b40bf3a06eddc241df113051ee1de6ab34c0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-06 22:17:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c0e24c367bfcc177ea6a78480812dfa1f6dfd8c2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-09 13:13:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9cd13af71da6ff7a9f5de7672ad26901c49fc77a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 16:21:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e2497f4c74843d3bb0ea1069b5e2640 : SUCCESS in 3m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06bede28c43b41829a64bff4befc9a48 : SUCCESS in 3m 47s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/54ad139f84ed41ad80e331a08c2e2ed7 : SUCCESS in 18m 59s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39876c10f47d4d269e6c434e9e8b2211 : SUCCESS in 17m 47s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c8b4ec050fc480bbffb45cba8934d83 : FAILURE in 14m 16s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/98ec7a1ff6984afe8ab8eefdcff14fd1 : SUCCESS in 52m 15s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8bf0b35990d0402d920d53ab07e08ca7 : SUCCESS in 57m 49s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a81632215844e3e896bb716aa4d86e2 : SUCCESS in 47m 12s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5a69715c1e144100a18687f643d5aca2 : SUCCESS in 48m 45s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/769f42f40cd647efb3fd7262245d4e35 : SUCCESS in 1h 24m 43s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ebba106ea7aa4879bdf30be46c56f27e : SUCCESS in 1h 41m 08s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ae77876f66d0432f9e345d1b09414c67 : SUCCESS in 1h 45m 59s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f5fb28cbb87a47158050dac33d2e889b : SUCCESS in 1h 01m 05s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/1ab2dfeb490540c9a9fee50dfc52aad1 : SUCCESS in 23m 04s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5b89cb9be3c4bed91ec0bab0380f38b : SUCCESS in 18m 05s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/80b66e7d1b4f4a9d9d4f85ad56691c6b : FAILURE in 46m 30s","accounts_in_message":[],"_revision_number":2},{"id":"f1bb2682a2b22a4e0957e74e189708f5b994b05a","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2020-11-11 22:02:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"71c5fa672f3edb871ada2283eb8b3f7da6b92e25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 00:47:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8779312ebf5a4caa9eee3ad2fd84a892 : SUCCESS in 3m 25s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a89fc12a51ae460a81124b591ef4ef07 : SUCCESS in 4m 02s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/53d9079447fe4931b0728edbc6a0227c : SUCCESS in 18m 37s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b6ffed508614f3c938da7b12fd8f401 : SUCCESS in 16m 45s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0a48ff827f042058441edb866dfa00c : FAILURE in 17m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/de5215cca9414281abbe5c7629a6e36b : SUCCESS in 51m 32s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df86f291f42048fb847b0c18f535111a : SUCCESS in 59m 21s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8348e0931e954260847c83198a18af17 : SUCCESS in 49m 47s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ec729198c7ed4aa2a023b1bdb2c6df7b : SUCCESS in 47m 58s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0e7e8e3c869442a9a24427ea85ec73ea : SUCCESS in 1h 16m 42s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/55c3cb5dff2649d0bd4affc284e0b3c5 : SUCCESS in 1h 49m 55s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7ab1175097b8498da9cbe67f07b7daa0 : SUCCESS in 1h 27m 42s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bedae0f35ec94e27a8a9201939520939 : SUCCESS in 59m 58s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/5e71d392a1374d6c867e3e800e1f50b2 : SUCCESS in 21m 53s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/207e53c447a946cbbb32a1379fca1e72 : SUCCESS in 18m 07s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0676ccf3674748cca98f3bd5d4f90098 : SUCCESS in 2h 38m 51s","accounts_in_message":[],"_revision_number":2},{"id":"accce83e134d74b1539e1e73cbcdd210370454fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 00:47:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"00017bd755cd26829f70bc16d807eead1e9ee953","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 03:34:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc6262c80e274fad9cd22a372cde93ac : SUCCESS in 3m 59s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1d21f2d89d942408f1f8695713554f5 : SUCCESS in 2m 59s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/51b77a2a80084117909ff4bf6f53d8cb : SUCCESS in 18m 04s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/829477d8fd4e46798fe5d9a57cb24722 : SUCCESS in 13m 56s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/981b8f05be154f5ab6757eaed48f2065 : SUCCESS in 1h 06m 01s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2a82a119265643478481eea473f36dc7 : SUCCESS in 1h 51m 59s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7d89a9c5abc34eb6b09699acb9d9ac15 : SUCCESS in 1h 29m 19s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6c86b26184694f9a9ed444b139c787d2 : SUCCESS in 54m 10s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/edb14b4d5c4f4fc99b298a2603207dba : SUCCESS in 22m 43s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69eaf98277014ba4aff6f12c8615337d : SUCCESS in 13m 36s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c4cf85b4326f45a2b840cf43c530b23d : SUCCESS in 2h 45m 53s","accounts_in_message":[],"_revision_number":2},{"id":"735cfb05aa3000e872da5d08b638dd61bdc5b5f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 03:34:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"515f58c18eb0ef460e98e45216a838ca524feb7c","revisions":{"f78ed44e5e7a39f343cb2abf6d05074d08fd78e9":{"kind":"REWORK","_number":1,"created":"2020-10-29 17:03:59.000000000","uploader":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"ref":"refs/changes/16/760416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/16/760416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/760416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/760416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/760416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/16/760416/1"}}},"commit":{"parents":[{"commit":"5e846a6e8edc3f6f2866610549b29fedc2f6a86d","subject":"Fix outdated files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5e846a6e8edc3f6f2866610549b29fedc2f6a86d"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2020-10-29 17:03:39.000000000","tz":-240},"committer":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2020-10-29 17:03:58.000000000","tz":-240},"subject":"Add content provider jobs for tripleo based jobs","message":"Add content provider jobs for tripleo based jobs\n\nRelated change:\nhttps://review.opendev.org/#/c/760249/\n\nChange-Id: I6ed8fe0cc60cd2775a8ae60ed22a4824f4bc7aad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f78ed44e5e7a39f343cb2abf6d05074d08fd78e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f78ed44e5e7a39f343cb2abf6d05074d08fd78e9"}]},"branch":"refs/heads/master"},"515f58c18eb0ef460e98e45216a838ca524feb7c":{"kind":"REWORK","_number":2,"created":"2020-11-06 14:22:45.000000000","uploader":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"ref":"refs/changes/16/760416/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/16/760416/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/760416/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/760416/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/16/760416/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/16/760416/2"}}},"commit":{"parents":[{"commit":"5e846a6e8edc3f6f2866610549b29fedc2f6a86d","subject":"Fix outdated files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5e846a6e8edc3f6f2866610549b29fedc2f6a86d"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2020-10-29 17:03:39.000000000","tz":-240},"committer":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2020-11-06 14:22:35.000000000","tz":0},"subject":"Build containers for single consumer job","message":"Build containers for single consumer job\n\nMake standalone job build containers when there is one single\nconsumer job in layout. Do not run content provider jobs in\nthis case.\n\nChange-Id: I6ed8fe0cc60cd2775a8ae60ed22a4824f4bc7aad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/515f58c18eb0ef460e98e45216a838ca524feb7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/515f58c18eb0ef460e98e45216a838ca524feb7c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
