)]}'
{"id":"openstack%2Fpuppet-neutron~760995","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fvictoria~I6ed8fe0cc60cd2775a8ae60ed22a4824f4bc7aad","project":"openstack/puppet-neutron","branch":"stable/victoria","topic":"new-ci-job","hashtags":[],"change_id":"I6ed8fe0cc60cd2775a8ae60ed22a4824f4bc7aad","subject":"Build containers for single consumer job","status":"MERGED","created":"2020-11-02 19:46:47.000000000","updated":"2020-11-11 23:17:53.000000000","submitted":"2020-11-11 23:17:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760995-1605136673101-e490bc4f","meta_rev_id":"12266f1cfedc33877fd5a5224b1bf6c4021b4c48","_number":760995,"virtual_id_number":760995,"owner":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-11 23:17:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},{"value":0,"date":"2020-11-11 17:01:04.000000000","_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2020-11-06 13:35:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":2,"date":"2020-11-06 22:17:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-10 03:55:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":1,"date":"2020-11-11 20:06:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-02 19:46:47.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2020-11-06 13:35:20.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-11-06 22:17:51.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-11-10 03:55:02.000000000","updated_by":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"reviewer":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"state":"REVIEWER"},{"updated":"2020-11-11 20:06:12.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-11-11 23:17:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b67b485e49c15593ccaec9131459c2d990a02ea0","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2020-11-02 19:46:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"760584457fd801c67ef5afe16e30fe4e3e29e56d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 23:09:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6778ff8fcf7349fbb365d6e705830c82 : SUCCESS in 6m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9d6a3f47644465c9b651a8ec5024ee9 : SUCCESS in 4m 00s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/7fb04c3bebab45d4a20d11d2c3495051 : SUCCESS in 16m 03s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f80a254915dc4e8e81dad1b15c6d8216 : SUCCESS in 14m 23s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64a7065fd7914d0d8e5c63b8f4658779 : FAILURE in 19m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e135c08d081d4da8aabc47a09dc9e939 : FAILURE in 5m 22s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2fef99a4b5c04187be5cb839a402841d : FAILURE in 9m 08s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56e9281ce70242518a437de3a7dfb030 : FAILURE in 9m 14s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4099e34f8e5c4dc0ad226bad489a9613 : FAILURE in 9m 24s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e6f6bdf2523b42c7b8d10f038bf7e247 : SUCCESS in 1h 26m 03s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/62a3fba17bea490db064feace1c69c69 : SUCCESS in 1h 27m 47s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5ce1d3135a024d44b9bb8c7c02e635b7 : SUCCESS in 1h 29m 15s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/caa5186467bf4017ba637ce1adab58ae : SUCCESS in 1h 08m 45s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/d1b77d6f34e64a259bce5486b323c976 : SUCCESS in 24m 17s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9946536975b4481a856873846c26ee9 : SUCCESS in 20m 26s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1311c9af598149959b2eb1af07974472 : SUCCESS in 3h 18m 20s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3671c094a2b14fef88e68d798e9c2ae6 : SUCCESS in 2h 06m 38s","accounts_in_message":[],"_revision_number":1},{"id":"492690f8b439a520f64e48eb56d1d23f8cdd6582","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2020-11-06 13:27:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0d4e0d26a041da732b6ac273ebe360c5562ed362","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-11-06 13:35:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"973d358ea17d591e7a5882568405c9de1da1892d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 15:14:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b61739296ad74504add1f38912b5b0e4 : SUCCESS in 5m 25s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ada1fd60b9704b59ae6c395a47a815d0 : SUCCESS in 3m 48s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/4075f1ca989847d69c2c93c675f20faa : SUCCESS in 21m 23s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c85c49e81ce34fe7b11446a968a5a10e : SUCCESS in 16m 06s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f40592552a24b90bbfb1a71a2e011d2 : FAILURE in 17m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/3e33a8c0e1c642d5a6446e38dbc1c056 : FAILURE in 10m 14s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29f19f8aa39d447aa962d3382c42723b : FAILURE in 5m 32s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35b573867eec4c36819a9e78b37a7e9f : FAILURE in 8m 53s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ff9ababc8f724c2e9c51cce5dc95eace : FAILURE in 7m 11s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/75bdade21b0644a4b22590a478541d46 : SUCCESS in 52m 29s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9f8c16c0f6714cd7a4e12c85b5239b41 : SUCCESS in 1h 37m 32s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/336a642125f84a7990ac94c356634e25 : SUCCESS in 1h 24m 25s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3688a373d1f645e6942725bf45c4a43a : SUCCESS in 1h 02m 57s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/a985e970b0b84e008426cb9314906862 : SUCCESS in 19m 36s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c68790602bda4f5794749a61504645a1 : SUCCESS in 17m 20s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/61312801a03a482989fc28ef1e9e19e0 : FAILURE in 57m 02s","accounts_in_message":[],"_revision_number":2},{"id":"fd578c4d51f4bdf90c166db4c089bf26b1bb7fe7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-06 22:17:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ed78259b05d431598a3b15ba0df034bbbc1c6f72","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2020-11-09 12:26:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f3a56d9b61fe6bb08798b2666b9943381624c0da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 14:24:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3839d3a0b2124171adadfde736c81ab8 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/355a89092e2e455f8f9ec9b13b1b9add : SUCCESS in 3m 15s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/8290beb51d80441a887fa8eb5aa6ac7b : SUCCESS in 16m 07s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fadac735c0a74480ba4befc71f597b05 : SUCCESS in 16m 59s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dcdf24e03ddd4085aeb46e39ee93a730 : FAILURE in 17m 47s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/9fc4a21111154eafb9a5213ddd36eb1a : FAILURE in 6m 22s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ac8352a31b149b99a369a86fa55309f : FAILURE in 5m 09s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca39dfd39b814b3b92225cdee6a08a1e : FAILURE in 6m 17s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/04952f155a9d47eea682225e4f8a4fde : FAILURE in 5m 04s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e4369555ac1b4cb09d4ddb7a83e0d51f : SUCCESS in 1h 19m 39s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/628ec8802f8f458b873f1018c384f3f3 : SUCCESS in 1h 50m 43s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8ff758579b0244fc9060e3cdcf88946a : SUCCESS in 1h 21m 58s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/15586339cddd4f919929739a09529d61 : SUCCESS in 56m 24s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/45dc6cb39473426a9b7a2e2288bc7471 : SUCCESS in 28m 57s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc58c0d618434fc0b47037a019242631 : SUCCESS in 17m 18s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f27eed05476d4e8797c30bcd258b651a : FAILURE in 48m 03s","accounts_in_message":[],"_revision_number":2},{"id":"c0820ef65abf1ff4f2739e52713c4374957cd6c1","author":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"date":"2020-11-10 03:55:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"404eefec24e88e05d4ad3bb8f4033e97676cba8a","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2020-11-11 17:01:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"883924ddcdaa3ebc8e580714bb0e8abe46ab7be1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 19:54:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b79a9e29d7134a938d314be559d5c11c : SUCCESS in 6m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82d9d47a4f0741d6a91ae5ecb99064c5 : SUCCESS in 3m 09s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0495ab80f88b4e8f91bdeb07c470d61d : SUCCESS in 15m 27s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5df0b87781c4ffbaf7ebfbce40f75c7 : SUCCESS in 13m 37s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/403ab033d64e41a7ac4ec9a4592e8fa5 : FAILURE in 15m 09s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/64110550a37848c89ae778abf89ed73c : SUCCESS in 49m 16s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/206490b8c8734d7c8b52c561dbabeb33 : SUCCESS in 53m 16s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9514e51b79be4ed3b6c55873f3e6b604 : SUCCESS in 52m 04s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/66c48d024da440868c20abcb8102f678 : SUCCESS in 1h 03m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c080b54f7f75498699c712e863396779 : SUCCESS in 1h 21m 55s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5a484e6075ca4e1aaddfe69c97b7e477 : SUCCESS in 1h 39m 30s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c18392b55bb042f785b116128b46134c : SUCCESS in 1h 19m 34s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ed27b6e1fd6a4f7490c2ea7e49f07608 : SUCCESS in 1h 05m 51s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/c22aa73c4eee46578f26bd56218f26be : SUCCESS in 18m 46s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2778a1ef02084e058e18cbf6d795920b : SUCCESS in 14m 18s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e9c00d1cc3f64916a27c2d8a593b46d6 : SUCCESS in 2h 42m 43s","accounts_in_message":[],"_revision_number":2},{"id":"9aaeaba2310f38ac96c09ef26f733d6cc1182789","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-11-11 20:06:12.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"83f6cc2d162f9a844f0be2d2f5e93cd11082c7a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 20:06:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"36d7640eeea2a4fa70b168d647be7cb6c163b85d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 23:17:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eef691e4a2a74c109949ef024cc65442 : SUCCESS in 3m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a8536387b53f48dfbe662e3b37d4b838 : SUCCESS in 3m 17s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1730db339c0f4ce39aa925b6d1c91602 : SUCCESS in 19m 49s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5011ace710494557b249e36cd1a2dd93 : SUCCESS in 16m 18s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cf208fcfad714add86048442f05df88f : SUCCESS in 1h 34m 46s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5a313bb470f74f24a534db9f0ea3df0c : SUCCESS in 1h 19m 38s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ce87f9e3306240c7bfef6491f1233854 : SUCCESS in 1h 34m 49s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ba8da8ead60b43a39c9c7ab2b9511d57 : SUCCESS in 56m 51s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/137fb09315954c17b83507915dc5e252 : SUCCESS in 24m 55s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d7d3ef0dd2a4c638ef5fd2115134476 : SUCCESS in 17m 12s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/21bc9d0cc157420f90e97caf9824959a : SUCCESS in 3h 05m 12s","accounts_in_message":[],"_revision_number":2},{"id":"036131c040816e33626ab5219facf8a92bfbe422","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 23:17:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"402e24d86de28b25b520ea017d1c16aedada108c","revisions":{"c06766ccdecc19f3395280c5ef3607a57bf38c5c":{"kind":"REWORK","_number":1,"created":"2020-11-02 19:46:47.000000000","uploader":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"ref":"refs/changes/95/760995/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/95/760995/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/95/760995/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/95/760995/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/95/760995/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/95/760995/1"}}},"commit":{"parents":[{"commit":"6ea43b72a45a1c8805d992cd7284ae2f18b93401","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/victoria\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6ea43b72a45a1c8805d992cd7284ae2f18b93401"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2020-10-29 17:03:39.000000000","tz":-240},"committer":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2020-11-02 19:46:44.000000000","tz":-180},"subject":"Add content provider jobs for tripleo based jobs","message":"Add content provider jobs for tripleo based jobs\n\nRelated change:\nhttps://review.opendev.org/#/c/760249/\n\nChange-Id: I6ed8fe0cc60cd2775a8ae60ed22a4824f4bc7aad\n(cherry picked from commit f78ed44e5e7a39f343cb2abf6d05074d08fd78e9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c06766ccdecc19f3395280c5ef3607a57bf38c5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c06766ccdecc19f3395280c5ef3607a57bf38c5c"}]},"branch":"refs/heads/stable/victoria"},"402e24d86de28b25b520ea017d1c16aedada108c":{"kind":"REWORK","_number":2,"created":"2020-11-06 13:27:53.000000000","uploader":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"ref":"refs/changes/95/760995/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/95/760995/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/95/760995/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/95/760995/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/95/760995/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/95/760995/2"}}},"commit":{"parents":[{"commit":"6ea43b72a45a1c8805d992cd7284ae2f18b93401","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/victoria\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6ea43b72a45a1c8805d992cd7284ae2f18b93401"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2020-10-29 17:03:39.000000000","tz":-240},"committer":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2020-11-06 13:27:50.000000000","tz":0},"subject":"Build containers for single consumer job","message":"Build containers for single consumer job\n\nMake standalone job build containers when there is one single\nconsumer job in layout. Do not run content provider jobs in\nthis case.\n\nChange-Id: I6ed8fe0cc60cd2775a8ae60ed22a4824f4bc7aad\n(cherry picked from commit f78ed44e5e7a39f343cb2abf6d05074d08fd78e9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/402e24d86de28b25b520ea017d1c16aedada108c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/402e24d86de28b25b520ea017d1c16aedada108c"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
