)]}'
{"id":"openstack%2Fpuppet-neutron~762827","triplet_id":"openstack%2Fpuppet-neutron~master~Icd834f4f4ca6578c836935614931af2a3e79bf7d","project":"openstack/puppet-neutron","branch":"master","topic":"rp-bw-default","hashtags":[],"change_id":"Icd834f4f4ca6578c836935614931af2a3e79bf7d","subject":"Ensure resource_provider_bandwitdhs is unset by default","status":"MERGED","created":"2020-11-16 12:03:50.000000000","updated":"2020-11-23 17:30:44.000000000","submitted":"2020-11-23 17:30:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762827-rp-bw-default","meta_rev_id":"15e989ceb1a048bceb98913b441fd4e44aeba68e","_number":762827,"virtual_id_number":762827,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"date":"2020-11-23 12:09:56.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-23 17:30:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-11-20 05:55:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2020-11-23 12:16:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-11-17 21:39:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-11-20 05:55:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-17 21:39:53.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-11-20 05:55:51.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-11-20 09:17:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-23 12:16:29.000000000","updated_by":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"reviewer":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"state":"REVIEWER"}],"messages":[{"id":"bc973a79b9f03e585ad983416f87a062a48be889","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-16 12:03:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18e065957b3570e1f6f7eebf620f83fcb9b560ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 15:06:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/356a4dc7e37d489f9b56bd21ca785f9b : SUCCESS in 3m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc891a30b64144559bf64f73bc895279 : SUCCESS in 3m 06s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9b1661933d8b4d9f9263aed1e5e6876f : SUCCESS in 15m 25s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/058720a3dce94ff8b8b37a3f1a1b9216 : SUCCESS in 13m 34s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e14f7d1039c245ea89ce5f24506ec896 : FAILURE in 15m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/32f73253518148fd96fdac71830f9555 : SUCCESS in 1h 04m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a25524f09a274925aadd3a7163556d7b : SUCCESS in 59m 39s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98bf2d54c826474e981470fa9d196548 : SUCCESS in 51m 32s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ebea8058eb684dd6b1d5ba6b16c68a69 : SUCCESS in 40m 00s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9f9a661596ac456084b668e0439c8871 : SUCCESS in 1h 30m 37s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e5533160773848298d6af35fd0957c15 : SUCCESS in 1h 38m 26s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e4a5ceffedc44317948456d18523ed49 : SUCCESS in 1h 29m 19s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fd6b5acb57af42d69c0eafa009e6015a : SUCCESS in 54m 31s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/d6e2419b277d46fd9d6ef12de5b23864 : SUCCESS in 19m 47s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3bbc9e8d9d4a485db4a3d9a234b65323 : SUCCESS in 17m 28s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4b498436fb0447088c2a2e9ed921be3c : FAILURE in 2h 58m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9f2036685f9e4bd51ea2fe7f0072adbd7ad5ce83","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-17 07:51:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b5f28685f7825f57ea0f9826c9e7321c5a09a70a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 10:26:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c6391cf575e4f05afe92e4a45f9e172 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e6ebcd800d7437e94c9e5a67c6d9705 : SUCCESS in 4m 15s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/2f632bb2af84414fb8a0a189d845b2be : SUCCESS in 16m 44s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/024f616a6d6f44188737d542e6f07873 : SUCCESS in 14m 50s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35686b58fc104cc9a0d54bcc81be6658 : FAILURE in 16m 47s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2095b22a7df64f02aa1333208d9597c2 : SUCCESS in 56m 45s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c7cd41082d4746858c28ec9b1a1edbdd : SUCCESS in 52m 57s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef18280367074ed28d47ce5b54f129b1 : SUCCESS in 46m 43s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/fc21a70fad6d49e0b6d7656773d644d2 : SUCCESS in 41m 06s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/40a87d6e6d0a4f2f898c5d1a98bbc77a : SUCCESS in 1h 12m 04s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c7251cb73dcf4c8783396488de692146 : FAILURE in 1h 38m 44s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/72a9330b14c94cb3ae5c65b35653f74e : SUCCESS in 1h 12m 02s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/93b10a7128644d51ab6340c8a0b51ea5 : SUCCESS in 49m 48s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/cbd520cd21474c6494bce4404eb29c81 : SUCCESS in 23m 22s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb417e64811f4f9986eb5f4edb5cd973 : SUCCESS in 15m 20s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/302a4af7ce6d42459c8881507eb3658c : SUCCESS in 2h 33m 37s","accounts_in_message":[],"_revision_number":1},{"id":"6bc0f38c978e3df0649affe547047499e47dbf93","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-17 13:12:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5ab39d3af0e3a0210a768525f50af75b3c2b619a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 15:58:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0124b24e3f204c209fc6e25660f7824c : SUCCESS in 3m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1eb2807f3e649c2a075e18a43bf2ed6 : SUCCESS in 3m 18s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/4e246bba2ef744449bb668418a2e14a2 : SUCCESS in 17m 38s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5546dcac7ddd4ef38147d47341fc2140 : SUCCESS in 17m 00s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b481dd6aefa146bdae590fadc7e4030f : FAILURE in 19m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/3b0596e98c904c1982f0e951d44b5792 : SUCCESS in 55m 20s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd9a617987eb43a1a400bde9c69768a2 : SUCCESS in 57m 23s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2211ffedc91a4ba4b1ef42706005c081 : SUCCESS in 49m 21s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/fe46a32c842c4bb4a93218de02e1a8d1 : SUCCESS in 48m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4cbb104bc30542c59a19fb8ca652126d : SUCCESS in 1h 29m 47s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/17585ea99b89421d9f8ff778ad7cc64c : SUCCESS in 1h 34m 40s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0c3b9b9f2fa541008203482056b214c5 : SUCCESS in 58m 24s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/61b53399db574737aa9545d0712ab281 : SUCCESS in 53m 17s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/8feb144f51564d4496372f3fad36774e : SUCCESS in 20m 08s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8bf8497db034e2eae4305f84684fe19 : SUCCESS in 15m 23s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c87e718b7be74671998e27d17e5ca122 : SUCCESS in 2h 40m 16s","accounts_in_message":[],"_revision_number":1},{"id":"7a76edbefa20fe581523bb6fab058810d7e809f9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-17 21:39:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1565b20795c84869f97de78f6317dffeb60e1605","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-11-20 05:55:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4b7bb70dfd9aba89449608e8fe753e2c20bbcb2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 05:56:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1130e3f8d5a9bdc683c53074ecfb43ae8dd79cdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 09:17:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03db6721772f412e923b89ddf823a984 : SUCCESS in 4m 59s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56ccf014c5434a02a9681fac33b4f660 : SUCCESS in 2m 57s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/00f20036a80c45ae9137f638b96733f9 : SUCCESS in 19m 24s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5882c012398417c88f8d03b3a28b79e : SUCCESS in 16m 52s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/70ab3c8f436746109ac13d8f800274d8 : SUCCESS in 1h 48m 14s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f91cece575e1448b926ec0b8ac139c65 : TIMED_OUT in 2h 00m 51s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d988bcb1d23f4c46a194573385b06d0c : TIMED_OUT in 2h 00m 54s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9df39c936b084bdcab9610cca7f0e8dc : SUCCESS in 1h 20m 45s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/fc00cbc412774d84ac7c0241860be2a5 : SUCCESS in 23m 56s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2cd51b94e5374457ae1b7ec2cf3863d0 : SUCCESS in 15m 50s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ebad26fef2c74a3ca0232dddc6df7e3c : TIMED_OUT in 3h 15m 19s","accounts_in_message":[],"_revision_number":1},{"id":"87e1ad2f3fd832b12ddcceb1611a6c62bb6e94a4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-20 14:36:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7a6ae15e47133bc9d84686ce5d0d018a13dbfe4b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-23 03:41:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"75d1c1fbe2a57bca55b6c8c9e70106b438bae2b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 07:39:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b43f9f174634bea9b3cd94fde601d5b : SUCCESS in 4m 16s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c033ac49c4a24d8ca43c422c962a8020 : SUCCESS in 3m 32s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c206f21ee7b743c89d5875107f1e87e7 : SUCCESS in 17m 01s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9bdafa7ed1164becb090d48424efff9f : SUCCESS in 11m 50s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abd9a2b8c65f43cbac1c1956fa794687 : FAILURE in 15m 48s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab80e814f8724a1da6edd71488a78755 : SUCCESS in 47m 42s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45ac4618a1f8433697303367b9d4e7c4 : SUCCESS in 40m 11s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f0577d7391b4660b3433e02e4cebece : SUCCESS in 54m 37s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d3d81fe2d9c4af4a9af089f0f4face2 : SUCCESS in 1h 23m 36s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/10ea20bb12bb454999dd5143d3517141 : SUCCESS in 1h 10m 02s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d3c41c70c40d4c939b2a7e20b6fb3551 : SUCCESS in 1h 09m 38s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9e5d594279834f8db81e1b1682615037 : SUCCESS in 59m 04s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d84d38cd61e34e80adf5c1bba9d12bc8 : SUCCESS in 43m 54s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/68c47e4c3bb141f982b33063c6cd2b08 : SUCCESS in 24m 55s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2eb102f598c44ba1ac36a92baa77d453 : SUCCESS in 17m 48s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fec6233c82fc4df3af58e905019f209d : TIMED_OUT in 3h 56m 21s","accounts_in_message":[],"_revision_number":1},{"id":"34311b97053d1077b94c024f8d6602054f51281e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-23 12:09:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3b018f7d4258db686265f9a4e775f30143c6ae29","author":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"date":"2020-11-23 12:16:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0d23c9eb4ea9f7f6868eb4a40dd6e41a0349d429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 14:56:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/433b1a2327b840988cbfe954b8c586b7 : SUCCESS in 3m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4de82c02c7d04db5837b379be5189e83 : SUCCESS in 3m 19s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c0163342ae0f4262bbd1d3c142f3aab7 : SUCCESS in 18m 37s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec52b63d14f14cceb3e8e6734fbbc7ef : SUCCESS in 13m 32s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7af43bf2548643379c0881d5337eeff4 : FAILURE in 16m 46s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09c60652f28d4176ba3cf6a98156ee5c : SUCCESS in 52m 48s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/098c21fd83cd429897c0aa5609371306 : SUCCESS in 55m 59s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0100e46928554da49661283c7dc9c362 : SUCCESS in 49m 08s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87991f878d5246d3b381c9f6c59da869 : SUCCESS in 43m 21s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/41e9b13be4b5492b85add6e61894c55f : SUCCESS in 1h 23m 55s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a8285ff87e224a1d9f8045606191399a : SUCCESS in 1h 31m 57s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/97450c87ac4f4a80902e422fcbf6b0b3 : SUCCESS in 1h 33m 42s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/db30631f9c10408b8bf13499b2f23e8c : SUCCESS in 1h 00m 18s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/d883052374ed4fba9da75421fe59402d : SUCCESS in 18m 58s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09450426816e4ef19cc87723f7c86c29 : SUCCESS in 16m 57s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d60cc604ccac43c3a8e447e1daa117b7 : SUCCESS in 2h 43m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ea7bedc33a68c73fab12b4bb3600dae725c730bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 14:56:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b19a73320cc64d37cadfd2cb1577d42cedb1bcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 17:30:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2aadac01938147339c41a36b46101073 : SUCCESS in 3m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf817ce32319499b80e89e333bb181c4 : SUCCESS in 3m 15s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0db78bcfba264933b75808540b362833 : SUCCESS in 16m 24s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fbfd1b24e25b4c53a4fe7328530eab2b : SUCCESS in 19m 59s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b5a3e564dd9640b1abb5c8e5b8253eb0 : SUCCESS in 1h 33m 43s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/226e56bce4f04384b0eef04dae7b10c2 : SUCCESS in 1h 54m 24s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/408fec9ce67f45e2b5d5b9176dfd1648 : SUCCESS in 1h 31m 45s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a3983bd5bb454384ab97bf89e4251604 : SUCCESS in 56m 13s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/c43d1d4f065b4239a3759aef9db6d02b : SUCCESS in 19m 22s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0be765b761434e468f75fc990fc2889b : SUCCESS in 17m 16s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ec9036ce53c14dadbc3b72ab509e9c44 : SUCCESS in 2h 30m 39s","accounts_in_message":[],"_revision_number":1},{"id":"15e989ceb1a048bceb98913b441fd4e44aeba68e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 17:30:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b7bc9f4f93231a321adba920f9778e447ceb0cf","revisions":{"5b7bc9f4f93231a321adba920f9778e447ceb0cf":{"kind":"REWORK","_number":1,"created":"2020-11-16 12:03:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/762827/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/27/762827/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/27/762827/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/27/762827/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/27/762827/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/27/762827/1"}}},"commit":{"parents":[{"commit":"43a2ed4380a8ef0a2e6c64d4ad08193da6f75eca","subject":"Add support for the keystone_authtoken/service_type parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/43a2ed4380a8ef0a2e6c64d4ad08193da6f75eca"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-16 12:01:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-16 12:01:19.000000000","tz":540},"subject":"Ensure resource_provider_bandwitdhs is unset by default","message":"Ensure resource_provider_bandwitdhs is unset by default\n\nCurrently puppet-neutron supports the resource_provider_bandwidths\nparameter for ovs agent and sriov agent, but it doesn\u0027t provide the way\nto clear existing settings.\n\nThis change makes sure that the resource_provider_bandwiths parameter\nis unset by default so that operators can reset the parameter to\nthe default value defined in neutron service itself.\n\nChange-Id: Icd834f4f4ca6578c836935614931af2a3e79bf7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5b7bc9f4f93231a321adba920f9778e447ceb0cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5b7bc9f4f93231a321adba920f9778e447ceb0cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
