)]}'
{"id":"openstack%2Fpuppet-neutron~762969","triplet_id":"openstack%2Fpuppet-neutron~master~I3d9f0b48a72f82f7d1a86aae94c1f5248b729987","project":"openstack/puppet-neutron","branch":"master","topic":"rp-bw-default","hashtags":[],"change_id":"I3d9f0b48a72f82f7d1a86aae94c1f5248b729987","subject":"Add support for the resource_provider_hypervisors parameter","status":"MERGED","created":"2020-11-17 08:04:24.000000000","updated":"2020-12-22 17:41:16.000000000","submitted":"2020-12-22 17:40:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762969-rp-bw-default","meta_rev_id":"4d2a415b7a6e7dc99c4780de85ac72c19147c8b7","_number":762969,"virtual_id_number":762969,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-22 17:40:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-11-20 05:56:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-11-17 18:32:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2020-11-17 21:39:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-11-20 05:56:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-17 18:32:04.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2020-11-17 21:39:57.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-11-20 05:56:10.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-11-20 09:17:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a7041b3e8ad365f4387c3c8d65a4a64ad47a95da","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-17 08:04:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84c9644c1aa7bb03204c173c07c4d11269e32854","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-17 08:11:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7640118d01f5315511e3051d451d6237734950cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 10:31:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1846249e282842f18d3ded0c0a157838 : FAILURE in 4m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ffd0ddc862b143d3857fdf8993f418bb : SUCCESS in 4m 01s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/4e61e2f963804cfe8b4d5a6523548766 : FAILURE in 10m 23s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb2724c7da1b448c97cc40483bc45525 : FAILURE in 8m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6afb3408e6c74f5eaf7800ec167576bf : FAILURE in 10m 58s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/60fdff8222b24fd2a96f542f42600325 : SUCCESS in 53m 25s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ffb12c3f5a64cd48b68df156a279627 : SUCCESS in 1h 01m 36s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66ba5de541064fd48b3f1182d656e715 : SUCCESS in 50m 25s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/3196d36e0795426ea95a60ee54ea98cd : SUCCESS in 43m 05s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7696019b170b4e9ea827328ecf81aca0 : SUCCESS in 1h 28m 44s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e4eaf5cf9a0244939088655ab151a402 : SUCCESS in 1h 12m 58s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/16a91870490a4a1c8fde1e90e9299c20 : SUCCESS in 1h 13m 53s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9dd9f7c8d0ce44cab541abbc783ee815 : SUCCESS in 52m 23s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/ac12fcfd7a184190b946796cc94d8ebe : SUCCESS in 19m 50s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f3d0ec4da6a4e43bb2a9422dd22b8c2 : SUCCESS in 15m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40c1635068e9472d84bca067a21a1e73 : SUCCESS in 4m 28s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/879dbe549aaf4e0b9ffabed75d85125f : SUCCESS in 2h 17m 50s","accounts_in_message":[],"_revision_number":2},{"id":"78d8f09bb389838afff0783a927923e6d3532dae","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2020-11-17 12:53:39.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c44edec9672789c5cbe1a6dfb9a1d69c79bdca77","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-17 13:14:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"21bc50db22af6c09eb12d5d998a4fca7af138f1f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-17 13:16:20.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"dbe53b31a9dfe949b8adb076690d4795c623fa22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 17:17:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc17216cedc24bbabcd7d327b806130d : SUCCESS in 3m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0814b6984a3421eb258a2e137872714 : SUCCESS in 3m 23s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/fecf2cafc8b24641b23b5fb892d79f64 : SUCCESS in 25m 57s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55ee2fea29f24d13af6248e8b1fdce0d : SUCCESS in 14m 07s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e5192e08eb74d9f9d99a385741ffc59 : FAILURE in 17m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6b19461a5d6546058ce40b7efde175a3 : SUCCESS in 54m 50s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6233e94294e408b88a03dbbfec0ba98 : SUCCESS in 55m 49s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3de5ee269e34447e8b966c2a6f68de34 : SUCCESS in 52m 19s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f2ceb980e84d485f820d60031ef92f55 : SUCCESS in 40m 20s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bb5a2401ed774b26af9837af669e7578 : SUCCESS in 1h 25m 48s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bcd641efdff549dabf6794fb62950ef9 : SUCCESS in 1h 57m 09s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5666a8c892014a509bb477290a15c0aa : SUCCESS in 1h 27m 41s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6ce426e0e2f8439aa60ee4fb3606e099 : SUCCESS in 49m 35s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/22755f02c1504c1986f6ed2ba4ed66a5 : SUCCESS in 20m 39s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/187c3f1c2087433fba3e8fce3b0375c7 : SUCCESS in 16m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce3d5b863f734b50a8e92f99cc7f3bcd : SUCCESS in 4m 24s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/af60e3ccc06d4cf8a9282a40b336acb7 : TIMED_OUT in 3h 57m 37s","accounts_in_message":[],"_revision_number":3},{"id":"ce1913c9f19feea65e265e89f1101169fcc94ffa","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2020-11-17 18:32:04.000000000","message":"Patch Set 3: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"df3f3b2028afc282cf8dc4824711c491fce28b57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 20:54:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24d342209f3144c9a9838a50beb9c4f6 : SUCCESS in 3m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ada01576d5234f9cada4775025687de1 : SUCCESS in 3m 58s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/cc5cac6cf04b460588d1affc1b83e312 : SUCCESS in 19m 00s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d8ad80cc3c149d5acd69443a3af46df : SUCCESS in 15m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eeae5b66dc614826bc593ee99e6c2302 : FAILURE in 18m 13s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ef6a514dc3fa4075b68da5db3c01772a : FAILURE in 34m 05s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66670ae5dea04595a911a090e68410be : SUCCESS in 1h 03m 02s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/251c71ba024f416fa3dfd3aa1eb91c47 : SUCCESS in 1h 21m 53s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2f85375313c643579d0280eaa6a6d7df : SUCCESS in 47m 37s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4521e601247149579a5c13459c545f63 : SUCCESS in 1h 16m 42s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0fe48fb95a27484d991f4976640294f5 : SUCCESS in 1h 41m 24s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6ae558fb47a545dcb80425947516c343 : SUCCESS in 1h 43m 32s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/88c2e0c86bb34cfbb51b9ba46fe5473d : SUCCESS in 52m 25s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/3252bd95dbaa49a480168329b7f60f98 : SUCCESS in 19m 42s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ab67e33032b478e9e70785d63f51df3 : SUCCESS in 16m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0c7727266564c9da0103cbf28c6af2a : SUCCESS in 3m 39s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fc54957cd2d646198a464edf7963e6dc : SUCCESS in 2h 17m 40s","accounts_in_message":[],"_revision_number":3},{"id":"7c4bb550b673ac05043c7454953d527c6fdab75f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-17 21:39:57.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8a755f6177c95c778c2ef08dfe98d84e7a9171bb","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-11-20 05:56:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"61b342143e2101161c0e574b6fa148efa9771797","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 05:56:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2cd27d49f7840a0ad6a595a914fe8546f340afd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 09:17:31.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"eea84f980050b1c6e4548eb285bd6546736a6e78","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-20 14:36:37.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7ce4fdcde6e4366880456ec81f40d2ec71381cf5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-23 03:41:47.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2ddc25130ca185c111d83e000e94e4b4f0aec2a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 06:10:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3635d072eb8a47f781d5a57f857b8602 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff96021918f24125b741ad33a156db82 : SUCCESS in 2m 50s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/30ab211d944e4e09ac3abca490643238 : SUCCESS in 17m 55s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57e2b07a557d4ec58604a408fd46e698 : SUCCESS in 15m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10c7d73b72b94f14a7fce40aa8ffa3c9 : FAILURE in 16m 37s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34ba659dd0854e6ea1a8cbe444581a1a : SUCCESS in 55m 34s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a72e38e38e8d40cfb540b5b1f4d57c9b : SUCCESS in 54m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05d26b70d16c4c68ab021c84c4161eef : SUCCESS in 47m 10s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57bd54c274a6431994094bd6f923c4fb : SUCCESS in 38m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9c8bea3475cd4cd79ba02bdbb8e2f642 : SUCCESS in 1h 23m 14s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/de883fc286a94fa2905b380832523366 : SUCCESS in 1h 40m 38s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/aa2c9630fb8e49a0bac6db708c54b043 : SUCCESS in 1h 42m 46s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/42a82a5d4f344c68af823915f449cc8d : SUCCESS in 56m 17s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/790e79fc88b24e2581c68413ecc1afb4 : SUCCESS in 20m 01s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ae70af3581548e792684b98db484aaa : FAILURE in 15m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d0b323344044a5daff6b962ccd9716c : SUCCESS in 3m 16s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/47190708f32941db93d3458e2b3e24af : SUCCESS in 2h 27m 59s","accounts_in_message":[],"_revision_number":3},{"id":"5da2865a3ec4945ea824a71dad029912862c5579","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 14:56:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5b08429f581e09edaaf9cd2fdcc7237db45409cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 17:33:15.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f911bd4d9f84fa98ab7ae2978c87da6 : SUCCESS in 3m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a58dfa565dbe403aa64117ccad80538e : SUCCESS in 3m 36s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/deb0e2426038428b95c0fe4404168b77 : SUCCESS in 20m 49s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f77c72aefd242e6bd60f853eb0d44b5 : SUCCESS in 14m 45s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fb7ca2bcdeb940e594fc487ee6297bba : SUCCESS in 1h 25m 59s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cd8d7408b9594406af70ad8f0e12e348 : TIMED_OUT in 2h 00m 54s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1aad6d8dc5f348e0b0ca7bae3cc784de : SUCCESS in 1h 28m 43s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ffa9965daa9c4e7ebeb9bafbab603c78 : SUCCESS in 56m 10s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/ec0c3a9f0c92497d963569ea5306094f : SUCCESS in 20m 04s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb58843a344e415fbb1d86e59d8002f7 : SUCCESS in 18m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbaa44ab71574a8590393c77766891af : SUCCESS in 4m 20s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/dd26f2deb6814af98083c5a1e08f998d : SUCCESS in 2h 31m 47s","accounts_in_message":[],"_revision_number":3},{"id":"3de3fd8a83730f6d160af42ef3eee5f17370f5ed","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-24 00:48:56.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6b94886ba7532dec58d91dd61b3939dd4e78cfe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 04:17:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7bcbd1e181954a20b7947bd5ec8daf62 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb4bae3ab0d84ca18e02966c1baa8a70 : SUCCESS in 2m 55s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1696e4e0b122411fa8b4eefb39309e35 : SUCCESS in 18m 28s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7babd7058bd464b9bef26e15c527ffd : SUCCESS in 16m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16d1774c436f4cdbaba0a38fe05ead60 : FAILURE in 18m 04s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55b35fbd8dbc4185934d3b55e8ae3dac : SUCCESS in 1h 01m 05s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39ae7d2217704ec1917aad35d02a46b5 : SUCCESS in 58m 06s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e32eac144cf463c9513617dbf49c2bf : SUCCESS in 48m 59s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc083049ee914a1c84aa4ebe5ad94e76 : SUCCESS in 44m 05s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/70e0e7fd08244c00af1578b480809402 : SUCCESS in 1h 35m 02s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/67c01465533d40848c7bcec60606e336 : SUCCESS in 1h 48m 59s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f9f11891a5d04bab89de9a5d653da165 : SUCCESS in 1h 45m 48s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b1baf652671244039d23ff90acb93ce8 : SUCCESS in 1h 03m 29s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/951b77f327c24219ae680673f0a10f54 : SUCCESS in 28m 21s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b4641abb4ec4dacb6a5b9ef987572a8 : SUCCESS in 16m 00s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e270fee92914353bd1c7c8c35bdd468 : SUCCESS in 3m 10s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d9b9ec0f611343dba9001d806caab6ae : TIMED_OUT in 3h 27m 03s","accounts_in_message":[],"_revision_number":3},{"id":"864afe41388fe7fdc33337bc31d2108d6fe28a4d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-24 15:28:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3dede2bac0f965e68136ca64509e84dcf1bda499","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 18:17:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3164cce7fc6f4605a22544d7baf5c2bf : SUCCESS in 3m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9333e51b78494a49bbac9844d020f5f1 : SUCCESS in 3m 02s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/77b9e274395f4e8ba72ca191f703c218 : SUCCESS in 13m 32s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88dac7558ad149dba14a7edd1a4b5f8a : SUCCESS in 13m 23s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5fcd202198d24954be552155442c0cce : FAILURE in 19m 30s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/79f898bdcc0b4883aca534c1ab1cd0e5 : SUCCESS in 52m 03s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/294bff7ec75344949c6eb51eb4d5363e : SUCCESS in 50m 37s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5422b30d1d9428797aacbce78750deb : SUCCESS in 41m 03s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/114fb53a82be4378abc186c727674896 : SUCCESS in 46m 46s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/00a5ca2502e74321918c42cb7e9a8798 : SUCCESS in 1h 17m 04s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/df4e46cd2eb242bcbadbf7ce6bc98f42 : SUCCESS in 1h 54m 53s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/915422c5f980435c99babf226561bfec : SUCCESS in 1h 27m 36s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8cbeeccde930439b985d183ac0bab3f1 : SUCCESS in 59m 36s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/d6ca82e5edc346e48adace37b88ec7a3 : SUCCESS in 21m 23s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f60e569678846f9add42f591c19da41 : SUCCESS in 15m 00s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c15834a83d3340d4b056e4f2a9f3faab : SUCCESS in 7m 06s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a31383afd26245bea0ec99d2c8909cc2 : SUCCESS in 2h 34m 25s","accounts_in_message":[],"_revision_number":3},{"id":"43c956cffac9f362c0061cca9e8c4bc884a80880","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 18:17:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"17812a60b9a3562cbacaa4dce9da9be39e3b4576","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 21:09:12.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4bbf230b1ae4b14a188db0e6719cea0 : SUCCESS in 4m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f502ae136f0341358390befbb511bd69 : SUCCESS in 4m 00s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/07eea1056b114590852ea2d7dfaefbb0 : SUCCESS in 15m 26s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5b3dc872be2479c91b5378345b6ac83 : SUCCESS in 14m 27s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d6c7acc4cd2546af9793c238418197e9 : SUCCESS in 1h 12m 53s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f5f38dfbd15049768e98eb0cb744b5d9 : FAILURE in 1h 36m 37s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5fd8c269aee04c24ae41d9dce8b0a121 : SUCCESS in 1h 39m 02s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/be24184d37fd41d5916fb41bd61983ba : SUCCESS in 59m 31s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/c700417f6fbf4c75b276a960ed556b6f : SUCCESS in 20m 08s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/456e3f3b97c941808104348128fc5058 : FAILURE in 14m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe5c863ca4fc46d2b96a53ff38cf785b : SUCCESS in 3m 17s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fad6b07dd8574736ba0e6449d863986d : SUCCESS in 2h 47m 54s","accounts_in_message":[],"_revision_number":3},{"id":"b628997fc15998dc3c1b550703f12220207fb19c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-25 08:01:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"34d05c8c61568cbb7132c42f8890456cf9f17d0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 11:28:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d19d56c167a043e082eb8affd24dcaba : SUCCESS in 3m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/772588434ed34930ac1646a300da76b1 : SUCCESS in 4m 58s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9099bc31b4a4456bb3f91ca17e42cefc : SUCCESS in 22m 32s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e3c4b1cf644417481109335c9e25ad1 : SUCCESS in 17m 25s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8b18012716b40ae96fc988225e82977 : FAILURE in 18m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ded86f199e5c403d88a635609fa9a00f : SUCCESS in 1h 03m 33s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19131adf173648bb9b58b82b2535592a : SUCCESS in 1h 17m 10s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5ca6a0fea3e4ea9bc2289124964f922 : SUCCESS in 49m 34s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03c823a6665f45c69968b4cb5d0d0ad4 : SUCCESS in 45m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/94b7939afa934be185a09e9ac87ba90e : SUCCESS in 1h 36m 31s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0d6c50d7945a4f20955cc0acdb0a973b : SUCCESS in 1h 46m 34s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/67ee5b4549af4bc58abceeed46e7a82d : SUCCESS in 1h 15m 10s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/846c95b784884f948940ebca1145505e : SUCCESS in 1h 00m 59s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/4c573e436f434a48b5573af267373b97 : SUCCESS in 26m 19s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f62b7a9943a4a929afed3cc08f1be99 : SUCCESS in 18m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c38c044814db45a29c397532e9d53ba0 : SUCCESS in 6m 12s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e46e622c59964b4b98a619fbdfbb74a5 : TIMED_OUT in 3h 18m 56s","accounts_in_message":[],"_revision_number":3},{"id":"ae0ffa38042555713ed8fbb15abc769ec127e4a5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-30 15:40:03.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"109f3247cb3454b2d9299b80cde381b8de31588a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 17:34:30.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9746690f6378469fb52ffc431d9e385e : SUCCESS in 3m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a27a14248b744290a3a84e05f4fe14b8 : SUCCESS in 3m 18s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/b038839c236e482f9975986e800bebd6 : SUCCESS in 18m 31s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bdd4130d09fb45f3a2a3433329088dc7 : SUCCESS in 17m 51s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ae22d2409624f53bd62c71eafeb5e6e : FAILURE in 18m 46s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3cf7df83b394486823e265fb01bf60f : SUCCESS in 50m 16s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a24e029a12d94932b381adb0f3eb9154 : SUCCESS in 58m 24s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39e3c94e3fdc4ae094e42d453425fde1 : SUCCESS in 52m 43s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fec1871afe05422587ae510d87e3d9c1 : SUCCESS in 46m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b77e06b124f444eda4c7ebde55b20957 : SUCCESS in 1h 31m 46s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e41c66b1b58f4a0c9aa8b0e33a8b716c : SUCCESS in 1h 26m 29s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cfd8a5b3d91e4ae88b2d2fc96db0533f : SUCCESS in 1h 21m 13s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ea290b53bf17437bbab99dc14209fe22 : SUCCESS in 51m 16s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/58a9a8305f9b4c9ea7438a5419618488 : SUCCESS in 18m 38s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5cb30eec6c604c77b4f7450464fd1616 : SUCCESS in 17m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44b166507b594fe69bd363c5ea7f0c11 : SUCCESS in 5m 01s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c2a91e4ae5084269bb0e34c5430a9163 : FAILURE in 8m 47s","accounts_in_message":[],"_revision_number":3},{"id":"7e182e45caa2ae1234a9d609b6927617cd786154","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-15 10:48:16.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4ca3bced7f51fc4ec3b602a81a8f1033d1777767","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 13:54:40.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34b144e50f164fa79ad9d9d72f9ce5c2 : SUCCESS in 7m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bcbbbf9ca669427688fabcbf1d9cbe74 : SUCCESS in 6m 56s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/eb155e0a75a3494bb359414dc1ec921e : SUCCESS in 27m 58s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cea1b4d40d954eb2984d264a8e765bf4 : SUCCESS in 21m 51s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1756abb80504f0f91ca7d9b18bd6358 : FAILURE in 18m 50s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb0753b126bf4593827fe1febb9a2af9 : SUCCESS in 1h 19m 51s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9567d588dbc04f39a9e20f3b6638ec74 : SUCCESS in 53m 53s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2104173ad434c3cbfa85265aaf0e1e4 : SUCCESS in 49m 50s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/052c70c7536b4498898a35e02ecb3f9a : SUCCESS in 1h 03m 07s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/81060f28d49c4afc8567c6cda7d70ce3 : FAILURE in 33m 16s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/438a1d9c4da54866a8ba61dc9a5e0652 : FAILURE in 34m 29s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/deb35ac23ac34278b8f5ec8e8cf77d8a : SUCCESS in 1h 47m 36s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5a9453153bf647eb9db656d3814af84a : SUCCESS in 1h 21m 56s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/476feb7c8b8e484a84891c564ff14b22 : FAILURE in 28m 24s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93db997a0ecc49008dc837bcba71154d : SUCCESS in 22m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0025b7a681141b5be1f7b78544fae7f : SUCCESS in 4m 04s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/120198abbac24643a03a6efcc7374693 : SUCCESS in 2h 55m 11s","accounts_in_message":[],"_revision_number":3},{"id":"26c847c1c7bba005535348ae38a398d58f0c9f0d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-22 12:26:11.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4177d9daff8ea363d08cab6c95467689da373b88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 14:52:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e6e3943ca814253bcb63db8e7f6e6f1 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32f700e2546b4651abb2d41f38488240 : SUCCESS in 3m 58s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/73b732e7f139406e87375f719557a166 : SUCCESS in 17m 37s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b96a21a179d74a488fca7836166f6167 : SUCCESS in 17m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f2eb3ed9ddf4359969c47f10f4867ca : FAILURE in 18m 10s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09e7bca22a7b465e91a61eac741c8a3e : SUCCESS in 56m 11s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff0d4f2d20114665a67fab497d63bd58 : SUCCESS in 53m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39e71f05797946fba6a2e221bc80e418 : SUCCESS in 39m 52s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/967443daaf2f489e9be55c6d4b83c1b8 : SUCCESS in 37m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9af25b68f9a04e22a022e35cd32d13d1 : SUCCESS in 1h 03m 52s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5c7f8e9162674b1ca1fc7614e9825021 : SUCCESS in 1h 32m 41s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d679eac55d164922880782c6386a6774 : SUCCESS in 1h 25m 35s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c8aaeb94940d4bc3b5844e34089e7536 : SUCCESS in 1h 07m 06s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/d849bb9840c443c3866b6ad431430a20 : SUCCESS in 18m 42s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72ef12f5ff1c4a00b41fadeb4a541334 : SUCCESS in 14m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14d4b15b956d4275aed9684d9812021f : SUCCESS in 4m 33s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fdb2da110f124b84a686c935eaebc9d5 : SUCCESS in 2h 21m 40s","accounts_in_message":[],"_revision_number":3},{"id":"f2d7ae5849d157337edbd9b65911e3cd24195ed2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 14:52:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9279a3cd13b10b356f3f1a86b08e10953a276ef6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 17:40:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85ffbd1e26f34967a8bb48b9aab47db4 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc9f84d706854366a4e3b07af01050e7 : SUCCESS in 2m 57s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/188ff58b3a5f472dac8209b898c0d8fc : SUCCESS in 18m 28s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30cf9ba91d5f4ebeb3f1041ee43df933 : SUCCESS in 16m 14s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/89fc099af2a442668ae5b381d91f4aa3 : SUCCESS in 1h 40m 29s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/43baab50146e405397df3f06f5afcf8a : SUCCESS in 1h 57m 28s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/54122abb79b247f6b0d06bc3a6d492c2 : SUCCESS in 1h 22m 02s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9e5943d689b74c54b144385019734b25 : SUCCESS in 1h 05m 40s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/50d42c1c79154b4c8afa513459e48f4b : SUCCESS in 17m 50s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90832259a56c4cb6b7364792c31a9fa8 : SUCCESS in 14m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/121c7a9eabb746a8940ac3327161c8b2 : SUCCESS in 3m 01s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/eabc568309364d628a26dc1be40905ec : SUCCESS in 2h 42m 18s","accounts_in_message":[],"_revision_number":3},{"id":"0db007d36f7d4cffa072865fe4e671cdf8db65bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 17:40:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"4d2a415b7a6e7dc99c4780de85ac72c19147c8b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 17:41:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/878cbb3d083b4b0f95023abbb9d88837 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"830c66cccaf1446ff59081630977033dbdc72b6a","revisions":{"6540d96edc49043ac0b17eead13d214a2f17f93a":{"kind":"REWORK","_number":1,"created":"2020-11-17 08:04:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/762969/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/69/762969/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/762969/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/762969/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/762969/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/69/762969/1"}}},"commit":{"parents":[{"commit":"5b7bc9f4f93231a321adba920f9778e447ceb0cf","subject":"Ensure resource_provider_bandwitdhs is unset by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5b7bc9f4f93231a321adba920f9778e447ceb0cf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-17 08:04:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-17 08:04:02.000000000","tz":540},"subject":"Add support for the resource_provider_hypervisors parameter","message":"Add support for the resource_provider_hypervisors parameter\n\nChange-Id: I3d9f0b48a72f82f7d1a86aae94c1f5248b729987\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6540d96edc49043ac0b17eead13d214a2f17f93a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6540d96edc49043ac0b17eead13d214a2f17f93a"}]},"branch":"refs/heads/master"},"bce9c109f518002168302448f237e4f4aeff0bc9":{"kind":"REWORK","_number":2,"created":"2020-11-17 08:11:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/762969/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/69/762969/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/762969/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/762969/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/762969/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/69/762969/2"}}},"commit":{"parents":[{"commit":"5b7bc9f4f93231a321adba920f9778e447ceb0cf","subject":"Ensure resource_provider_bandwitdhs is unset by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5b7bc9f4f93231a321adba920f9778e447ceb0cf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-17 08:04:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-17 08:11:05.000000000","tz":540},"subject":"Add support for the resource_provider_hypervisors parameter","message":"Add support for the resource_provider_hypervisors parameter\n\nChange-Id: I3d9f0b48a72f82f7d1a86aae94c1f5248b729987\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bce9c109f518002168302448f237e4f4aeff0bc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bce9c109f518002168302448f237e4f4aeff0bc9"}]},"branch":"refs/heads/master"},"830c66cccaf1446ff59081630977033dbdc72b6a":{"kind":"REWORK","_number":3,"created":"2020-11-17 13:14:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/762969/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/69/762969/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/762969/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/762969/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/762969/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/69/762969/3"}}},"commit":{"parents":[{"commit":"5b7bc9f4f93231a321adba920f9778e447ceb0cf","subject":"Ensure resource_provider_bandwitdhs is unset by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5b7bc9f4f93231a321adba920f9778e447ceb0cf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-17 08:04:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-17 13:13:34.000000000","tz":540},"subject":"Add support for the resource_provider_hypervisors parameter","message":"Add support for the resource_provider_hypervisors parameter\n\nChange-Id: I3d9f0b48a72f82f7d1a86aae94c1f5248b729987\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/830c66cccaf1446ff59081630977033dbdc72b6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/830c66cccaf1446ff59081630977033dbdc72b6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
