)]}'
{"id":"openstack%2Fpuppet-neutron~765805","triplet_id":"openstack%2Fpuppet-neutron~master~I87dba40544a76991f5fa16705ab1af00f75e5dc9","project":"openstack/puppet-neutron","branch":"master","topic":"agent_config","hashtags":[],"change_id":"I87dba40544a76991f5fa16705ab1af00f75e5dc9","subject":"Support arbitrary configurations for ovs agent and sriov agent","status":"MERGED","created":"2020-12-07 13:20:42.000000000","updated":"2021-01-06 03:50:13.000000000","submitted":"2021-01-06 03:48:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765805-agent_config","meta_rev_id":"359616cac32409bbc5bd7e2a37747caa7532bea1","_number":765805,"virtual_id_number":765805,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-06 03:48:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"all":[{"value":1,"date":"2021-01-04 16:43:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2021-01-05 15:35:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-01-05 15:50:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-01-05 15:50:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-07 16:25:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-04 16:43:33.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2021-01-05 15:35:54.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-01-05 15:50:10.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"f067f3416487a08aba5651ed5aaadc1ac50123d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-07 13:20:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdf95319c9113446a3fbc305baee7fb0f1c01368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 16:25:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51260bb181e1440c9dff005cae3ec88f : SUCCESS in 2m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5271ac6d1c354004874f452a8dfb4fce : SUCCESS in 2m 43s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/5f8307b5c6434e4ea5a87cf80d55c9f6 : FAILURE in 19m 10s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f16f25fc94ea4642b2e849e58ef7de9b : FAILURE in 19m 00s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4bb9880f11ef4e1497c97232ce3a7938 : FAILURE in 19m 06s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24f060b7445e43cbbf108d76d2f46f7b : SUCCESS in 58m 08s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4471aa19f0345af8b5aff8eca7b58cb : SUCCESS in 53m 26s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48e0b5c8c55a427c93ef957cfde31b74 : SUCCESS in 59m 15s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9db79fe3ac4e46acbe1a083f99c97702 : SUCCESS in 40m 21s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/00041a71733c416d937a28ca2d7a5663 : SUCCESS in 1h 21m 07s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a71479c044514be2927e0ed1f0537717 : SUCCESS in 1h 36m 56s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cf45b23280594289a750aa75a271a18c : SUCCESS in 1h 25m 34s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/691ec1d449a24c578c1be7144cd1a551 : SUCCESS in 1h 14m 00s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/1fddc92942b34a0580a5ff3e2235a0f6 : SUCCESS in 21m 10s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/325d1a8bb340433bafe982e220e1cf42 : SUCCESS in 13m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4a125ab85e74da1b7271973cf0bc3d1 : SUCCESS in 2m 55s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8ba71543be1b42afb2a3758089e35070 : SUCCESS in 2h 57m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3315d1958f9881abc33c98afa2c03d8a1a4ae0e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-21 12:13:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"78741b76f8ac01786e96e287172e4a452f1cc5ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 15:01:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d125152a3084fa38bf857fa19d994d7 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6fbfa98e9694ad38bf0c5c0e67833cb : SUCCESS in 3m 26s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9d5bbaa6219543f1b5d2ed583489fe04 : SUCCESS in 19m 27s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96de3a7463bb426999baf6bd2ced02c6 : SUCCESS in 15m 29s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b18ce0367a354fca8b724c0fda5c2a04 : FAILURE in 18m 27s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1da88ef7981445688d22fffa2d53686d : SUCCESS in 49m 07s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9371e0b36cc344a9903bd055fe77fb5b : SUCCESS in 1h 03m 52s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10104612d9904102b0182c6a56a1cf67 : SUCCESS in 46m 51s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d076d20844e4599a9039efcafa0154b : SUCCESS in 44m 42s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e258fabdf0d840bd8c0d94925b7af60d : SUCCESS in 1h 32m 51s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a9d719db4ea94b818f250a9919f6ae7a : SUCCESS in 1h 31m 47s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/102dac9e41e64ca8b2d3e59ecfea9b99 : SUCCESS in 1h 29m 43s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4ce1cdc1556d40b9988e10cba6fb8e6b : SUCCESS in 1h 01m 26s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/1af7d19898b9423a87740e8741fd6da9 : SUCCESS in 19m 42s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efe3c4dd616f440bbd685914c51d1795 : SUCCESS in 18m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9aee2930b9c14095a53f6bb5060939ff : SUCCESS in 3m 46s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/73b30996af6145c9915afa00ee31f45a : SUCCESS in 2h 35m 56s","accounts_in_message":[],"_revision_number":2},{"id":"345861613566dffe174d3cdd1db2bcce2f40d049","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2021-01-04 16:43:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"79db78a3182ed3d82a0da8ad2fab04f4c3a55618","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-05 15:35:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"373f5ec73fb6a2b26163747655c4864da8b40590","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-01-05 15:50:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"df880f02c8f692760463b1765b55583d99fac5e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 15:55:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4e37fa56eb0ed85c8896b2e1b5aba0aa27f98b00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 03:48:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2526dd18a68d4e64a1b3e58c07879202 : SUCCESS in 3m 59s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5b9dfbd8f0848ee9cb0ba36ec3b98ce : SUCCESS in 2m 54s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/736e4059421f484f98016068628332ff : SUCCESS in 17m 20s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5072c785dfa04e2abeca0d58602b3135 : SUCCESS in 11m 23s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1003ba883801493da381307111afdf17 : SUCCESS in 1h 19m 08s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/befbb79140a045f0b0d3a26db5ba52b1 : SUCCESS in 1h 05m 34s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9e881b48d2274b6bb6cb80f3501755e7 : SUCCESS in 1h 31m 35s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/49a48ffa95224d489a8020fdcf6c84c2 : SUCCESS in 1h 07m 06s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/7558fffa3e544e8188ab1a916e978ca5 : SUCCESS in 20m 14s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6c8b22a0c0341889fc3a1ee2e1162f0 : SUCCESS in 12m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5c2861449de49ce8b0febfd67655062 : SUCCESS in 3m 30s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f3396e81dd85496da2661e59ccabecf5 : SUCCESS in 2h 32m 07s","accounts_in_message":[],"_revision_number":2},{"id":"41ae2b47886aacdabd7af289d67e26e5f9b72431","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 03:48:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"359616cac32409bbc5bd7e2a37747caa7532bea1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 03:50:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4ae2c88aa7a42139c911f3011a147c4 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"34eef40377f32d44838ea1a3d7471f07c0fe6528","revisions":{"f0b5d3603ee0abbad3e9239d7a59554ba36c3133":{"kind":"REWORK","_number":1,"created":"2020-12-07 13:20:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/765805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/05/765805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/05/765805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/05/765805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/05/765805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/05/765805/1"}}},"commit":{"parents":[{"commit":"40767b70b75a038df037960537fee02ac489384e","subject":"Prepare Wallaby M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/40767b70b75a038df037960537fee02ac489384e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-07 13:19:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-07 13:19:44.000000000","tz":540},"subject":"Support arbitrary configurations for ovs agent and sriov agent","message":"Support arbitrary configurations for ovs agent and sriov agent\n\nThis patch introduces new parameters to the neutron::config class,\nto allow us to add arbitary configurations into openvswitch_agent.ini\nand sriov_agent.ini .\n\nChange-Id: I87dba40544a76991f5fa16705ab1af00f75e5dc9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f0b5d3603ee0abbad3e9239d7a59554ba36c3133"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f0b5d3603ee0abbad3e9239d7a59554ba36c3133"}]},"branch":"refs/heads/master"},"34eef40377f32d44838ea1a3d7471f07c0fe6528":{"kind":"REWORK","_number":2,"created":"2020-12-21 12:13:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/765805/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/05/765805/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/05/765805/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/05/765805/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/05/765805/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/05/765805/2"}}},"commit":{"parents":[{"commit":"40767b70b75a038df037960537fee02ac489384e","subject":"Prepare Wallaby M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/40767b70b75a038df037960537fee02ac489384e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-07 13:19:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-21 12:12:53.000000000","tz":540},"subject":"Support arbitrary configurations for ovs agent and sriov agent","message":"Support arbitrary configurations for ovs agent and sriov agent\n\nThis patch introduces new parameters to the neutron::config class,\nto allow us to add arbitary configurations into openvswitch_agent.ini\nand sriov_agent.ini .\n\nChange-Id: I87dba40544a76991f5fa16705ab1af00f75e5dc9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/34eef40377f32d44838ea1a3d7471f07c0fe6528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/34eef40377f32d44838ea1a3d7471f07c0fe6528"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
