)]}'
{"id":"openstack%2Fpuppet-neutron~768438","triplet_id":"openstack%2Fpuppet-neutron~master~Ib91b8215ff8f07d7e81382cd197d583b6657cc42","project":"openstack/puppet-neutron","branch":"master","topic":"quota","hashtags":[],"change_id":"Ib91b8215ff8f07d7e81382cd197d583b6657cc42","subject":"Deprecate parameters for LBaaS quotas","status":"MERGED","created":"2020-12-24 07:32:06.000000000","updated":"2021-01-05 04:51:11.000000000","submitted":"2021-01-05 04:50:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768438-quota","meta_rev_id":"72fcb1bb5ad11748a47a84bf6274c8905a42cbf0","_number":768438,"virtual_id_number":768438,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-05 04:49:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2020-12-27 18:45:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-01-04 14:42:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-01-04 14:42:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-24 10:13:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-27 18:45:46.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-01-04 14:42:09.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-01-04 20:30:34.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"CC"}],"messages":[{"id":"a6aa885d50fda976dc1a29197cca60681b0b2f43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-24 07:32:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02d5013cfdf9de3902bcb991943abcd650f389d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 10:13:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89a4866f40534373b843562d43c9686a : SUCCESS in 3m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b01a7e2011c44aab5d536ae3efe9098 : SUCCESS in 3m 12s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/6c7edf175f5d4fc8a8ba51b672aed68a : SUCCESS in 14m 00s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50b6ba35708540b2acc0ee90aec200c2 : SUCCESS in 11m 21s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f683d4b3964a4ceca6c6fb08782c2c57 : FAILURE in 13m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9329cd2d6c2444f2b0e458409d44a3bc : SUCCESS in 1h 05m 34s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/963be2f4c6f74d81a209765d2b05bc74 : SUCCESS in 57m 37s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c06f973e67a5400c9106d863b72c9675 : SUCCESS in 56m 37s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8378d6f44f2d4230a9c99a47129788b4 : SUCCESS in 46m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/52aed26c428d4032aedc332daf2cf33f : SUCCESS in 1h 09m 11s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/356adf40079f4ecb851a8e25f81d97f4 : SUCCESS in 1h 33m 01s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a79c22e2ae5e44a5b91ca5a61d278542 : SUCCESS in 1h 28m 35s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f664318435694e3eaba8b82a3a27c35f : SUCCESS in 1h 04m 16s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/ddec11c765cb408b945e747b4f22bbff : SUCCESS in 24m 36s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f82751b6dc54f2cb5b30775c6516c29 : SUCCESS in 18m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/056c83670dba45f3b723bda585c62e21 : SUCCESS in 4m 48s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5b832b100dec478ca6a6a7eaad65c6db : SUCCESS in 2h 38m 36s","accounts_in_message":[],"_revision_number":1},{"id":"3782f0232d050ea57b9ef8519695a513c6364cb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-25 00:39:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"af9130f6e1de444e8f4abe5bb58650706f7120d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 03:10:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80b6fdee740242edb80159d14d406a6f : SUCCESS in 3m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49df45dd41204419ba03d219ba97bd2b : SUCCESS in 3m 04s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/12b642e721ea4d85a90d8ccd7670d5e9 : SUCCESS in 17m 34s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39577484b50e46dbba98bf29ca42bfba : SUCCESS in 12m 34s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4623c990f4d14fe0bbc70b2fb4325123 : FAILURE in 14m 04s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/363633682ecb43c8bafd5300bce01a43 : SUCCESS in 43m 40s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88adb75c4c254ae29f99d514672df452 : SUCCESS in 59m 44s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dcd6d015956a457594e4b2d08aec812c : SUCCESS in 39m 24s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6ce274ccd554e2a92801c886e1d022d : SUCCESS in 48m 03s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/98e38d4515d14dfba5d097ccf944f4a5 : SUCCESS in 1h 01m 33s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fea1608f92fb49e38477ced6b951e2cf : SUCCESS in 1h 02m 10s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8f5efae821db4bee8abe456e86f41d4a : SUCCESS in 1h 00m 55s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/42d39480ac93469ea55a160d3ec1e094 : SUCCESS in 46m 42s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/8162045033234747bda85a387b6affc8 : SUCCESS in 18m 25s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c9b50a9076148588fa05eb3317debfc : SUCCESS in 15m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62cc0ea547854be0b479932e88b1b351 : SUCCESS in 3m 08s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8c64a4626441453cadc6527964deb026 : SUCCESS in 2h 27m 34s","accounts_in_message":[],"_revision_number":2},{"id":"9917c182693b48053a39e18ec677761376ee5162","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-12-27 18:45:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e44cc5676f6b71738c711192ff1c9dae539ffb9c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-04 14:42:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"20ab8de2e326711beec9add6e83dadb2a9059cc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 14:42:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ecc398c5677a4e3a5c08cfc099a592bf187e98cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 17:41:24.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6311dea301d459b905e612991c7f1fa : SUCCESS in 3m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c97553cb59140e0ab369680a1f0ca85 : SUCCESS in 3m 21s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/b1dbb3f0a9974dd2995eb51d753bb3ae : SUCCESS in 19m 17s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe0f26194e9749d7a4b6f12133253aaa : SUCCESS in 17m 06s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5bc58f32182e4925aef56176ddaca9b4 : SUCCESS in 1h 25m 34s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1f068e09dbeb4c5dbb6a21577d2f2aca : SUCCESS in 1h 38m 54s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b7c6afefbdca46d79bb57231d3f8849a : SUCCESS in 1h 30m 33s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3018d018afa342fbbb4aeea3c58736c2 : SUCCESS in 1h 07m 05s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/816c3ba01e1548769a3aaa4149407c4e : SUCCESS in 18m 34s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f06704987ed64b33b33f3508ada2091c : SUCCESS in 15m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef25acc6d9de4480a7d62d401c4ddb5c : SUCCESS in 4m 36s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/01d27deb573d4bd6ad23b6a4fb4d218f : FAILURE in 2h 52m 56s","accounts_in_message":[],"_revision_number":2},{"id":"722bac64d234d5cf1cc6277a55b2395ebb66b87e","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-01-04 20:30:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"429bcf58ba2dd4ac05ad259b834164231e7466ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 23:19:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e26eeb7b61b475085b7b477c8cc6b99 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f1d2ccb08fc494b84c1f10b021214e1 : SUCCESS in 3m 15s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/82f59676370e4654ba144281cdd7afce : SUCCESS in 18m 35s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f165a13beb046ef9c769ad2e3c0892b : SUCCESS in 17m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/922d38f4b6cb4689b94605e955ac534c : FAILURE in 12m 37s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d57827b12254146b567e9c68bfe0aea : SUCCESS in 44m 20s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8924e536a9149bc8bf3854e5607def1 : SUCCESS in 53m 04s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/68b54b3f412c4b0faee5cd1b822208ad : SUCCESS in 50m 59s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92da8f57dc654f7096e5fa4e80ef3942 : SUCCESS in 36m 18s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/afae44a1c5ee45dc9fff062f8c4f6a55 : SUCCESS in 1h 35m 14s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3107962586704ceab6d3f851cd64731d : SUCCESS in 1h 35m 54s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/528a405861e14054b522c492c62c8e47 : SUCCESS in 1h 30m 03s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7b3860c5d3524b2d8a66cfbc3cef000e : FAILURE in 1h 20m 15s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/5a16985e1cc74cad9df0c2d489a895e8 : SUCCESS in 18m 09s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/568d550ef4f34c6a966bbdbaf1274099 : SUCCESS in 14m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fb328f18c2741cfb1d656e60df18458 : SUCCESS in 2m 56s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f9ccfcd7be914a6aa05ba8b54423430d : SUCCESS in 2h 43m 42s","accounts_in_message":[],"_revision_number":2},{"id":"4b1b680d9c5980e36415988c38c91e3ddab6f1eb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-04 23:21:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f2051d852fb29ae5ced65b1d37605fc9002e05db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 02:17:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe592300551c4fb18d2db47693fbc6d5 : SUCCESS in 2m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a78bdf64ff8b4970aeb13050af87fed3 : SUCCESS in 2m 50s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/ede94fb5bdda421db13066445586e60b : SUCCESS in 21m 16s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/511750c0529a43cc8b68624727132a2e : SUCCESS in 15m 02s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d8a6a79001446d9a8e7fa96c1bfe879 : FAILURE in 19m 07s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8dff02b09a0f42859bbdf660f2278fca : SUCCESS in 49m 14s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa5dff66e520400f969e6f9a1aacc2f7 : SUCCESS in 55m 13s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ca455c11d954516be379e72240696fe : SUCCESS in 46m 00s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5466760db1cb484dbe969aa3d3d4e17d : SUCCESS in 36m 42s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b7ed8411e31f4754a65bef8cda1ab762 : SUCCESS in 1h 17m 25s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9ffc2ae5e195440aad1bf2b7b97a2192 : SUCCESS in 1h 25m 40s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/598550484ba84cc4b2eb002868c44e39 : SUCCESS in 1h 23m 58s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cbcc1d9dbb9041bdbfff941f92720ad2 : SUCCESS in 1h 02m 50s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/f0eb367ca6de49a79d8fcdef9352a637 : SUCCESS in 18m 33s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f97d686bdbd8467e982df091fe284d03 : SUCCESS in 14m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e5edb6bfd154629a7e5db78305166ee : SUCCESS in 3m 04s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5941ffb26f024d8493025561e573aee3 : SUCCESS in 2h 50m 41s","accounts_in_message":[],"_revision_number":2},{"id":"ff1569ae209b6abcbb91dae65a04971a954dc042","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 02:17:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9b7a34bac33e3c5469202299f578d76ff5bd4922","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 04:49:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f5563da5fd74bc88c1557af24bc84a9 : SUCCESS in 4m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/feaea9f993ef4b059573dbc89893e256 : SUCCESS in 4m 09s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/f968f7d4875c43b3bf1a50664487a01e : SUCCESS in 17m 04s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a8b1c665a59b4c0495ced3e820c723b7 : SUCCESS in 16m 20s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7b0330f3f3bb40fa9cf1e42266664095 : SUCCESS in 1h 20m 18s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/50db5e228b6841e98ef864576bc5d641 : SUCCESS in 1h 30m 49s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9e17129ba2cf4b3fbc5a16e4ecd36530 : SUCCESS in 1h 26m 12s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/103fc1461e474e3c9b65f3267cf6096b : SUCCESS in 1h 16m 25s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/255bb2757ad646b998b27bcaf8f8e35d : SUCCESS in 18m 56s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0a3baf93f584e88b22600c5326569dd : SUCCESS in 14m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0404d0a11bf840c781a95433a22e5fd9 : SUCCESS in 3m 53s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3caf04d6d8ed4dc9867221bf34e2efe2 : SUCCESS in 2h 28m 06s","accounts_in_message":[],"_revision_number":2},{"id":"65afcde076c663b3a14f5b278de57beddfb671be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 04:50:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"72fcb1bb5ad11748a47a84bf6274c8905a42cbf0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 04:51:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78865b642b344322b759f82698a8560e : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4ca9ac106243433eebca1492218d72718617d1d0","revisions":{"40ffaa9f3232166ece66b56fc27de3083e7c7c58":{"kind":"REWORK","_number":1,"created":"2020-12-24 07:32:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/768438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/38/768438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/38/768438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/38/768438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/38/768438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/38/768438/1"}}},"commit":{"parents":[{"commit":"c87498ac3b3167c1d16cf363a0ec0cd644293d49","subject":"Merge \"Add support for the resource_provider_hypervisors parameter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c87498ac3b3167c1d16cf363a0ec0cd644293d49"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-24 07:30:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-24 07:30:48.000000000","tz":540},"subject":"Deprecate parameters for LBaaS quotas","message":"Deprecate parameters for LBaaS quotas\n\n... because Neutron LBaaS was deprecated a while ago.\n\nChange-Id: Ib91b8215ff8f07d7e81382cd197d583b6657cc42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/40ffaa9f3232166ece66b56fc27de3083e7c7c58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/40ffaa9f3232166ece66b56fc27de3083e7c7c58"}]},"branch":"refs/heads/master"},"4ca9ac106243433eebca1492218d72718617d1d0":{"kind":"REWORK","_number":2,"created":"2020-12-25 00:39:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/768438/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/38/768438/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/38/768438/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/38/768438/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/38/768438/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/38/768438/2"}}},"commit":{"parents":[{"commit":"c87498ac3b3167c1d16cf363a0ec0cd644293d49","subject":"Merge \"Add support for the resource_provider_hypervisors parameter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c87498ac3b3167c1d16cf363a0ec0cd644293d49"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-24 07:30:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-25 00:38:37.000000000","tz":540},"subject":"Deprecate parameters for LBaaS quotas","message":"Deprecate parameters for LBaaS quotas\n\n... because Neutron LBaaS was deprecated a while ago.\n\nChange-Id: Ib91b8215ff8f07d7e81382cd197d583b6657cc42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4ca9ac106243433eebca1492218d72718617d1d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4ca9ac106243433eebca1492218d72718617d1d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
