)]}'
{"id":"openstack%2Fpuppet-neutron~768836","triplet_id":"openstack%2Fpuppet-neutron~master~I8891efc94ceafe7ee7c789b68c410629fc240c91","project":"openstack/puppet-neutron","branch":"master","topic":"plugins-ml2","hashtags":[],"change_id":"I8891efc94ceafe7ee7c789b68c410629fc240c91","subject":"Replace vnic_type_blacklist by vnic_type_prohibit_list","status":"MERGED","created":"2021-01-02 02:09:28.000000000","updated":"2021-02-23 22:54:46.000000000","submitted":"2021-02-23 22:53:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768836-plugins-ml2","meta_rev_id":"f709b0c1c09df1384f76a753a7e838390c0fc7d7","_number":768836,"virtual_id_number":768836,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-23 22:53:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-02-23 14:44:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-02-23 17:54:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-02-23 17:54:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-02 04:44:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-23 14:44:15.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-02-23 17:54:30.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"89cabbe142f9743c3b8b42f776ebc1ca3b782670","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-02 02:09:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"523469fa3dbf1ef132d182e6739812465fc476b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-02 04:44:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a6f8931275745589add971204072af7 : FAILURE in 3m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7650499b9dfc4310acbdbcce72cad237 : SUCCESS in 2m 44s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/ce8c32853ff5405ba31b8b264cb9bd7a : FAILURE in 19m 28s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/68b8ee073a44487193f0103753ca7267 : FAILURE in 17m 12s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eed82ece162e413a8ce7be6bda1cdfea : FAILURE in 18m 21s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b05c6c9639a6408fbfe315586ef5514d : SUCCESS in 52m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/400ec929353f4b3aa82cabbb1470b3a1 : SUCCESS in 1h 00m 03s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25e3a53569ef4720ab8ca2a98100996d : SUCCESS in 48m 05s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9dc0e33e48444a12885d3f1dcf289de2 : SUCCESS in 41m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8280a200fc424f5f85599c11d9fe29a3 : SUCCESS in 1h 28m 55s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1d5b5fcf4cab4cb8a8ba5256552ac4d8 : SUCCESS in 1h 38m 36s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0c5b09346cd34c49936ed82b009ac79f : SUCCESS in 1h 36m 48s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6e0020f27b874f9eb2ade0d2e9c6359e : SUCCESS in 1h 09m 58s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/0c1df5d10e87430782d881ac33027819 : SUCCESS in 26m 13s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3afdea480fe34f9487b22f2a00dcd6f4 : SUCCESS in 14m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9e4fdf61b104fcdbb700d44424dc7ed : SUCCESS in 3m 07s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2ba8d7132a014a4caaebf771e0053e6a : FAILURE in 2h 33m 09s","accounts_in_message":[],"_revision_number":1},{"id":"1e0c4fae2a6ab9b7b1a04527b17e296f9cfc9f6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-02 05:46:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9effc5da1e57d19634273d899d495e3ef28c2f4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-02 07:25:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4ef485e5455dc1e10f24c0ec3de0273e4da41e23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-02 09:46:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/930f99bff57d422ab4000a2358778a3c : SUCCESS in 2m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b67b5195cb5468986b4a6c3ccf5af14 : SUCCESS in 2m 41s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c48380ab736e4bb4ae8bd59235a8b723 : SUCCESS in 18m 14s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afd90e3b9b004e7182422affa70a70e1 : SUCCESS in 17m 29s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6280904b757a4c90b2e1ed538b749ec7 : FAILURE in 17m 30s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43355358d9df42c5b88fce497db2a316 : FAILURE in 27m 25s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8bcc14acf6734873a25e7bb4600ccc0b : SUCCESS in 53m 07s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b06135ac8b4245018f8d7dd5cbabfb56 : SUCCESS in 45m 17s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a4249aa897245cda597f333c2208a00 : SUCCESS in 41m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ce1a50eb90d745bca50962cc7a22d7fc : SUCCESS in 1h 42m 19s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b065c5301927404b92ed423199a2ab2b : SUCCESS in 55m 04s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/52451bdb15aa4ef48c8955de5db456ca : SUCCESS in 1h 43m 07s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9a72f9c74eff416fa39df3710dbbf626 : SUCCESS in 56m 55s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/3b066c53fe90489e8e15e6af0510c27a : SUCCESS in 20m 31s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60e5dbcbae764d35af64166cf3fc559d : SUCCESS in 13m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c24772141b4471c93c37b1f096460ef : SUCCESS in 2m 47s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8a2b88ed7e294bcda6354684f975060d : FAILURE in 2h 19m 53s","accounts_in_message":[],"_revision_number":3},{"id":"ff98b65bb737d0231e18dda0466ceaf0d9cd9dbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-03 05:50:05.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"35abb385d8f139b5322fcf68c4324da31bbf4dd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 08:20:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2ff2ffcd16f4740bf2887ee90e14284 : SUCCESS in 2m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02d161213d764b9b90501219fc005944 : SUCCESS in 3m 26s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e83e3754be3b49a6a76e7ee1d9c59be6 : SUCCESS in 15m 35s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d2e6b85cf34459a89e10a3dd521eb68 : SUCCESS in 15m 12s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03f1ef655137431d8482234ee45dd079 : FAILURE in 18m 50s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3972d6e94d14b2488ddc3ea61114821 : SUCCESS in 52m 31s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12b5b12ac5f443afae53168914962cfa : SUCCESS in 52m 05s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88f30075ac39479d9f6a625dbcb171a7 : SUCCESS in 43m 08s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/228454533f2f452f94cd41e70f5a2706 : SUCCESS in 46m 56s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/50b7a07a675843ae9583eb019a93fbf0 : SUCCESS in 1h 27m 59s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/540bf404c389486085ff0123394a2da3 : SUCCESS in 1h 23m 29s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/25b06aaa00c64691a09e580fa375d233 : SUCCESS in 1h 14m 47s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1af7f247fbd94f618198b3c81889bdf7 : SUCCESS in 1h 11m 23s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/59b5f2bb64ba4434a4752db994ae5e52 : SUCCESS in 18m 43s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7985ea50feeb47279fef6d6a2f57b9ee : SUCCESS in 20m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1aa01cbc6cab44f5a38b01263cd9abf1 : SUCCESS in 4m 17s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/96af89a8698c425c98021de1a6b93423 : FAILURE in 2h 28m 50s","accounts_in_message":[],"_revision_number":4},{"id":"9b1f2751ee05b248fa5329cedef24c6db12780dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-26 12:33:51.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"aa49576efaa31903f068e1c5e4e6ebcee3236489","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 15:42:45.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b6ca8e172264d5e868eff04125960a0 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/68d8759becb74d5694e1b7488fbe3d9f : SUCCESS in 4m 19s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/8639a97d5067444b961d0b5a29d6856b : SUCCESS in 14m 13s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3c85f24dea84f288f615ed9a630cc2f : SUCCESS in 10m 49s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/351380305a444014aff9054d042a1739 : FAILURE in 17m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1557facb6bc4d0e91c491748b063014 : SUCCESS in 51m 20s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47620b621ff84f6491dc75a4a1193f79 : SUCCESS in 59m 18s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a77ee3de8d0144d1941255ff63ba2a6b : SUCCESS in 54m 11s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/026d539239ee45199f966e1af850e708 : SUCCESS in 45m 02s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e71392d61fd54f8292c29af5d0c4bada : SUCCESS in 1h 11m 09s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/116a3ce2d5db491083b26fed2dbb6c80 : FAILURE in 1h 43m 41s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/40a4805b04104e0d9895df85129dada1 : SUCCESS in 1h 23m 35s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/40331c79c8be41a1915402f2839093a0 : SUCCESS in 1h 05m 41s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/0c7b32a07c2a43809d7645c714540b53 : SUCCESS in 27m 59s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1adc5a092e4d457e94c84aa08d24d802 : SUCCESS in 13m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23843d3923ef4de59cd78bc1dfc115cb : SUCCESS in 3m 30s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f4b6920628ba4688a2731d2bd67526bf : SUCCESS in 2h 59m 55s","accounts_in_message":[],"_revision_number":5},{"id":"b7458003f01671385bb07a19a8008b25702a283e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-15 14:15:41.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"6b1cd36a3d1c8f24e6a80155d4c0b5fa503408bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 16:52:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a455bfcd7ad47ea93ba49a01c1fd0d5 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2bf4dc3b74ff40a1a115880e431bfa25 : SUCCESS in 3m 33s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/b7b1a864a40442b4a68e7070b62a34d6 : SUCCESS in 14m 01s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/ab95cc6e92a544a08512d48bcd8d32f8 : SUCCESS in 14m 19s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/228e54b16df54e7aab624a1e8937acbe : SUCCESS in 11m 50s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f34eaeaf7e8a4e82a11ce5ed9db4d622 : FAILURE in 14m 55s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f09b03a24f24ffab221bee3c907a10b : FAILURE in 43m 24s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34581102b45d486481b356ca39be1079 : FAILURE in 39m 57s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db27e7ee3d674fdc88cc498833f16855 : FAILURE in 32m 41s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/945bbb9f264b4034a8d47b8d44f480c9 : FAILURE in 24m 51s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4d317d9bf51649219226cee475e61b74 : SUCCESS in 1h 20m 35s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/27f8aa87fad849b6bc9e9c759f61f628 : SUCCESS in 1h 36m 22s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d677642e66ec4ca18f29e6a0f23be631 : SUCCESS in 1h 23m 03s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dd7955caca01469fb53ea23f2b7810c9 : SUCCESS in 1h 06m 10s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/24a96d7148514a74988ce61d0948521d : SUCCESS in 1h 14m 18s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4327c56df1444cbaaea7614fe769d7cc : SUCCESS in 1h 18m 34s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/91b316471a1049c1943a98813c8f8edb : SUCCESS in 1h 13m 46s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a1ddb3efa37d458aa6455d855240b6be : SUCCESS in 1h 00m 34s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/ea1f953e215e4f2d9302dd5dd366a8e9 : SUCCESS in 19m 24s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6dceb7af5b7642a49322d9e119cc57f1 : SUCCESS in 14m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dba022c4fdd445008c507be1cf9e84eb : SUCCESS in 3m 35s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2cd859aa868248b8b2ce658190352758 : SUCCESS in 2h 25m 39s","accounts_in_message":[],"_revision_number":5},{"id":"f84ee4815b3f7e157abd01fa789b8f52d320336f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-02-23 14:44:15.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e2916dfcfdbdd81091b4bf546ec1d762b6baabf6","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-02-23 17:54:30.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"4949ad8ba1f45fdbc9e635e7746a19679429ec9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 17:54:51.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"242143161182488f2401873eaf4ec1cb27cab7a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 22:53:21.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19681d4a0d004e60bed0ad348eaedf7f : SUCCESS in 4m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54144234b1fd406f82ce743bf72ee891 : SUCCESS in 4m 18s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d8705c75f78e433b9a2e8850c77ea926 : SUCCESS in 16m 06s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afc9a10f0e164d82ad8d76727fab41e3 : SUCCESS in 13m 35s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/880dc031082847aebaf03d8a9e9089d3 : SUCCESS in 1h 32m 21s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e5814e981fd24ec58019a8e756d3b294 : SUCCESS in 1h 47m 54s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/760e2841d3604619bfb2cdb43d61a8b9 : SUCCESS in 1h 25m 07s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/53917adad9dc4767a6fae6e77efc8cc7 : SUCCESS in 1h 02m 56s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/57c4393dc019462aa7df53ea30982fd8 : SUCCESS in 23m 56s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/752c07039f1f419aac4cdf967ef7f2fc : SUCCESS in 15m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07809745989f49e398576eb38a08e523 : SUCCESS in 3m 57s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2cc97d1d9786413988554048b2305e4a : SUCCESS in 2h 47m 44s","accounts_in_message":[],"_revision_number":5},{"id":"d4154d2b3b6e0e311ac80105fab8c013f07c9a19","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 22:53:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"f709b0c1c09df1384f76a753a7e838390c0fc7d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 22:54:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf41ee8bea9541cb96c079ffbd7bb980 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2c5ed70823415e4cfe35ddf1555ee63df61be9b3","revisions":{"47a84c4c9a30967a643d9f7c174bd37a809dfff1":{"kind":"REWORK","_number":1,"created":"2021-01-02 02:09:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/768836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/36/768836/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/768836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/768836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/768836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/36/768836/1"}}},"commit":{"parents":[{"commit":"d8b65c34dc64f81c02750fa87127af9b8a4d56a3","subject":"Ensure vnic_type_blacklist is unset by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d8b65c34dc64f81c02750fa87127af9b8a4d56a3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-02 02:05:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-02 02:07:48.000000000","tz":540},"subject":"Replace vnic_type_blacklist by vnic_type_prohibit_list","message":"Replace vnic_type_blacklist by vnic_type_prohibit_list\n\n... to follow renaming in neutron[1].\n\n[1] 055036ba2bc1fd1cfc54ef63dc6cb82037111950\n\nChange-Id: I8891efc94ceafe7ee7c789b68c410629fc240c91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/47a84c4c9a30967a643d9f7c174bd37a809dfff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/47a84c4c9a30967a643d9f7c174bd37a809dfff1"}]},"branch":"refs/heads/master"},"cfd12ce5efb00b67c10e77464df6994226afebc2":{"kind":"REWORK","_number":2,"created":"2021-01-02 05:46:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/768836/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/36/768836/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/768836/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/768836/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/768836/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/36/768836/2"}}},"commit":{"parents":[{"commit":"708bafde5758ed733c7c3c916209cff81ba75008","subject":"Ensure vnic_type_blacklist is unset by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/708bafde5758ed733c7c3c916209cff81ba75008"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-02 02:05:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-02 05:46:13.000000000","tz":540},"subject":"Replace vnic_type_blacklist by vnic_type_prohibit_list","message":"Replace vnic_type_blacklist by vnic_type_prohibit_list\n\n... to follow renaming in neutron[1].\n\n[1] 055036ba2bc1fd1cfc54ef63dc6cb82037111950\n\nChange-Id: I8891efc94ceafe7ee7c789b68c410629fc240c91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/cfd12ce5efb00b67c10e77464df6994226afebc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/cfd12ce5efb00b67c10e77464df6994226afebc2"}]},"branch":"refs/heads/master"},"105f305f99c842d75eb0050f6e1366e757582566":{"kind":"REWORK","_number":3,"created":"2021-01-02 07:25:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/768836/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/36/768836/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/768836/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/768836/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/768836/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/36/768836/3"}}},"commit":{"parents":[{"commit":"708bafde5758ed733c7c3c916209cff81ba75008","subject":"Ensure vnic_type_blacklist is unset by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/708bafde5758ed733c7c3c916209cff81ba75008"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-02 02:05:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-02 07:25:17.000000000","tz":540},"subject":"Replace vnic_type_blacklist by vnic_type_prohibit_list","message":"Replace vnic_type_blacklist by vnic_type_prohibit_list\n\n... to follow renaming in neutron[1].\n\n[1] 055036ba2bc1fd1cfc54ef63dc6cb82037111950\n\nChange-Id: I8891efc94ceafe7ee7c789b68c410629fc240c91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/105f305f99c842d75eb0050f6e1366e757582566"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/105f305f99c842d75eb0050f6e1366e757582566"}]},"branch":"refs/heads/master"},"71f4fdfc048f6b7b8f99858b63f5e138fbd186c4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-01-03 05:50:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/768836/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/36/768836/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/768836/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/768836/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/768836/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/36/768836/4"}}},"commit":{"parents":[{"commit":"c1a3259fde02a5a06dda87538eb0a10162a902da","subject":"Ensure vnic_type_blacklist is unset by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c1a3259fde02a5a06dda87538eb0a10162a902da"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-02 02:05:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 05:49:57.000000000","tz":540},"subject":"Replace vnic_type_blacklist by vnic_type_prohibit_list","message":"Replace vnic_type_blacklist by vnic_type_prohibit_list\n\n... to follow renaming in neutron[1].\n\n[1] 055036ba2bc1fd1cfc54ef63dc6cb82037111950\n\nChange-Id: I8891efc94ceafe7ee7c789b68c410629fc240c91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/71f4fdfc048f6b7b8f99858b63f5e138fbd186c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/71f4fdfc048f6b7b8f99858b63f5e138fbd186c4"}]},"branch":"refs/heads/master"},"2c5ed70823415e4cfe35ddf1555ee63df61be9b3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-01-26 12:33:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/768836/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/36/768836/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/768836/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/768836/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/36/768836/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/36/768836/5"}}},"commit":{"parents":[{"commit":"0d34714dad01e7f59ca06abf4486a997498dd8b1","subject":"Merge \"Prepare Wallaby M2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0d34714dad01e7f59ca06abf4486a997498dd8b1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-02 02:05:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-26 12:33:51.000000000","tz":0},"subject":"Replace vnic_type_blacklist by vnic_type_prohibit_list","message":"Replace vnic_type_blacklist by vnic_type_prohibit_list\n\n... to follow renaming in neutron[1].\n\n[1] 055036ba2bc1fd1cfc54ef63dc6cb82037111950\n\nChange-Id: I8891efc94ceafe7ee7c789b68c410629fc240c91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2c5ed70823415e4cfe35ddf1555ee63df61be9b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2c5ed70823415e4cfe35ddf1555ee63df61be9b3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
