)]}'
{"id":"openstack%2Fpuppet-neutron~768960","triplet_id":"openstack%2Fpuppet-neutron~master~I4630ad619382d00132e6b4ab0dc43bf72146073b","project":"openstack/puppet-neutron","branch":"master","topic":"typo","hashtags":[],"change_id":"I4630ad619382d00132e6b4ab0dc43bf72146073b","subject":"Fix typos in comment lines","status":"MERGED","created":"2021-01-03 05:50:58.000000000","updated":"2021-01-05 19:05:17.000000000","submitted":"2021-01-05 19:05:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768960-typo","meta_rev_id":"9f035d0b4e5605a5f871ab541e52989eda213ff1","_number":768960,"virtual_id_number":768960,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-05 19:05:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-01-05 15:34:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-01-05 15:46:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-01-05 15:46:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-03 08:24:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-04 20:42:12.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"CC"},{"updated":"2021-01-05 15:34:14.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-01-05 15:46:37.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"df72cf7e4b3ed99d2cd66c99e95a6909291f4924","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-03 05:50:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4898aea362e951e5f92f68f7a6d2303fbf1c494b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-03 05:51:22.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c3a3253b901b879c4a329eca5ead1467462af204","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 08:24:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/620fa5e5995143779da831dfdb0086ce : SUCCESS in 2m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94c517a225ed4d00acc9e04ad417e629 : SUCCESS in 2m 56s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/90db44db00634e5cbbcd5b51a9e7bff7 : SUCCESS in 18m 53s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ece57b964854d7388ef89674c4ee221 : SUCCESS in 16m 05s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48d23bea371140f796a43ee281299880 : FAILURE in 16m 15s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a37e5e211fd4d098912b463cc826400 : SUCCESS in 1h 09m 40s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/206f3b294f4c4e83b7129242182f1033 : SUCCESS in 1h 09m 50s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfeafc3055764c9db9b4763c3b326446 : SUCCESS in 1h 02m 58s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c5b05e8c7ff42c48eb30fcdbff09d47 : SUCCESS in 46m 09s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8a0d3d81ea9444618ab851172b66c39f : SUCCESS in 1h 29m 20s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/52f226a557504bc8a603c39bb4ff969a : SUCCESS in 1h 41m 46s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/97163504ee3148fcb4989da382398c18 : SUCCESS in 1h 29m 11s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f27006b42a554788928a47eb8eff4252 : SUCCESS in 1h 16m 22s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/c3594338694d4831a26513b2be989a91 : SUCCESS in 17m 44s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1115dc7e8b01420c8323c0ae68297c15 : SUCCESS in 13m 31s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4c7e1d6e879740bf82a48fb69aaa99fb : FAILURE in 2h 28m 31s","accounts_in_message":[],"_revision_number":2},{"id":"f0870d4353ef1b24c0f1e827d56c7f3363da4b7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-03 22:23:45.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"8b7b5bde40055578c0044bc7615d5cd0e7ed4648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 00:50:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8adea29436e42abbc7eb7577eb17c3f : SUCCESS in 3m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28c48baba4144d1bbf54a3e91e95dfa3 : SUCCESS in 2m 50s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/79ce545fe1574e858a6107f69bdf8534 : SUCCESS in 18m 55s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0fbb74297598471e85bee983134f1f2e : SUCCESS in 15m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/70247b1ba38a46c2a01562c5b0d0cf0a : FAILURE in 16m 46s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9fd93fe6902f499a9d16168007ee994c : FAILURE in 28m 16s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6a1c9bc7b2f4081b514c0d02b5001ea : SUCCESS in 52m 32s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/758aef4a38284fd4922124020d9346ef : SUCCESS in 46m 06s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28b247841a264d1e8d7246cbf0bc00cb : SUCCESS in 39m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/96509330078b45989baa548e1ce60e9a : SUCCESS in 1h 29m 42s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6fd18915903a4e128d213f9a87f81acb : SUCCESS in 1h 38m 31s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/502a89ba33344d22a1627069c2b2a6f7 : SUCCESS in 1h 27m 32s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4de405494c534fff87bb5d718de5ea4d : SUCCESS in 1h 05m 47s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/be5d2aba7a86445689e7020d2539f6ba : SUCCESS in 18m 52s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93661fb5d3eb4f5eab67c45db7fee4ab : SUCCESS in 14m 54s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8b9204cc428842db8767ded50ef4bd00 : SUCCESS in 2h 21m 20s","accounts_in_message":[],"_revision_number":3},{"id":"f5812b67c1158e44b7d821221d1014b2c0061c86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-04 13:42:18.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"0e0f86dfdb391641bb8b91ec528c949a0b5271fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 16:43:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2550c15b070b4f90a559121f13a60308 : SUCCESS in 4m 53s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a20a9753cb1d4225bd4f6d89f37ef8c9 : SUCCESS in 3m 33s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/fa21feb757ef48edb7e963f040438d39 : SUCCESS in 16m 40s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6417ebe031414f05a4ac4b164938a9ae : SUCCESS in 12m 49s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/451f1d1d5dee4fd5b897f127bd45d3ad : FAILURE in 19m 31s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12a1af0e5af84c508e31cfeb31e12114 : SUCCESS in 1h 02m 19s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aefca6127684403c8724efedbc9df4fb : SUCCESS in 56m 34s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e1f1fa6af504673a25192a4bdfacc6c : SUCCESS in 45m 04s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34020aa7ec5b4ec19ff761f1425ade81 : SUCCESS in 40m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/769b96dc77da42ceba8e1c3736154ba5 : SUCCESS in 1h 30m 15s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c25d81b8cda14d66acc6e1e88395e860 : SUCCESS in 1h 35m 48s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/071f6aeb83894c52a001ba0de2fdd328 : SUCCESS in 1h 35m 28s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f775041f3b3245d79d2dd425702f2e0e : SUCCESS in 1h 06m 53s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/6ae9fe0e96f04666b3b2653bf478667e : SUCCESS in 26m 35s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b803543d22de4b14acb79c15cb444c25 : SUCCESS in 16m 04s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/64b0e7a25e714a4b82cc3313abd8e7f2 : FAILURE in 2h 53m 04s","accounts_in_message":[],"_revision_number":4},{"id":"56f2a98179d48b4d98bdceb255239588fee61e8c","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-01-04 20:42:12.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e88c392625b1c407767ba099b89614ce8855e973","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 00:05:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a139e4583662448a803af8c4b43b26c2 : SUCCESS in 4m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7221f1ece1824091a089429c50411362 : SUCCESS in 3m 43s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/2699f6c979104a6cbf285a8528893597 : SUCCESS in 14m 46s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1bfbcd99f680473d98f88d6f178ee538 : SUCCESS in 16m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ed6c110118445ce9e5eb82726fdbd8c : FAILURE in 12m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14bbb91b95e6456c97c30ec5400fabb9 : SUCCESS in 57m 11s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ccc594bd5a9475dbe7d4d30e4a43393 : SUCCESS in 44m 59s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9458def167d04b3bb2ef5e9fa6ceb976 : SUCCESS in 58m 55s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a96f90aa32aa46fd90402b2d6b1e2446 : SUCCESS in 47m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/63fe05940fe847a09f1aebb9af8964a0 : SUCCESS in 1h 36m 21s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/854479b148664403b19fa5c2251ef628 : SUCCESS in 55m 47s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a83705cd93ef4a6bbf89bf96be043245 : SUCCESS in 1h 33m 14s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9ed8f3f5f6064b82a64b29a8942d6b84 : SUCCESS in 1h 12m 09s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/0418325dc73d4295b7344602007bbdeb : SUCCESS in 36m 18s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18a9bf0778b7486c98b670395974c6c4 : SUCCESS in 15m 25s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7634912b749347d5943188b1646acc19 : SUCCESS in 2h 34m 28s","accounts_in_message":[],"_revision_number":4},{"id":"339576e20017893b52caf3c2921514f1a9276128","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-05 15:34:14.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"111c35864f2a5b8a9f8feb34f6ba6db38f3ad3cf","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-01-05 15:46:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"0e0a2026f9ccbff54ad36e0854adf99ef049760d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 15:46:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"66be1943bad8d026e74d794ab038b8cce93193a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 19:05:16.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/accc48c3d44045d9baf0482de6e166a2 : SUCCESS in 5m 25s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2c5836b937d4f4e88f67d81f8572985 : SUCCESS in 3m 18s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/67ee140da7c141fab9991039b39f3f6d : SUCCESS in 21m 16s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa0ad3cd524f49b58ae18c0d8d3a0bcb : SUCCESS in 17m 43s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4844971d785c43b49c7d46e9084aeb05 : SUCCESS in 1h 25m 19s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6ea6dd471bb34aef87745a06f550ebda : SUCCESS in 1h 59m 25s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/34b1a92b58a54b86a1e1ddfa51da4167 : SUCCESS in 1h 36m 15s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7aefee046ded4b278aa2d40f88b7adf1 : SUCCESS in 1h 25m 51s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/2603348539fb40748806bfa1cb2e73c2 : SUCCESS in 19m 21s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/565ab98eebe24903b4295aad5eb9ebb2 : SUCCESS in 22m 04s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3cf70d453a074d28ace988c9be7fc004 : SUCCESS in 3h 10m 56s","accounts_in_message":[],"_revision_number":4},{"id":"9f035d0b4e5605a5f871ab541e52989eda213ff1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 19:05:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ec1a5f24c3cea1ef2e84b82623e29c472c960a2d","revisions":{"d593cc7e147e54a5ca6faa9b50f52b1b395961fd":{"kind":"REWORK","_number":1,"created":"2021-01-03 05:50:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/768960/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/60/768960/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/60/768960/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/60/768960/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/60/768960/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/60/768960/1"}}},"commit":{"parents":[{"commit":"c87498ac3b3167c1d16cf363a0ec0cd644293d49","subject":"Merge \"Add support for the resource_provider_hypervisors parameter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c87498ac3b3167c1d16cf363a0ec0cd644293d49"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 05:50:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 05:50:36.000000000","tz":540},"subject":"Fix a typo in comment lines","message":"Fix a typo in comment lines\n\nChange-Id: I4630ad619382d00132e6b4ab0dc43bf72146073b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d593cc7e147e54a5ca6faa9b50f52b1b395961fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d593cc7e147e54a5ca6faa9b50f52b1b395961fd"}]},"branch":"refs/heads/master"},"02534e314d04a07b9f37455bd9028353a78c3b66":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-03 05:51:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/768960/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/60/768960/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/60/768960/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/60/768960/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/60/768960/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/60/768960/2"}}},"commit":{"parents":[{"commit":"c87498ac3b3167c1d16cf363a0ec0cd644293d49","subject":"Merge \"Add support for the resource_provider_hypervisors parameter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c87498ac3b3167c1d16cf363a0ec0cd644293d49"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 05:50:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 05:51:08.000000000","tz":540},"subject":"Fix typos in comment lines","message":"Fix typos in comment lines\n\nChange-Id: I4630ad619382d00132e6b4ab0dc43bf72146073b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/02534e314d04a07b9f37455bd9028353a78c3b66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/02534e314d04a07b9f37455bd9028353a78c3b66"}]},"branch":"refs/heads/master"},"b81ac79f1dc20ee46300da96991d5079ed116fec":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-03 22:23:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/768960/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/60/768960/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/60/768960/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/60/768960/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/60/768960/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/60/768960/3"}}},"commit":{"parents":[{"commit":"18c3fd2b04a4a998df15ec894e8922baf7b9df41","subject":"Disable neutron_tempest_plugin in TripleO job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/18c3fd2b04a4a998df15ec894e8922baf7b9df41"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 05:50:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 22:23:45.000000000","tz":0},"subject":"Fix typos in comment lines","message":"Fix typos in comment lines\n\nChange-Id: I4630ad619382d00132e6b4ab0dc43bf72146073b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b81ac79f1dc20ee46300da96991d5079ed116fec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b81ac79f1dc20ee46300da96991d5079ed116fec"}]},"branch":"refs/heads/master","description":"Rebase"},"ec1a5f24c3cea1ef2e84b82623e29c472c960a2d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-01-04 13:42:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/768960/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/60/768960/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/60/768960/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/60/768960/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/60/768960/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/60/768960/4"}}},"commit":{"parents":[{"commit":"2314c53abfdbf800de9c08dbf3af8a9215b5f93f","subject":"Add OVS bridge_mac_table_size parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2314c53abfdbf800de9c08dbf3af8a9215b5f93f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 05:50:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-04 13:42:18.000000000","tz":0},"subject":"Fix typos in comment lines","message":"Fix typos in comment lines\n\nChange-Id: I4630ad619382d00132e6b4ab0dc43bf72146073b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ec1a5f24c3cea1ef2e84b82623e29c472c960a2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ec1a5f24c3cea1ef2e84b82623e29c472c960a2d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
