)]}'
{"id":"openstack%2Fpuppet-neutron~769010","triplet_id":"openstack%2Fpuppet-neutron~master~I3139cc238036ec681d4eda18bb0aebc9f1404c4e","project":"openstack/puppet-neutron","branch":"master","topic":"ovn-vif_type","hashtags":[],"change_id":"I3139cc238036ec681d4eda18bb0aebc9f1404c4e","subject":"OVN: Deprecate the vif_type option","status":"MERGED","created":"2021-01-03 08:15:41.000000000","updated":"2021-01-15 12:12:28.000000000","submitted":"2021-01-15 12:11:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769010-ovn-vif_type","meta_rev_id":"7d6ba67cf24e46e449a7ef3e05380504c8119bd3","_number":769010,"virtual_id_number":769010,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-15 12:11:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-01-05 15:34:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-01-05 15:46:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-01-05 15:46:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-03 10:44:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-04 20:42:25.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"CC"},{"updated":"2021-01-05 15:34:03.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-01-05 15:46:59.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"a34ad270a525d927aec2bcd2eff3edf237c9f15c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-03 08:15:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6cac589cd54586d9ed007aaf7af63408f221c6ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-03 08:16:08.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6cfe902c2b63e6e1db68e876b2aede4c6a84f8f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 10:44:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1293a3169f154bf590b04ed030899cf8 : SUCCESS in 3m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fbfdeeffc3434eaf91dece9a74d61f28 : SUCCESS in 3m 10s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0d6d5311d23944068a01eb677eb09d2a : SUCCESS in 19m 26s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/285e9eab47ca4d05a7ac1aa35a1bd9f9 : SUCCESS in 15m 48s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/180281e348704093954a663935bb6a9f : FAILURE in 18m 29s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a772895740c40728129c4755ee7eea5 : FAILURE in 37m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8aa86dc3e03b43aa9162863916cdfd99 : SUCCESS in 1h 02m 30s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db318d275e4d432198afa54d06ea4656 : SUCCESS in 50m 24s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f5b029667f648f2b707a7f28f7d6394 : SUCCESS in 42m 16s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1ebeb2d6960343d88777b67ec5dc8f5f : SUCCESS in 1h 22m 30s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/359f2a8100fe44b6960a4440af8f16b9 : SUCCESS in 1h 32m 02s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ceaeb1b4168142f3bea3e9f78acc1022 : SUCCESS in 1h 28m 15s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/71af5b0e931342a1a748dcf5586e8699 : SUCCESS in 1h 06m 58s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/1bce6e38e8a842778b7142727347ff11 : SUCCESS in 18m 15s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f738f14ae969484ab4f6da1995ddd722 : SUCCESS in 14m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5cefea53af74a72ab1e9c816ad371b7 : SUCCESS in 3m 39s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c2e99ca7647048fc8785a0404fc6ec2c : FAILURE in 2h 26m 54s","accounts_in_message":[],"_revision_number":2},{"id":"d6021a7e1f3254f7d329725783bfadcfcdd9c456","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-03 22:22:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"42cbdd60bc4d9f95bda149f4db3d5befe375c29b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 00:49:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce8d873c2c5c48bc95a8dad7b58002a1 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c3092ac51d3491ebd630c86f84e7ee4 : SUCCESS in 3m 14s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/b30e23d1334c47acbd194fd1f8253be7 : SUCCESS in 18m 48s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc6b6b7b469e4a1b849ebba539822231 : SUCCESS in 16m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a4c4bd39f2a4ef898c5985745c3af8b : FAILURE in 17m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/145cce35278349c79b165f823b3956dd : FAILURE in 28m 39s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e57c7201e1834413a5e94aa326e7f2f8 : SUCCESS in 53m 02s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8865175293243659c132852a4822b9a : SUCCESS in 48m 04s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a9a727dd36d41beb3af83728af302b4 : SUCCESS in 49m 05s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ce6478bd68644376af13ef15e765bb04 : SUCCESS in 1h 24m 09s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9e45aed2086b4f11ac7023259d51ea8e : SUCCESS in 1h 01m 21s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5334e9be24d5423b9e01b914af0685a3 : SUCCESS in 1h 07m 18s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fb98f7e87ea94c62b7d78b3155951358 : SUCCESS in 1h 16m 09s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/f1a6de5b2d504deea524f14a0c8ae800 : SUCCESS in 23m 04s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7685bc010bac433bbc038209d292b5d1 : SUCCESS in 14m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/707290a47fbf4327beef209460432aa2 : SUCCESS in 3m 04s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1053457f454942289d43a62f965813fc : SUCCESS in 2h 21m 26s","accounts_in_message":[],"_revision_number":3},{"id":"3a006b21a0d5625ab4889eb6ef5eb52643a3c1ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-04 13:42:05.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"266674174b705dcb4e80db365355024dd96b903d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 16:23:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9862619846294804b6e42fafbd3225aa : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a00c9b60ef8d47ccac2d1187d430db31 : SUCCESS in 3m 25s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d9b5551492af4feca7082e5bf8d6080e : SUCCESS in 19m 21s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcc6beb8c92d4005b15ca5ef364d1a90 : SUCCESS in 14m 45s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3379c153e857405db7e242fde27490c2 : FAILURE in 21m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0688aae6a1f461fbd37a7178644e03a : SUCCESS in 1h 03m 01s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22e4652c6d21466fb3033633a77c7632 : SUCCESS in 51m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c11e429703a84cd185133e98b0368d97 : SUCCESS in 55m 45s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fbcd0923299045788d6134d067c39ea2 : SUCCESS in 43m 14s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8c689bd780ac4ac382c2454e9a5e5a21 : SUCCESS in 1h 16m 31s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d9c9af95065944bc9fc0abb7ffbff5e9 : SUCCESS in 1h 39m 03s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/80af597be4a14e1f93b6a0311d95166c : SUCCESS in 1h 34m 50s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/618c0297c3e64c21ba9510a9006438fe : SUCCESS in 1h 07m 09s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/41dc15bfc8aa4dd08cf958944319dcfb : SUCCESS in 20m 33s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df53a8579e3d4a3883818c452aa75204 : SUCCESS in 14m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa52aaf1958248f08e311b02ee5f6560 : SUCCESS in 4m 16s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/faffe020483648559bfe9500f4f806fc : FAILURE in 2h 34m 50s","accounts_in_message":[],"_revision_number":4},{"id":"3a3304572245730972b34925b882f9ca5b5186e7","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-01-04 20:42:25.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2ef04c874459a8125b29ee5aca698886e0a7a19b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 23:20:11.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/70428a4b626644739df4cb0ea752d7f5 : SUCCESS in 3m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/409897fcf61b4a168acb6a4cb99f28d9 : SUCCESS in 3m 02s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/5168b9c4e6ea4182be0f82d7a55daabe : SUCCESS in 21m 29s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc5d919a3cce4814ad63bee1efb30aa8 : SUCCESS in 13m 55s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e4ab69790fe4dfc887708aef3678b8c : FAILURE in 14m 05s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9bf1ec3329b49e1995ff1cdab9b5d45 : SUCCESS in 58m 30s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25fd2f4576d24c6687353d54174be0c0 : SUCCESS in 58m 07s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1de212ba644440feb01b24a74bb2a2e9 : SUCCESS in 54m 29s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c70d5821e5f648cf8c34b720ddbd313a : SUCCESS in 46m 18s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f23d858921df404496b9d390949fa9ad : SUCCESS in 1h 21m 51s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fba9597899e149bb8d49741533947b91 : SUCCESS in 1h 31m 06s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4a81219d6d6b474e8b8b539c5f39b69a : SUCCESS in 1h 13m 29s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dc7ca10a08d1490e9f7dd3c46db18698 : FAILURE in 1h 37m 14s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/1b7027a838144dadbc9c60a3724cdbdf : SUCCESS in 20m 08s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22e4078aaf2f449491079d91308b4dd4 : SUCCESS in 14m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/506eaec3c0be456698f9a8be73366020 : SUCCESS in 3m 33s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/84fe59c19664443b9d29689455a86a26 : SUCCESS in 2h 32m 07s","accounts_in_message":[],"_revision_number":4},{"id":"5c28302a49b4235697dfdf9642c4ccd40b1fef26","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-04 23:22:54.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a1b626b2afb35459f4f441d80c6c9a533e165356","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 01:57:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cca3b4cd829f4ba18753054c0d9461ec : SUCCESS in 3m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7eaf073e312417396e27659ad5806e0 : SUCCESS in 3m 04s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/fcc1e495f3974d1cb2ad8a54bcc622c4 : SUCCESS in 17m 22s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abf8808633134580b9fdf70b8b9e5746 : SUCCESS in 15m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9aca9c222d7c4f0b8ab95023f8835fc7 : FAILURE in 17m 09s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eeccb97421a04da89173d524a0523fd6 : SUCCESS in 1h 11m 17s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c0669f11d0f45a490f94dd6330694d0 : SUCCESS in 51m 51s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db87061747314113b1f45ec66f675c9e : SUCCESS in 50m 45s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/118a7ac344404319b7c4b42e1b7149eb : SUCCESS in 43m 19s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/41d225c381c74afabe0ae817531dfb3f : SUCCESS in 1h 37m 16s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c37e63a02be240369e804489e492125b : SUCCESS in 1h 31m 40s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6e4e07fa307644febade71c5608914ca : SUCCESS in 1h 21m 08s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b5d1429e2d6a46afac930f2ac059e4d0 : SUCCESS in 1h 02m 06s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/017a52e9b59a4d6ab004bb2c51fbca80 : SUCCESS in 18m 50s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9d9ed12d33744d5b428c6168926c20e : SUCCESS in 16m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c7627d9de0e4127b9e517b5ab9a5a5c : SUCCESS in 3m 53s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b17a70582ec34a958407edd9cb63f5c6 : SUCCESS in 2h 30m 02s","accounts_in_message":[],"_revision_number":4},{"id":"f61827a35ef0d05512f7d14b1572ed8f4c052118","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-05 15:34:03.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e62b6644585a1874e11f1f4ebd636aeeed447fea","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-01-05 15:46:59.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2bc4b557bc5e119a1c5a17a43e3a616d3cde4521","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 18:51:37.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"17fc0669edc0e19335aa918a69faf4545039d30a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 22:29:18.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f77782f6c6b743fba143d34784000c40 : SUCCESS in 3m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d0b52f1c05f4e62a9114b44396b652e : SUCCESS in 3m 36s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/cde9a1f553094ece811efe5cd8696e4d : SUCCESS in 17m 36s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af7cbe0381594fc0add5b27f51325749 : SUCCESS in 11m 49s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c108c49c55d0425380ae811e1870629a : SUCCESS in 1h 18m 16s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f63c5a8a00c7430eb2e042b524d711c9 : SUCCESS in 1h 25m 13s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/13ded5c78d4c4a30831c5eca4d516b54 : SUCCESS in 1h 28m 09s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d37de62d5c5b49379c21d6b6bd278cc6 : SUCCESS in 1h 14m 28s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/a0de3378c2f544938c4ee6a22ed16bf1 : SUCCESS in 21m 45s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40ea9ab6f1084813a2b9a6094dbf3bbd : SUCCESS in 17m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83fa3d7747bb49ae90d17368fd4857f5 : SUCCESS in 3m 44s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/66f57b3cc3b845c1abeb996312cf3363 : TIMED_OUT in 3h 20m 02s","accounts_in_message":[],"_revision_number":4},{"id":"5fa2a2db458569eb8ad304b0ffb6935e71789e0f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-15 06:31:05.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"0b40b7f04aa3b30d3fb4623e01bc827905753520","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 09:58:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a754b4774dbf4d328c6ccd90266e81f4 : SUCCESS in 4m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93b493fee17c4b659251bb476c6027b0 : SUCCESS in 3m 35s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/f5b52561445f45faadcc8bc2d759630e : SUCCESS in 14m 22s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed1ecc0dd30b435d9066fa9d988eb0f0 : SUCCESS in 10m 53s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c28fc60c2b9847f7a60bab4c087b259c : FAILURE in 11m 48s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4bd908056674d39ac5aa7ece6a95d8c : SUCCESS in 57m 04s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f1683fee69347b9b7d9ef5bb022db71 : SUCCESS in 1h 01m 46s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae9013c822394a5d984812a6783eeef9 : SUCCESS in 54m 08s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/229a5b61152a4ba19f404dacc24a3f72 : SUCCESS in 46m 30s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bca7be4308ae4989b535c1593b347a34 : SUCCESS in 1h 20m 02s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4140e8eb920145eb9e7712c4bda0556e : SUCCESS in 1h 23m 37s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/53501ddf5773466ead0d4159899a9e79 : SUCCESS in 1h 36m 12s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/71e1f878f60c45b883920bf1c4e80877 : SUCCESS in 1h 05m 33s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/5ac39af33e854b85bcdda14a25cc5964 : SUCCESS in 22m 38s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e0457e21731490fa4fa4058a17a2eeb : SUCCESS in 15m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3fd010e8fbb4176bf8a2723eee9841f : SUCCESS in 3m 08s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1f3c327534a64fab86422b9be462dc86 : SUCCESS in 2h 46m 33s","accounts_in_message":[],"_revision_number":4},{"id":"4aa4599e67f22d5598155f513916c29d54878b55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 09:59:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4b8bfbe4cfa233911727d857c02ce7a7bdd1a4fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 12:11:15.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f22b0b713ac143aba4b90246b86e4920 : SUCCESS in 3m 53s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19964249e17f4ae6a248f12004a38a04 : SUCCESS in 3m 20s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1afb25df2c5c47ed97ede0b6eaa9976e : SUCCESS in 14m 36s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2faef960143343398d81aaee4e0927fe : SUCCESS in 12m 24s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fc2362e344c642c0aaca1c9c56b7450e : SUCCESS in 1h 33m 17s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/962b0c6704cb492e8fbba72924b6d194 : SUCCESS in 1h 33m 33s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/77a45e76cee043b8a1f0de54f0004be1 : SUCCESS in 1h 26m 09s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/555e25b6b5f74c2393745a9c62d2ec55 : SUCCESS in 1h 08m 10s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/d2020180c5044649bb9e354c0656571c : SUCCESS in 17m 13s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/791fbd9293ea4fc29c24e14211baceb1 : SUCCESS in 15m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df65c61f1233486e974a31e90f8de70d : SUCCESS in 4m 00s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ada78bd9809f4bf7ac56814b2d2268b0 : SUCCESS in 2h 08m 19s","accounts_in_message":[],"_revision_number":4},{"id":"6023788e3254b071f9030c457672ee1bb58eaf1e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 12:11:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"7d6ba67cf24e46e449a7ef3e05380504c8119bd3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 12:12:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3af8293ea1bc404390d0955a37157d36 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1ba7a40514850b059afbfc22da90cfea33ceaab5","revisions":{"e67d25442dba01e8d562677d48c39865ef1a106c":{"kind":"REWORK","_number":1,"created":"2021-01-03 08:15:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/769010/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/10/769010/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/10/769010/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/10/769010/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/10/769010/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/10/769010/1"}}},"commit":{"parents":[{"commit":"eefa0dbfe6f6883b72c4652b57f71cd18bcebdeb","subject":"OVN: Deprecate the ovn_l3_mode option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/eefa0dbfe6f6883b72c4652b57f71cd18bcebdeb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 08:14:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 08:14:38.000000000","tz":540},"subject":"OVN: Deprecate the vif_type parameter","message":"OVN: Deprecate the vif_type parameter\n\n... because it was already deprecated in networking-ovn and has no\neffect now[1].\n\n[1] 5be7dfeb637984f75937a21b994c1375dfa77d87\n\nChange-Id: I3139cc238036ec681d4eda18bb0aebc9f1404c4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e67d25442dba01e8d562677d48c39865ef1a106c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e67d25442dba01e8d562677d48c39865ef1a106c"}]},"branch":"refs/heads/master"},"cae18b72bfe578222d3ba73f148df78a5dc283a9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-03 08:16:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/769010/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/10/769010/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/10/769010/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/10/769010/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/10/769010/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/10/769010/2"}}},"commit":{"parents":[{"commit":"eefa0dbfe6f6883b72c4652b57f71cd18bcebdeb","subject":"OVN: Deprecate the ovn_l3_mode option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/eefa0dbfe6f6883b72c4652b57f71cd18bcebdeb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 08:14:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 08:15:54.000000000","tz":540},"subject":"OVN: Deprecate the vif_type option","message":"OVN: Deprecate the vif_type option\n\n... because it was already deprecated in networking-ovn and has no\neffect now[1].\n\n[1] 5be7dfeb637984f75937a21b994c1375dfa77d87\n\nChange-Id: I3139cc238036ec681d4eda18bb0aebc9f1404c4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/cae18b72bfe578222d3ba73f148df78a5dc283a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/cae18b72bfe578222d3ba73f148df78a5dc283a9"}]},"branch":"refs/heads/master"},"84dd78b7aefbb6b3c8f6fb64e3dbaab5aae785fa":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-03 22:22:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/769010/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/10/769010/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/10/769010/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/10/769010/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/10/769010/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/10/769010/3"}}},"commit":{"parents":[{"commit":"83719eb21c7be65fef7a8e8546a6e5a35159e9c9","subject":"OVN: Deprecate the ovn_l3_mode option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/83719eb21c7be65fef7a8e8546a6e5a35159e9c9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 08:14:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 22:22:41.000000000","tz":0},"subject":"OVN: Deprecate the vif_type option","message":"OVN: Deprecate the vif_type option\n\n... because it was already deprecated in networking-ovn and has no\neffect now[1].\n\n[1] 5be7dfeb637984f75937a21b994c1375dfa77d87\n\nChange-Id: I3139cc238036ec681d4eda18bb0aebc9f1404c4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/84dd78b7aefbb6b3c8f6fb64e3dbaab5aae785fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/84dd78b7aefbb6b3c8f6fb64e3dbaab5aae785fa"}]},"branch":"refs/heads/master","description":"Rebase"},"1ba7a40514850b059afbfc22da90cfea33ceaab5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-01-04 13:42:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/769010/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/10/769010/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/10/769010/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/10/769010/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/10/769010/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/10/769010/4"}}},"commit":{"parents":[{"commit":"4f09774c6b47375533ef6120d5418172137c4c2d","subject":"OVN: Deprecate the ovn_l3_mode option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4f09774c6b47375533ef6120d5418172137c4c2d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 08:14:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-04 13:42:05.000000000","tz":0},"subject":"OVN: Deprecate the vif_type option","message":"OVN: Deprecate the vif_type option\n\n... because it was already deprecated in networking-ovn and has no\neffect now[1].\n\n[1] 5be7dfeb637984f75937a21b994c1375dfa77d87\n\nChange-Id: I3139cc238036ec681d4eda18bb0aebc9f1404c4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/1ba7a40514850b059afbfc22da90cfea33ceaab5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/1ba7a40514850b059afbfc22da90cfea33ceaab5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
