)]}'
{"id":"openstack%2Fpuppet-neutron~769065","triplet_id":"openstack%2Fpuppet-neutron~master~I90a205df7817a28574e494ec7e1d497ee320a493","project":"openstack/puppet-neutron","branch":"master","topic":"xenapi","hashtags":[],"change_id":"I90a205df7817a28574e494ec7e1d497ee320a493","subject":"Deprecate parameters for XenAPI support","status":"MERGED","created":"2021-01-03 12:10:26.000000000","updated":"2021-01-07 04:36:02.000000000","submitted":"2021-01-07 04:34:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769065-xenapi","meta_rev_id":"7dcbb4e08638d94edbd738192b19ac556f723fce","_number":769065,"virtual_id_number":769065,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-07 04:34:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-01-05 15:33:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-01-05 15:47:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-01-05 15:47:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-03 14:36:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-04 20:42:39.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"CC"},{"updated":"2021-01-05 15:33:53.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-01-05 15:47:28.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"b01a995ab584f48c30c66ae1b016e8b8058160bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-03 12:10:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0f3d5d0083dd8448cde68bba9522efcc84b19eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 14:36:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54b4420c63f74820974fcc3d44ed04a8 : SUCCESS in 2m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13e8fc37469d4523b331f818e10dd261 : SUCCESS in 2m 40s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e77593b315be42719c7cbbaeeeb7cd30 : SUCCESS in 16m 58s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4fad8119c9345abaa3484eb061ed8e9 : SUCCESS in 15m 23s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53ff5c8d448242c0a710640580bcb22c : FAILURE in 17m 23s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/960da0ab741a45e9b0e4f15e5a8fb57d : SUCCESS in 52m 45s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02412e53029c484cbb50e64aa3bbecea : SUCCESS in 48m 19s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5fd5a49a589b4c3ca43f5727af55b2b2 : SUCCESS in 50m 40s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91bbe9985fa3499ca22594ee5cb524e8 : SUCCESS in 43m 12s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/114211e3f9614274881d14330e7aecf5 : SUCCESS in 1h 05m 17s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b57b93bc85484ec5a23e659860465ce2 : SUCCESS in 1h 28m 24s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c15e42a6b2be46ff928df144c47fda53 : SUCCESS in 1h 26m 55s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/eafabda162dc46938e0703ce0d602ae9 : SUCCESS in 48m 23s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/e05fe2b20c444ccf82b185fa2ba8013e : SUCCESS in 17m 29s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5d3bb35fbdd4a22835936428a396fef : SUCCESS in 15m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2647cfcb5b5d4220993edeed8edb1f01 : SUCCESS in 3m 12s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e4763d618c3f4c0b8fe91f9513fadfc3 : FAILURE in 2h 24m 44s","accounts_in_message":[],"_revision_number":1},{"id":"0d32bd31e8c96298a092161d8a7247a1a05a6e8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-03 22:20:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d3744aa78f4ea20234580b6638cc4d4e405e28b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 00:37:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16aee5a6b1584dda9e0ca5fefba22f24 : SUCCESS in 3m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c61f51a71e2e4d229381816557086201 : SUCCESS in 2m 37s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/34ac433414044e218c0be429832a05d9 : SUCCESS in 17m 41s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83b03b1676c14f8bb97b9008f72424a9 : SUCCESS in 14m 06s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55c69a1686a44d02a3d43a0014345473 : FAILURE in 17m 18s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5faead62eef400798af5ddcb6d0fe2f : SUCCESS in 1h 03m 29s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c0591296be5433e8bb066cf43e8a2d4 : SUCCESS in 51m 46s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8cf03dc21a342a390ed5956f5c3e93b : SUCCESS in 57m 38s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56c029055c6e4ed0a8d0349e0909f18c : SUCCESS in 43m 06s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e1b12d405249444eac5a6d41548132d9 : SUCCESS in 1h 20m 03s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/25152ff668d949c79becaee47e1a5648 : SUCCESS in 1h 27m 40s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ad248fc7d8344ff0b62cf79019ecaf1b : SUCCESS in 1h 17m 10s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/12f13e11c071419c98ee246c9bc48542 : SUCCESS in 1h 08m 13s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/f5f4e6bbb0e84ff797fc68e53876873b : SUCCESS in 18m 11s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3703eb5423e64b54af686ac902cf1dc9 : SUCCESS in 18m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71a0d9e3766a408cb8902f40b9919931 : SUCCESS in 3m 01s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6c475a1cb0ab469d8f64da87d361c4b1 : SUCCESS in 2h 06m 00s","accounts_in_message":[],"_revision_number":2},{"id":"33582059b1289c0c70e1713bbc28598d54e29211","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-04 13:41:07.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0bbe0aab51cfd530c01e2091bd2fcf4fe5668dd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 15:54:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e4d757bfd73435ca254c50736f4c0cc : SUCCESS in 3m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73fe9a09377a4680b7667193cf147839 : SUCCESS in 3m 25s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/df8f2bb76d004cdab074e7c437eaffe8 : SUCCESS in 17m 29s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfea6d669baa4f58836cecd558705c6c : SUCCESS in 17m 15s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9dc869f7580c49e7b5d3e9816d9c56d6 : FAILURE in 17m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49b53bd0b13e4727b73a8f1c3514e27f : SUCCESS in 52m 42s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f9fb9d64c064386bafd4c5d2faaaa1e : SUCCESS in 53m 45s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abb602fd006b4eb09fc263c3ff76ef41 : SUCCESS in 50m 38s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57cf0317a679487284dad5bb5331a3f3 : SUCCESS in 37m 25s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c2e3e2a930f144b797fdcce7492c255c : SUCCESS in 1h 29m 12s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f1936658631e452fb2b87355b407ef77 : SUCCESS in 1h 31m 45s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5190d50d453d42fbb134c0e93bdc4bec : SUCCESS in 1h 39m 13s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3eefc6f3e22540cd81e9649ec1274c71 : SUCCESS in 1h 05m 21s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/d6612be1888645eaafe40c0692a9e381 : SUCCESS in 20m 40s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a309b7528ea04b339ecab7a068d698b8 : SUCCESS in 15m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53abd92c61a74a74b8570c2b69886410 : SUCCESS in 3m 32s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/047172dbec0e4f2a8ed2577752c22b5b : FAILURE in 2h 11m 58s","accounts_in_message":[],"_revision_number":3},{"id":"d4ce4c8ef0c16b61bf48d1619d39e34a16d14f7d","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-01-04 20:42:39.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"81dba7a37a9d20d2b719fc587f36d1b57511bd08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 23:38:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a07d1f479944460181ea4d08deedec7c : SUCCESS in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c10dca36c2f84f7a8c3bed2af24e09fc : SUCCESS in 3m 05s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1e1c221848eb4208bf7bb4c71b972b2a : SUCCESS in 18m 08s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dec5763f68d449c693c5c6bf8e1dc63e : SUCCESS in 14m 20s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2525295f6f0f47c2b6c49b4ab332501a : FAILURE in 17m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb6231f3f43d4e938dd4b61d02a01a53 : SUCCESS in 1h 07m 34s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf08f29932b1439295ffc6c609457985 : SUCCESS in 1h 04m 33s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd8a83a22fc84e52a841fedef930b9df : SUCCESS in 56m 49s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/397f9e8afddf400e86b3bf4c0610eaf3 : SUCCESS in 48m 31s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d33aacd4b7234c04bb425d51cbe8631f : SUCCESS in 1h 35m 33s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7f455e74b8a94598acabea9bda010eaf : SUCCESS in 1h 37m 26s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/97181a2e40df45228f58a426fd45f5d6 : SUCCESS in 1h 27m 29s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ddc93ccf098c4fca9735fe8156cdf0a9 : SUCCESS in 1h 11m 35s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/a9c5c10ef258496097c080a212a50d8d : SUCCESS in 21m 22s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c58654852d2b4b2eb51ab14926c5a7c3 : SUCCESS in 17m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51e53ef777044071b84a9715b524b2a5 : SUCCESS in 3m 22s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7ee5987f30844038ad6760f70e0a48b4 : SUCCESS in 2h 50m 32s","accounts_in_message":[],"_revision_number":3},{"id":"8d1dc9ed135f27ebe472c9be650561de28d15a76","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-05 15:33:53.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"80bc78e308255a3799fe34da16ef84dff9e5a3d8","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-01-05 15:47:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bbc42f301769f0d75f867e16f75302019d87f037","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 15:48:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"63e3d6cf2458dde0a800e374fc06cdc78f5d68a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 19:14:06.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17a7cb1e02ce4fb6810f7f29e7b9aee8 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8fbffe824d8e451ba67c6470f4b9d564 : SUCCESS in 3m 06s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9822c640b6e747d99f4f8341801bc478 : SUCCESS in 21m 45s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/571395e3d74248ccb72d8d938ac47236 : SUCCESS in 15m 58s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6f742b7082fa419db414b2721dcab225 : SUCCESS in 1h 23m 33s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ea16d8f3090e42448239eb5559454330 : SUCCESS in 1h 20m 54s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/92decafaee034a72889012495550a735 : SUCCESS in 1h 22m 22s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b1abc9e28fd14d01a5de834a752f9d5a : SUCCESS in 59m 28s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/28c07e40e33042af8a4f4d2f5ba2e04f : SUCCESS in 20m 47s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14d6ba65f2a44ad294bad3a372568b57 : SUCCESS in 14m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d6d10fd7ef34b2da91cbd1f8119262d : SUCCESS in 6m 06s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bcfa0444fe254eb5884047bb6b6a8168 : TIMED_OUT in 3h 16m 29s","accounts_in_message":[],"_revision_number":3},{"id":"e0ef7156addda89bb528ffab7c89448202c48aea","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-06 23:09:20.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"409307d730c79642e7c62f941466632d761779b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 01:49:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b528cb85743e42d7954ccc892da73ec9 : SUCCESS in 3m 16s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22996ff792e2463ca690ed8b10fa8c78 : SUCCESS in 2m 58s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/fd0bc3a4cdce4e30b628a1dc81ad862d : SUCCESS in 19m 27s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/632336af48774ecea1a8c938246f4509 : SUCCESS in 16m 21s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/adcd1bde7baf45659ed64aa5cf6dfeb2 : FAILURE in 18m 25s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1aa2918815a34b93a9ffc74e88c0387f : SUCCESS in 54m 12s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8aba450353b4480392eb7d0e69764dfd : SUCCESS in 55m 34s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/755bc72a81604b14bbefe6e5139c9010 : FAILURE in 15m 55s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d501b5e9ef014c77b84e5c67f2d616ed : SUCCESS in 40m 20s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ff7a32f5f9a242019f900a57a23cf692 : SUCCESS in 1h 25m 49s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5dc7b40836ea4815b2c33d09409963e9 : SUCCESS in 1h 39m 08s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bad1e7b0913d406d9837d6ab4508b1a5 : SUCCESS in 1h 23m 27s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3c22a2294f8941388f567e571a7d8082 : SUCCESS in 1h 04m 14s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/63d9ef2a627145a59b4df9ed1e64ece3 : SUCCESS in 18m 32s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95c6d76f2eb5421b9f1e7184355becb9 : SUCCESS in 13m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cfdfe85d0b54442bf6645b542678920 : SUCCESS in 3m 22s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3eacb4e30450481fa9ae65e685b1f4f9 : SUCCESS in 2h 35m 14s","accounts_in_message":[],"_revision_number":3},{"id":"5da290b2263f0f92999420d4142d6691fef807f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 01:49:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f2a97dea61f4a6a57f08c6ed257f30ca6ba1b79a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 04:34:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4584fcde5464b96af7637c4daefd80f : SUCCESS in 3m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39bda922a80c4af19ea34cb1a7072647 : SUCCESS in 3m 03s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/497e85e7deb14bdd9fafb3626139d570 : SUCCESS in 20m 31s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e978b61f92be43349b592c18d14a2ebd : SUCCESS in 16m 22s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6561b31d0ac44d93bb15a030b7f596fa : SUCCESS in 1h 22m 25s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e85f90380a0c47e5a9dd75ac2a2bc77f : SUCCESS in 1h 28m 18s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a5ed5bf133164390aad70762c59ff73a : SUCCESS in 1h 28m 09s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cfd1b5fa25f341749fe89e3ae49a4ea5 : SUCCESS in 1h 03m 24s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/035e18038137409680f6883b98f8c48b : SUCCESS in 18m 53s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/561bdc221f0441f99e530108edc2e916 : SUCCESS in 13m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d66f7797479142f681ab7dca651e3659 : SUCCESS in 3m 13s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d22867f1c07b44418bfd54a95bbb2b8e : SUCCESS in 2h 40m 02s","accounts_in_message":[],"_revision_number":3},{"id":"e324dcdbf2557e9936f246b48ef3547e81562573","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 04:34:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7dcbb4e08638d94edbd738192b19ac556f723fce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 04:36:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39ab11bb0a53405bb1f4aa6b656dfe8b : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1e01fb86bc0d1fd9ba792a2cf1917c973cc35558","revisions":{"aa31e6808b22b40a3e50e266771c914c99737925":{"kind":"REWORK","_number":1,"created":"2021-01-03 12:10:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/769065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/65/769065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/769065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/769065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/769065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/65/769065/1"}}},"commit":{"parents":[{"commit":"2314c53abfdbf800de9c08dbf3af8a9215b5f93f","subject":"Add OVS bridge_mac_table_size parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2314c53abfdbf800de9c08dbf3af8a9215b5f93f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 12:09:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 12:09:23.000000000","tz":540},"subject":"Deprecate parameters for XenAPI support","message":"Deprecate parameters for XenAPI support\n\n... because it was already deprecated in neutron[1].\n\n[1] a6dbf97242caa3be646e8eb6b1502b5e59e123fd\n\nChange-Id: I90a205df7817a28574e494ec7e1d497ee320a493\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/aa31e6808b22b40a3e50e266771c914c99737925"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/aa31e6808b22b40a3e50e266771c914c99737925"}]},"branch":"refs/heads/master"},"dff146379e4bb8523c2c67b71e0cf75458d1258d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-03 22:20:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/769065/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/65/769065/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/769065/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/769065/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/769065/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/65/769065/2"}}},"commit":{"parents":[{"commit":"18c3fd2b04a4a998df15ec894e8922baf7b9df41","subject":"Disable neutron_tempest_plugin in TripleO job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/18c3fd2b04a4a998df15ec894e8922baf7b9df41"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 12:09:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 22:20:59.000000000","tz":0},"subject":"Deprecate parameters for XenAPI support","message":"Deprecate parameters for XenAPI support\n\n... because it was already deprecated in neutron[1].\n\n[1] a6dbf97242caa3be646e8eb6b1502b5e59e123fd\n\nChange-Id: I90a205df7817a28574e494ec7e1d497ee320a493\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dff146379e4bb8523c2c67b71e0cf75458d1258d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dff146379e4bb8523c2c67b71e0cf75458d1258d"}]},"branch":"refs/heads/master","description":"Rebase"},"1e01fb86bc0d1fd9ba792a2cf1917c973cc35558":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-04 13:41:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/769065/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/65/769065/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/769065/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/769065/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/769065/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/65/769065/3"}}},"commit":{"parents":[{"commit":"2314c53abfdbf800de9c08dbf3af8a9215b5f93f","subject":"Add OVS bridge_mac_table_size parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2314c53abfdbf800de9c08dbf3af8a9215b5f93f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 12:09:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-04 13:41:07.000000000","tz":0},"subject":"Deprecate parameters for XenAPI support","message":"Deprecate parameters for XenAPI support\n\n... because it was already deprecated in neutron[1].\n\n[1] a6dbf97242caa3be646e8eb6b1502b5e59e123fd\n\nChange-Id: I90a205df7817a28574e494ec7e1d497ee320a493\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/1e01fb86bc0d1fd9ba792a2cf1917c973cc35558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/1e01fb86bc0d1fd9ba792a2cf1917c973cc35558"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
