)]}'
{"id":"openstack%2Fpuppet-neutron~773373","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fstein~I1d96899de29cdd994f2a45b701923b0177a7edd8","project":"openstack/puppet-neutron","branch":"stable/stein","topic":"bug/1868569","hashtags":[],"change_id":"I1d96899de29cdd994f2a45b701923b0177a7edd8","subject":"Add ovs/igmp_snooping_enable support","status":"MERGED","created":"2021-02-01 12:01:08.000000000","updated":"2021-02-02 14:44:05.000000000","submitted":"2021-02-02 14:44:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773373-bug/1868569","meta_rev_id":"f3ba3a26e2a47e05981b8909720e34ec16d2c907","_number":773373,"virtual_id_number":773373,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-02 14:44:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"all":[{"value":1,"date":"2021-02-01 14:28:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2021-02-01 16:03:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-02-02 11:27:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-02-02 11:27:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"reviewers":{"REVIEWER":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-01 12:01:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2021-02-01 12:22:55.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-02-01 12:22:55.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-02-01 12:22:55.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-02-01 14:28:15.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2021-02-01 15:00:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-02 11:27:30.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"bc6e5c947cf588a9e2ece4fed20f9e5020737f2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-01 12:01:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"252a83ebb84212287d97ee101c21f1d2adc00603","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-01 12:14:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b6400dc324203bcd2c9e0a52554f9ce6369a63b6","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2021-02-01 14:28:15.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6b2237a8336271671c047ff1a0865f7447f04a8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 15:00:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44269c30da7249b2aa051a10313573bb : SUCCESS in 4m 11s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f2f715f364348e29df15847ae683f6a : SUCCESS in 4m 22s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5c6b4e6c753432a896cbdad6e95e818 : SUCCESS in 3m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/249b667b70f8492784974fe8cbee49da : SUCCESS in 4m 59s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/d075419a7a394ffdbf9634e72d5be831 : SUCCESS in 30m 25s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/9d67e90605aa4242ae775979067957a1 : SUCCESS in 26m 46s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12f3647f7b7144fb9d1f76704ecbb80d : SUCCESS in 13m 19s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e8f9c54e70d4edaa97c2ad3e1ad2c1c : FAILURE in 24m 22s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/652b293edbfa4c7c855b6803524ce92f : SUCCESS in 1h 04m 20s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b926b49fd9fb4d388d81d2c191d5042a : SUCCESS in 1h 01m 42s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/20245b534eb64cb0b1f9456c6e04eaf0 : SUCCESS in 1h 00m 28s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/618604858742440fa6a4e9baa8a34994 : FAILURE in 1h 21m 42s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/326bd80857c94976bb647b88d9bcb2a4 : FAILURE in 28m 36s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/beb4c454af9c4b3ca3528edfc9b564aa : FAILURE in 27m 38s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5afb0cf963964c8c9eb1ced59c0984a6 : FAILURE in 24m 13s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7b7cb5030ffd44ce8390c819e38cd10d : FAILURE in 16m 30s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/51095b96364447f8ace7b4fea20b748f : SUCCESS in 16m 33s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66458387685245779db325a839f88197 : FAILURE in 6m 23s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0bec256664fb41a0ba5a29f333731973 : SUCCESS in 2h 39m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"489bf2d5da6b19d9dbd8c5aa747a202cd7977772","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-02-01 16:03:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4cb1fa76319a8d09a1e3b6054c21e769ff136eb1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-02-02 11:27:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ddca3ea145304ffbfc07f7fc88b4b0407c142ab3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 11:30:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"24a9dadc2d8e8800758e1b5694890ee2e4bec7a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 14:44:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/752a4d8e9c7a483b80db30a826017623 : SUCCESS in 4m 33s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95645658f618446ca7ccec5c2d768e7a : SUCCESS in 4m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf2e8231f4e14213869f321ee848ccba : SUCCESS in 4m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41d713089d1745d6bb3c7f00470081fd : SUCCESS in 3m 59s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/cc57da2748ea4f9989b8bc07e91ed3b9 : SUCCESS in 35m 24s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2ca4fe5ba8b44c76ae244bedffb87f7a : SUCCESS in 31m 09s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3a8937a382c43bd98de0ba9a4cf4444 : SUCCESS in 14m 42s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/255aa6dd864b4a99bc91b8f1baae7969 : SUCCESS in 59m 25s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/fc61a585b52e47e291233f9dcdd14133 : SUCCESS in 1h 03m 15s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2baa40d346de4ac7bb86cf8d3bcb6a64 : SUCCESS in 1h 00m 20s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/df750dabd1ca4b63afdeb93e3147eef1 : SUCCESS in 19m 20s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0aea544576094b11a16f47a2d6cd9e6b : FAILURE in 7m 42s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3138b730ba5d4b36bdcec0ffba0ae803 : SUCCESS in 3h 06m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f3ba3a26e2a47e05981b8909720e34ec16d2c907","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 14:44:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4d5fd04844b58c84be7428a5bfcaab1b2f4d7eab","revisions":{"806c9b6964a6e85fe4af62412dc724a17fa7b037":{"kind":"REWORK","_number":1,"created":"2021-02-01 12:01:08.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/73/773373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/73/773373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/73/773373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/73/773373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/73/773373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/73/773373/1"}}},"commit":{"parents":[{"commit":"c2f22df9fa41cfc632b22d3b6dad542f8a66f669","subject":"Add agent/explicitly_egress_direct support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c2f22df9fa41cfc632b22d3b6dad542f8a66f669"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-12 11:32:13.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-01 12:00:47.000000000","tz":0},"subject":"Add ovs/igmp_snooping_enable support","message":"Add ovs/igmp_snooping_enable support\n\nThis patch is adding support to the \"ovs/igmp_snooping_enable\"\nconfiguration option in the neutron.conf and ml2_conf.ini configuration\nfiles.\n\nBoth, ML2/OVS and ML2/OVN reuse the same configuration option to enable\nIGMP but, for ML2/OVN this option could can be set in the neutron.conf\n(since it will be consumed on the controllers) and and for ML2/OVS it\u0027s\nconfigured for the agents (ml2_conf.ini)\n\nConflicts:\n      manifests/agents/ml2/ovs.pp\n      manifests/server.pp\n      spec/classes/neutron_agents_ml2_ovs_spec.rb\n\nRelated-Bug: #1868569\n\nChange-Id: I1d96899de29cdd994f2a45b701923b0177a7edd8\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 382352a893c2a693eb10329365357282d68d75e4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/806c9b6964a6e85fe4af62412dc724a17fa7b037"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/806c9b6964a6e85fe4af62412dc724a17fa7b037"}]},"branch":"refs/heads/stable/stein"},"4d5fd04844b58c84be7428a5bfcaab1b2f4d7eab":{"kind":"REWORK","_number":2,"created":"2021-02-01 12:14:13.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/73/773373/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/73/773373/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/73/773373/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/73/773373/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/73/773373/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/73/773373/2"}}},"commit":{"parents":[{"commit":"c2f22df9fa41cfc632b22d3b6dad542f8a66f669","subject":"Add agent/explicitly_egress_direct support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c2f22df9fa41cfc632b22d3b6dad542f8a66f669"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-12 11:32:13.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-01 12:13:44.000000000","tz":0},"subject":"Add ovs/igmp_snooping_enable support","message":"Add ovs/igmp_snooping_enable support\n\nThis patch is adding support to the \"ovs/igmp_snooping_enable\"\nconfiguration option in the ml2_conf.ini configuration file.\n\nConflicts:\n      manifests/agents/ml2/ovs.pp\n      manifests/server.pp\n      spec/classes/neutron_agents_ml2_ovs_spec.rb\n\nRelated-Bug: #1868569\n\nChange-Id: I1d96899de29cdd994f2a45b701923b0177a7edd8\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 382352a893c2a693eb10329365357282d68d75e4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4d5fd04844b58c84be7428a5bfcaab1b2f4d7eab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4d5fd04844b58c84be7428a5bfcaab1b2f4d7eab"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
