)]}'
{"id":"openstack%2Fpuppet-neutron~774073","triplet_id":"openstack%2Fpuppet-neutron~master~I56ece3f7af97bc28570b75ddbfeaaef6c35086c0","project":"openstack/puppet-neutron","branch":"master","hashtags":[],"change_id":"I56ece3f7af97bc28570b75ddbfeaaef6c35086c0","subject":"Add puppetlabs-apache to dependencies","status":"ABANDONED","created":"2021-02-04 12:01:11.000000000","updated":"2021-03-18 13:50:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ab513247c46732385c93552dc511f5bd3cb2cea7","_number":774073,"virtual_id_number":774073,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-18 13:43:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":-1,"date":"2021-02-05 09:59:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-04 12:11:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-05 09:59:16.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"cf6ff4f6ad8f36f399420ae3de256859d185c8b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-04 12:01:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f97b493325790a17e9044e3e2719d364fd54a551","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 12:11:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/dd977a695e4647a2b58f185322ccbf97 : SUCCESS in 2m 06s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f6255e8285f487bb2783a376e36e1ae : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f06c5c5cf2fdd0f083cd23cffe7c2b72eddd5068","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-02-05 09:59:16.000000000","message":"Patch Set 1: Code-Review-1\n\nMatch minimum version stated in puppet-openstacklib?","accounts_in_message":[],"_revision_number":1},{"id":"6b63249cf5308e8cf76d5012274f504d9db8249b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-07 07:12:19.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e Match minimum version stated in puppet-openstacklib?\n\nI used the version stated in puppet-nova.\nIt seems that most of puppet modules except for puppet-openstacklib and puppet-hroizon have puppetlabs/apache\u003e\u003d3.0.0 in its requirement. Should we also fix them ?","accounts_in_message":[],"_revision_number":1},{"id":"282ded72f3f8d24b2d036a0747d4b1c1d594b6ce","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-02-10 09:16:18.000000000","message":"Patch Set 1:\n\nMaybe lets bump them all while we are at it?","accounts_in_message":[],"_revision_number":1},{"id":"38aceead4b1e5673bbbf36c9592f1033bd11be6b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-18 13:31:51.000000000","message":"Topic missing-dependency removed","accounts_in_message":[],"_revision_number":1},{"id":"b34a4ce301a1ceb37cc573825e8e135b5041299c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-18 13:31:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"1f32156df4568a2aa960a8a393087c98f4816f04","tag":"autogenerated:gerrit:restore","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-18 13:33:00.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"9d2aceaba809f772fe006501dbe74212a20d3646","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 13:43:18.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"ab513247c46732385c93552dc511f5bd3cb2cea7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-18 13:50:15.000000000","message":"Abandoned\n\nI made a different commit with the comment addressed https://review.opendev.org/c/openstack/puppet-neutron/+/781465","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2be0c503623010094e0d3a8b6645daf2d70ac02","revisions":{"a2be0c503623010094e0d3a8b6645daf2d70ac02":{"kind":"REWORK","_number":1,"created":"2021-02-04 12:01:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/774073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/73/774073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/73/774073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/73/774073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/73/774073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/73/774073/1"}}},"commit":{"parents":[{"commit":"135713be4d3268e1c0bd5e6ef2d011d0f3b96967","subject":"Remove duritong/sysctl dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/135713be4d3268e1c0bd5e6ef2d011d0f3b96967"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-04 12:00:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-04 12:00:10.000000000","tz":540},"subject":"Add puppetlabs-apache to dependencies","message":"Add puppetlabs-apache to dependencies\n\n... because it is required to run neutron api by apache+wsgi.\n\nChange-Id: I56ece3f7af97bc28570b75ddbfeaaef6c35086c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a2be0c503623010094e0d3a8b6645daf2d70ac02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a2be0c503623010094e0d3a8b6645daf2d70ac02"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
