)]}'
{"id":"openstack%2Fpuppet-neutron~780467","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fussuri~I3d9f0b48a72f82f7d1a86aae94c1f5248b729987","project":"openstack/puppet-neutron","branch":"stable/ussuri","topic":"rp-bw-default","hashtags":[],"change_id":"I3d9f0b48a72f82f7d1a86aae94c1f5248b729987","subject":"Add support for the resource_provider_hypervisors parameter","status":"MERGED","created":"2021-03-16 09:22:49.000000000","updated":"2021-03-18 17:38:42.000000000","submitted":"2021-03-18 17:34:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780467-rp-bw-default","cherry_pick_of_change":780085,"cherry_pick_of_patch_set":1,"meta_rev_id":"f96e5ae43c68a831384dbda245a913647a335857","_number":780467,"virtual_id_number":780467,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-18 17:34:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2021-03-17 12:25:33.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2021-03-17 13:47:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-03-18 14:40:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-03-18 14:40:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-16 11:58:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-17 12:25:33.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2021-03-17 13:47:38.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-03-18 14:40:52.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"728ccb9c2c14fe109b42f5d315cefeeea3ff64f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-16 09:22:49.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"40c0bb06ecd7edb49d92dfe7fdc0e3d9b8b95159","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-16 09:23:23.000000000","message":"Topic rp-bw-default-stable/ussuri removed","accounts_in_message":[],"_revision_number":1},{"id":"7ea991d8bb041b16af3c36ee4cd32a7692c29b95","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-16 09:23:28.000000000","message":"Topic set to rp-bw-default","accounts_in_message":[],"_revision_number":1},{"id":"cf869f8472e9e4c634e8522393153d6223f6e13a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 11:58:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73d97b71e2fa40809e57f92caae8e847 : SUCCESS in 6m 43s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e2331a021394d3a82c222b9e46f860c : SUCCESS in 6m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ae33d6d13ff4c93b7106754b8eed1c2 : SUCCESS in 4m 26s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/48de45f1b4ac4cd0b367b70be5e6a855 : SUCCESS in 14m 33s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/4ad985317eab446fa7e83ccf3f208379 : SUCCESS in 14m 36s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/375ac2d9f6d9448c96ed0ac5fc0178d5 : SUCCESS in 11m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb440842be724894981da70efbe4ae88 : FAILURE in 19m 48s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d1bfdf4ce83a4733a42f1171909dc93e : FAILURE in 36m 28s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7751baaed0d490caf1eb34406aefde4 : FAILURE in 34m 32s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e7ce0976e634eb09bf98c4c4d3566aa : FAILURE in 35m 06s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a83c36719b6c45f1b95bf1cf29a7820b : FAILURE in 29m 19s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/04e12f4a24af4e4e92b6ffb21f6250c1 : SUCCESS in 58m 56s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/66760ca5b6374b939b518311e03c3a07 : SUCCESS in 1h 02m 14s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e6ac6ef79ab04c218e951d8ef824c912 : SUCCESS in 1h 02m 09s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8f972cbc54b74bfbb1a5a63ef47195e5 : SUCCESS in 43m 22s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/110752ef27e64e0caa914fb40c7b0773 : SUCCESS in 22m 30s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/933070dacaed4085a224dbccd51521b6 : SUCCESS in 17m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2300c1063a45492693bcaa7cd1c9d202 : SUCCESS in 5m 01s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/859270833eac42f294f7fcbce58533c5 : SUCCESS in 2h 33m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1b93cf45b24146525c5c012da8edfd8a7b1af64c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2021-03-17 12:25:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b415bfd0a5c31882e5e3a04963a12d52687fb939","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-03-17 13:47:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9e1176857e3e8ec16f2f352653ae65eeed845003","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-18 14:40:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"67250e4a582d632b2251b7f038c2a047bb1715f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 14:51:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"19a0c56f10f4d7c5c843eaa349c9c54822b7b391","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 17:34:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d27bf73add5d4a7fb2eeebc42fb22a68 : SUCCESS in 5m 21s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86fe4be0db844e6cb331a467421dab46 : SUCCESS in 5m 59s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72bba85816f64d638cecaf037647b33c : SUCCESS in 4m 14s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/ddf8d97a1a9f4aca8fb38e39c14cc844 : SUCCESS in 19m 17s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/0a2b7d2ef8594d4486bc523e92437dfd : SUCCESS in 19m 47s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7d3b1a0ef894dbf8e48e6e5461ad98f : SUCCESS in 13m 29s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dfd246027e364424a0f5d0903a79b21a : SUCCESS in 49m 45s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4d10378598aa4271a4cf3918cedd9c16 : SUCCESS in 1h 03m 12s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d32c8f0259684910999d6e451452ad16 : SUCCESS in 1h 09m 56s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3113cc99fa474cd0b4e423fa487d93f6 : SUCCESS in 48m 30s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/2502cd63965c4c589c7ca6b90fb1381c : SUCCESS in 26m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/355f47253b624a18a0631903c340ab08 : SUCCESS in 5m 47s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fb79c9036bc041a7a6cfb2b415757453 : SUCCESS in 2h 36m 49s","accounts_in_message":[],"_revision_number":1},{"id":"0a6dec4623d33d8e3c017ae213d21613bd335e75","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 17:34:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f96e5ae43c68a831384dbda245a913647a335857","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 17:38:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57549fa1223b4ad4a27f2dae26121ae7 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"192fc513dc124ae4d619b6e02bf7348fcf42bc90","revisions":{"192fc513dc124ae4d619b6e02bf7348fcf42bc90":{"kind":"REWORK","_number":1,"created":"2021-03-16 09:22:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/780467/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/67/780467/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/780467/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/780467/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/780467/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/67/780467/1"}}},"commit":{"parents":[{"commit":"e77d34f037559fee522ac8811e07b3416f94e26e","subject":"Ensure resource_provider_bandwitdhs is unset by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e77d34f037559fee522ac8811e07b3416f94e26e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-17 08:04:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-16 09:22:49.000000000","tz":0},"subject":"Add support for the resource_provider_hypervisors parameter","message":"Add support for the resource_provider_hypervisors parameter\n\nChange-Id: I3d9f0b48a72f82f7d1a86aae94c1f5248b729987\n(cherry picked from commit 830c66cccaf1446ff59081630977033dbdc72b6a)\n(cherry picked from commit 7c0ffac91c3207a94d6c53ded2fe860a92a3572a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/192fc513dc124ae4d619b6e02bf7348fcf42bc90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/192fc513dc124ae4d619b6e02bf7348fcf42bc90"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
