)]}'
{"id":"openstack%2Fpuppet-neutron~780758","triplet_id":"openstack%2Fpuppet-neutron~stable%2Ftrain~I3d9f0b48a72f82f7d1a86aae94c1f5248b729987","project":"openstack/puppet-neutron","branch":"stable/train","topic":"rp-bw-default","hashtags":[],"change_id":"I3d9f0b48a72f82f7d1a86aae94c1f5248b729987","subject":"Add support for the resource_provider_hypervisors parameter","status":"MERGED","created":"2021-03-16 09:28:24.000000000","updated":"2021-03-18 17:36:47.000000000","submitted":"2021-03-18 17:35:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780758-rp-bw-default","meta_rev_id":"9a56cd8c122519cf87cbcaafe832a92d413eaac4","_number":780758,"virtual_id_number":780758,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-18 17:35:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-03-17 13:47:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-03-18 14:41:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-03-18 14:41:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-16 10:37:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-17 13:47:13.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-03-18 14:41:10.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"3935b595680cb5877b114f840e747bc51ac9c7a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-16 09:28:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6dfc9a91511959a2f324a30d354cf853e475c5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 10:37:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c59663b695a549bea4824afd89fef3db : SUCCESS in 5m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1de89b30365b462ca94d3f75e51120f5 : SUCCESS in 5m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b173cb8944f042d7a46a61c2a8d387b9 : SUCCESS in 5m 19s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e62b21ae80c340ffb965128403b76065 : SUCCESS in 33m 55s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/e6fa292276de47b0ae5613927123b392 : SUCCESS in 17m 45s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/6267603ac06349a8a95697b1a1e31cb2 : SUCCESS in 16m 07s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/609a5dc540644a47b3c1e045b3b7e664 : SUCCESS in 14m 53s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ded2d868dfb46eea9be78024fbf8864 : FAILURE in 25m 02s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/ce7930a203ad4220a827b798a56228ab : SUCCESS in 57m 25s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/5db8dd8ad10d4ad3a30c5664f0b18d49 : SUCCESS in 1h 01m 55s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/1387c7aaf28747178b8b41a529d1793f : SUCCESS in 1h 00m 00s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/0d375c15322c44309eaafc5ad823d5e9 : SUCCESS in 47m 27s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/c91e6821001e4cfab7b9fef6a7414947 : FAILURE in 31m 41s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/906825bd90f94d618bab159859f4ea1d : FAILURE in 34m 36s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb8baef844e34d5cb51ad07788b8f1d1 : FAILURE in 28m 05s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/3e48f1e377344319b912ef3609b7b6a5 : FAILURE in 22m 27s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5e8607dfdd9a4ecc8826fc48d3dc44c2 : SUCCESS in 54m 19s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2d77ea0aa68f44a2b88eb7bd9da40ce5 : SUCCESS in 57m 20s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f64edea178a74c06979158f528b7dd68 : SUCCESS in 58m 15s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/148c69579d1b453da61403f8aaddfd33 : SUCCESS in 41m 44s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8342461fc4c044f08982565d3d4426b9 : SUCCESS in 20m 29s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2dccc4331ad40bd88a236c7fa8b6c40 : FAILURE in 9m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c82b67517e234f9a90e66dfcc7db1377 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"5c705cc0fdc52ed07e1ee161ed670b922e57d962","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-03-17 13:47:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4ee7861ed388a5038332abec1e8bebf236064750","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-18 14:41:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3da20719e6329e8d619c3f09b63762491617b87b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 14:51:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aef54ae7ec52806f311350db061a0ce966fd985d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 17:35:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da1c54bb8c994108b57756e2327fbba1 : SUCCESS in 8m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28005891fecd4121a0a9946b9996e997 : SUCCESS in 5m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53317a7ceb4643379056d4d15ca1448a : SUCCESS in 6m 36s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/274603d562384932b8d17efd1d2b3ac3 : SUCCESS in 35m 24s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d7d5965a36f403b8937e2ff372bf416 : SUCCESS in 16m 29s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/886f264e4018492db54b01e22db9c626 : SUCCESS in 56m 58s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/186b616e71f34eb493ccb5f8e6c99356 : SUCCESS in 57m 26s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/12812c4e03f04564873822df9ad1c2dc : SUCCESS in 1h 01m 31s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/6505d1261eae4530b502a4e95eb65617 : SUCCESS in 48m 17s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e75c1feca61e4522905aa9e88177c3ee : SUCCESS in 20m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76719a8319654298ab93f9041ae562d0 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"5bf8d7e9494ca2d4d46e5ad03d9169dd91912a23","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 17:35:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9a56cd8c122519cf87cbcaafe832a92d413eaac4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 17:36:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3fdd12fbcc8482cb29f782aaa7891fd : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c774dcb85ea936cec5fd7827b49eb9718764ee8c","revisions":{"c774dcb85ea936cec5fd7827b49eb9718764ee8c":{"kind":"REWORK","_number":1,"created":"2021-03-16 09:28:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/780758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/58/780758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/58/780758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/58/780758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/58/780758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/58/780758/1"}}},"commit":{"parents":[{"commit":"89e5b0d030305ba322799a3a3951cd218ca73f9b","subject":"Ensure resource_provider_bandwitdhs is unset by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/89e5b0d030305ba322799a3a3951cd218ca73f9b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-17 08:04:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-16 09:27:32.000000000","tz":540},"subject":"Add support for the resource_provider_hypervisors parameter","message":"Add support for the resource_provider_hypervisors parameter\n\nConflicts:\n\tmanifests/agents/ml2/ovs.pp\n\nChange-Id: I3d9f0b48a72f82f7d1a86aae94c1f5248b729987\n(cherry picked from commit 830c66cccaf1446ff59081630977033dbdc72b6a)\n(cherry picked from commit 7c0ffac91c3207a94d6c53ded2fe860a92a3572a)\n(cherry picked from commit 192fc513dc124ae4d619b6e02bf7348fcf42bc90)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c774dcb85ea936cec5fd7827b49eb9718764ee8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c774dcb85ea936cec5fd7827b49eb9718764ee8c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
