)]}'
{"id":"openstack%2Fpuppet-neutron~786993","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fussuri~I0650085d2aa2c3492b175c565976705c977fd806","project":"openstack/puppet-neutron","branch":"stable/ussuri","topic":"fix-l3-agent-extensions-list-stable/wallaby-stable/victoria-stable/ussuri","hashtags":[],"change_id":"I0650085d2aa2c3492b175c565976705c977fd806","subject":"Fix L3 agent\u0027s extensions list","status":"MERGED","created":"2021-04-20 00:20:43.000000000","updated":"2021-04-27 07:32:00.000000000","submitted":"2021-04-27 07:32:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786993-fix-l3-agent-extensions-list-stable/wallaby-stable/victoria-stable/ussuri","meta_rev_id":"9d52844404eaa74382e94fa407671042ed53abd5","_number":786993,"virtual_id_number":786993,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-27 07:32:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-04-27 04:52:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-04-26 15:48:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-04-27 04:52:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-20 00:20:43.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-04-20 03:40:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-22 15:17:59.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"a66e8611731b64af388b8b0ec061be441dfcad2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-20 00:20:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"953eece457dfae410042d139e2ff726dfcb3aa50","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-20 00:22:23.000000000","message":"Topic fix-l3-agent-extentions-list removed","accounts_in_message":[],"_revision_number":1},{"id":"a9db6d1ebe00839ae26a6e549049b1d015b12258","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-20 00:22:26.000000000","message":"Topic set to fix-l3-agent-extensions-list","accounts_in_message":[],"_revision_number":1},{"id":"c74d6eb07cc6abb6e272a9b12796b9201fb534e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 03:40:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7043dba048948ffa51b25b6cde78f1e : SUCCESS in 3m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e41f1450d2e741ad89a38ff73974bde3 : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f172ea8dc2549718db8c21e3e4d79ae : SUCCESS in 3m 20s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/cd3940cb4fbf47ebac4f230753245491 : SUCCESS in 10m 59s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/13f7058ae6714cb495d0ffff25f39f40 : SUCCESS in 11m 22s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d7bab6821404cdcb5f6798e91e8f683 : SUCCESS in 9m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7c7f0c122c54588b3c0b878a35973c3 : FAILURE in 10m 21s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/19df6dcf5aad4102a6ccca5e7e38a06e : SUCCESS in 51m 41s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/719e4767a533493085a54a6cc30dd36d : SUCCESS in 38m 26s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47abefc706c24efc8be5ffa048cfedad : SUCCESS in 37m 48s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/53b94810a33847198b9cf56a8310ec62 : SUCCESS in 37m 34s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3e5dffce63b641e189b88b4d386f2d42 : FAILURE in 52m 26s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e2d0ffb59f854bc48848db5512f087c5 : SUCCESS in 1h 03m 10s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c8ad0137fa054ae0bdb987decd802f40 : SUCCESS in 48m 11s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b5a14123e062429eb11477e7c0dbe4ac : SUCCESS in 33m 42s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/ad605ebbda734e9e850b7a71a552ada4 : SUCCESS in 19m 14s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/15403debb5b149e1804ee94f1c060995 : FAILURE in 21m 05s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e1c85cbbe6c4a1ea5cb9ab9e56cbaeb : SUCCESS in 15m 14s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/520107b582c84c758665aef15a53e7ef : TIMED_OUT in 3h 14m 05s","accounts_in_message":[],"_revision_number":1},{"id":"4af319e4c79ede33402e7e9be4ffe1214485f396","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-20 17:14:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eee7cc55bee820bbfb6302db8d3a299018766f25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 20:07:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73f44a7fe7174f14b3295eeffb6e0bd1 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab21d06fff964fe2afa72c6c1fcb1c76 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5d51c8d69f64476be5debb49e4e14f3 : SUCCESS in 4m 08s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/ec837adac91a4e298eef35a51d0f64e6 : SUCCESS in 14m 40s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/992e63f1e8fd4e708764d22c2c29c797 : SUCCESS in 20m 06s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf58b49dea4d425cad59de0cf7909b7b : SUCCESS in 14m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78747d6a19cd483cb1e6299c1a5cb117 : FAILURE in 15m 52s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/9ec7d29ffdfd4785ae1dca49328ca101 : SUCCESS in 55m 00s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18d9fd8f45664cd38c33c9978d8d5402 : SUCCESS in 50m 15s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74319b2afe6c451c849fa7219999fabd : SUCCESS in 45m 40s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8b74a8de2858474aab0442e89beea9fd : SUCCESS in 37m 24s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e05d24d63e53456c9ee95a59b7b4265d : SUCCESS in 49m 10s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ac1ecbd08d734fedb64c3374523c5ad4 : SUCCESS in 1h 08m 05s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/aa1dfacf87964ee8ac01fb9f009e2c12 : SUCCESS in 1h 03m 32s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7de59b78f28441ad80e657833a408caf : SUCCESS in 45m 20s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/2af12d69b24044769636a29748d5d9ae : SUCCESS in 23m 57s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/097ecdf3fdca4878b07a9b0022f0ff34 : FAILURE in 21m 36s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25f05e43e8d04714bf9628075a7560bd : SUCCESS in 17m 51s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bf6bc7e288f14b6b8691505ab35e5bf3 : SUCCESS in 2h 40m 41s","accounts_in_message":[],"_revision_number":1},{"id":"eaac00d562bfcac4150936f85654de87555338ee","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-04-22 15:17:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cb4eeeb3c10ae04dabb5c3550cba2fac0172d73b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-26 12:08:30.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"f5c66409a36aada82b2986cd6d33bcad76ddafd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-26 12:10:59.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"040258cdbf4f224fe560d20107d49e7d71464070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 14:46:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57a93fb38603405e9d74593a45ba0149 : SUCCESS in 3m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a02254ba06348ef8a6b7494bdf868cc : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/adbcabd2d4d7404cb25e668b6d172c7d : SUCCESS in 3m 23s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/6ea6596af99446d1aa6b4eaf4c913c4f : SUCCESS in 12m 38s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/21a185092e7a48f78cbd04bad1d0ac49 : SUCCESS in 13m 32s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b41da3f278c94c6c82c178752cfe2a13 : SUCCESS in 13m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c993522a8b54cd09cf5db219d67cc14 : FAILURE in 15m 59s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/47de7158782247d4a6e3e24ad6c8aa5e : SUCCESS in 53m 48s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba2af7fcfeac4ebb83023fa7e078e22d : SUCCESS in 1h 01m 13s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27e9be1e80e646ccba4aa24e3f38751e : SUCCESS in 43m 37s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/94d181e87dbb4cffa90f5572f522907e : SUCCESS in 52m 03s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/13cc6e39c01f47809cf6fd7b8b0a0345 : SUCCESS in 1h 00m 43s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5e4bc953cc304b09b0d8d51184c83f6f : SUCCESS in 59m 33s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c9544f29cb424ccd91d3390b64db053d : SUCCESS in 54m 46s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dcb9636219fc4086aee18156490ca9d9 : SUCCESS in 49m 26s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/acf4e391c021453087907a3f1f06b42a : SUCCESS in 21m 15s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/ce53caf1fb6345d0813837397e9087c1 : FAILURE in 19m 16s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9536f4653d6e4231a1ab2efa40a1a380 : SUCCESS in 15m 26s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d813e0cc053f460c9e187e3a657503bd : SUCCESS in 2h 33m 54s","accounts_in_message":[],"_revision_number":2},{"id":"87539a672745619925e95e8157cad3690f598be0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-04-26 15:48:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8729b72414acb23179e7eeaebf3bd93b4000907d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-27 04:52:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ace09e71dcaa3af660a47c96b05b96211e7f0eb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 04:53:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a15720d3daa45655a61ee3a61bfc6a7986162719","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 07:32:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/802acb1c93a049fca39790e726a24507 : SUCCESS in 3m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7101e8bd211f4611beb2692276ffafc1 : SUCCESS in 2m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36536584fad5400ca2fac226f47d0c99 : SUCCESS in 3m 18s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/fff8eeec964449c8a13e99f08ee20e60 : SUCCESS in 11m 17s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/238b43fdd739484193c445d75cb7328f : SUCCESS in 11m 21s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50a98b7aa6c449a8aa8645100ed9c216 : SUCCESS in 9m 37s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f8644b527f654590b31ab9a57a43b4a6 : SUCCESS in 56m 47s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c5a531a332b843379d6f3a28f44ecd45 : SUCCESS in 1h 15m 55s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3844fef3c3ed4028a3cf6a8e8f4e6aba : SUCCESS in 1h 07m 16s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0d06e6e4ece54a8fa1af8aa5296ca284 : SUCCESS in 43m 28s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/2df1867c97344af1813c3a594188e619 : SUCCESS in 22m 15s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/72e616eb81064cdc8cad18b4a55475f4 : SUCCESS in 2h 37m 54s","accounts_in_message":[],"_revision_number":2},{"id":"9d52844404eaa74382e94fa407671042ed53abd5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 07:32:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"82324318ca5af337c4878588554965872d458ce4","revisions":{"a6990a12a62b331298ea54cf9bf109f420fd0e2d":{"kind":"REWORK","_number":1,"created":"2021-04-20 00:20:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/786993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/93/786993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/786993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/786993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/786993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/93/786993/1"}}},"commit":{"parents":[{"commit":"6bf094faa02bf56e20ffac580867891fac59fbc0","subject":"Prepare a new release for ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6bf094faa02bf56e20ffac580867891fac59fbc0"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-13 13:48:48.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-20 00:20:23.000000000","tz":540},"subject":"Fix L3 agent\u0027s extensions list","message":"Fix L3 agent\u0027s extensions list\n\nL3 agent\u0027s extensions list should be passed as an array\nto the neutron::agents::l3 class and then it should be converted\nto the comma separated string in the L3 agent\u0027s config.\nThis patch changes expected type of the variable and\nadds that such conversion from an array to string.\n\nConflicts:\n\tmanifests/agents/l3.pp\n\nChange-Id: I0650085d2aa2c3492b175c565976705c977fd806\n(cherry picked from commit d29611f08a7dd0a41e0495d2c1678eb37d6b5a66)\n(cherry picked from commit 27619a9fc01a35865c64cb345e752c9f756e4811)\n(cherry picked from commit f63aeb672dc75c78cb8c4f9c34b83360b3435d60)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a6990a12a62b331298ea54cf9bf109f420fd0e2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a6990a12a62b331298ea54cf9bf109f420fd0e2d"}]},"branch":"refs/heads/stable/ussuri"},"82324318ca5af337c4878588554965872d458ce4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-04-26 12:10:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/786993/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/93/786993/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/786993/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/786993/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/786993/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/93/786993/2"}}},"commit":{"parents":[{"commit":"6bf094faa02bf56e20ffac580867891fac59fbc0","subject":"Prepare a new release for ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6bf094faa02bf56e20ffac580867891fac59fbc0"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-13 13:48:48.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-26 12:10:43.000000000","tz":540},"subject":"Fix L3 agent\u0027s extensions list","message":"Fix L3 agent\u0027s extensions list\n\nL3 agent\u0027s extensions list should be passed as an array\nto the neutron::agents::l3 class and then it should be converted\nto the comma separated string in the L3 agent\u0027s config.\nThis patch changes expected type of the variable and\nadds that such conversion from an array to string.\n\nConflicts:\n\tmanifests/agents/l3.pp\n\nChange-Id: I0650085d2aa2c3492b175c565976705c977fd806\n(cherry picked from commit d29611f08a7dd0a41e0495d2c1678eb37d6b5a66)\n(cherry picked from commit db151245184f2188e8e4ef8be6b5418eba4e926b)\n(cherry picked from commit 9029fe04cc2f504930613000f2169befc3540255)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/82324318ca5af337c4878588554965872d458ce4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/82324318ca5af337c4878588554965872d458ce4"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
