)]}'
{"id":"openstack%2Fpuppet-neutron~790548","triplet_id":"openstack%2Fpuppet-neutron~master~Ibd4d25b1f4e658667a522db5dd450fa47867e92f","project":"openstack/puppet-neutron","branch":"master","topic":"puppet-lint-latest","hashtags":[],"change_id":"Ibd4d25b1f4e658667a522db5dd450fa47867e92f","subject":"Fix lint errors with the latest lint packages","status":"MERGED","created":"2021-05-11 03:18:01.000000000","updated":"2021-05-12 03:58:12.000000000","submitted":"2021-05-12 03:58:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790548-puppet-lint-latest","meta_rev_id":"e4ec8309528ef68dc05bc7e52d4b03b8d565718c","_number":790548,"virtual_id_number":790548,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-12 03:58:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-05-11 09:43:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-05-11 13:38:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-05-11 13:38:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-11 06:16:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-11 09:02:43.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2021-05-11 09:02:43.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2021-05-11 09:02:43.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-05-11 09:02:43.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"8e3408ff5a94975bd6d1c94f4f64fe58c337b12d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-11 03:18:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab36ef11a58d23ff9c2690de3f494ddd5f68c0c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 06:16:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0db59651371a424ca1e576b27bdd0f43 : SUCCESS in 3m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/135fd3496ae2473d83441e675c5ae9b0 : SUCCESS in 3m 37s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9cc0c210fff431887426f3c348ba534 : SUCCESS in 3m 29s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/086d9a4c6ccd4696be1cf3b9c6278a10 : SUCCESS in 16m 46s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/7312d8c16d4a45f98ba2c26e024d8b3d : SUCCESS in 17m 07s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acd10c91814642088f44afb6ce49e1ec : SUCCESS in 14m 14s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/803b13c52fa84899a4cf566a27026583 : SUCCESS in 15m 05s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57f0132a3c1e4e7d9ff97f0c6d0d4e11 : SUCCESS in 14m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6b059ba31be447eb55b91beab6b6981 : SUCCESS in 48m 40s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1404516bb154ad6b6298c88e0bb3401 : SUCCESS in 51m 54s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c476059a1a104c14a4c3ea91d4b5b638 : FAILURE in 29m 08s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39b1b6a293ef435db56658a5de60a026 : SUCCESS in 37m 53s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/22a502aa5ba549a0acc614fff253be60 : SUCCESS in 21m 25s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f50640dba6ce44589bfa25a4edfff71d : SUCCESS in 1h 14m 21s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/80f97a7f2227435583a167dbc79669e2 : SUCCESS in 1h 39m 24s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/464bb32f42e749ada7ac8c0b9087845a : SUCCESS in 1h 19m 59s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/227531956e684494ba89a57662470c04 : SUCCESS in 1h 02m 57s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/276d98a0029c4a49885a4865441ca408 : SUCCESS in 19m 52s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa460b12b2df4d6e9bbe25eac779529f : SUCCESS in 1h 13m 53s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0092ab977bb74900b23d01c82ad18b42 : SUCCESS in 1h 10m 56s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/15501013976e42f59e7ca9e7cd759161 : SUCCESS in 1h 09m 31s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a1af27429e2a4cc18babccf675af2450 : SUCCESS in 1h 01m 46s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f37b7921b99e4161aa463b3c4db6dbf6 : SUCCESS in 20m 37s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fda26bd483f44ddf94c1b0636d858363 : SUCCESS in 14m 45s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ddaea2aa434411fb54708cf3cdec1b5 : SUCCESS in 20m 23s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93320fc1560141e6ae41efd350893806 : SUCCESS in 17m 00s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0783b4ba68a0422ebce97d42cb7403b8 : SUCCESS in 16m 41s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f07d2ececdfd429f902f47877f5a3837 : SUCCESS in 2h 53m 55s","accounts_in_message":[],"_revision_number":1},{"id":"37ff62d33021e138dea428b56a24a696095cd0e6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-11 09:10:13.000000000","message":"Topic lint removed","accounts_in_message":[],"_revision_number":1},{"id":"360dbba2ff1c229aab552e7a68d985d1e266bbb5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-11 09:10:20.000000000","message":"Topic set to puppet-lint-latest","accounts_in_message":[],"_revision_number":1},{"id":"81b034c86a982453ab807b44879178bb52e4ff58","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-05-11 09:43:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c9b09abec8103bfb00d226fcde03e65adc7b6a1a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-11 13:38:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b68b96176e0b6f37f9df65708a2b28f5f09b78b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 13:39:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"535f26da6707aeb316197169f39b8a59dc2c48c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 16:58:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aca90b369f0644b29dda736679437eae : SUCCESS in 4m 26s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6417330e882d416c9dd4549cb13719e4 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1dbbcde6f294c1793c52448f0339773 : SUCCESS in 3m 52s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad8d47ff03a34aa5a17d5e9957973139 : SUCCESS in 18m 29s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/d77db6665d15450ea339fde94ebb88f7 : SUCCESS in 18m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f60673c764aa4694bfab2285b7dbb808 : SUCCESS in 17m 29s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0cf2b783b07c4ccbb3a9d612802fbd93 : SUCCESS in 15m 46s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/408419316627451b928e82135d903bb5 : SUCCESS in 1h 26m 23s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fbc54ecb0cdf4efeba565b10aa41761b : SUCCESS in 1h 27m 53s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/faecef964cd44a6182efcaa734980ab5 : FAILURE in 34m 22s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a4d78c83d5614240b43e7d2fce60426a : SUCCESS in 1h 07m 49s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/679173a38fc5419ea485f89d2025c66a : SUCCESS in 22m 51s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e989a1efc5f940c093e5fad8034a523b : SUCCESS in 17m 22s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/27419289a60044ccbbf37111a0854832 : SUCCESS in 3h 13m 17s","accounts_in_message":[],"_revision_number":1},{"id":"02a637cb181487620476ccc290ad61f078f37981","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-11 22:58:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"139c71fe4718e730bf71b6204242e6218a20362c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 01:53:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b04715ba33549b3aec7cc09882c1e95 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/503d7eef3f7b4f3fa62e2fd5e1b4fe43 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98c5e12ab39c4c11af3bf13ece7fe9fe : SUCCESS in 3m 55s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/69542d75662841ec81acd9e00eb1fa4b : SUCCESS in 15m 05s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/84c1dabfdc8e4795bfbd0acbadd48c90 : SUCCESS in 16m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0fb197815eae4663ae25dd4f558e263c : SUCCESS in 15m 37s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb5d99648368493099d3a8665dc84a8d : SUCCESS in 15m 04s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea3bc4f52f9b449d9ce97216127afcea : SUCCESS in 16m 00s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1037b23f8a54313884947e790289889 : SUCCESS in 59m 20s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f81b3451aea14f6ea95fddca3deb009d : SUCCESS in 56m 07s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cad861d567c14d18afdd33ff005d2706 : FAILURE in 31m 17s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eae850b016744b88a364eca8df935f36 : SUCCESS in 45m 04s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e82024e26f51483ebcf746f72d7d45a2 : SUCCESS in 21m 25s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5c256affa4ec4128bd6a43226e466d52 : SUCCESS in 1h 28m 01s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a6d0e58071747b08fb9bce55fde9da9 : SUCCESS in 1h 34m 36s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2e8920ff61543c29d7df8c4feb61c81 : SUCCESS in 1h 31m 55s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d32d203122fb47d7969ae0b16270dc4b : SUCCESS in 1h 09m 10s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fe41183aa33c4e62b592baf0081a3b1c : SUCCESS in 21m 41s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/819dd34868944785a9db9e2a7131851f : SUCCESS in 1h 22m 10s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7ebcfb55e5f943f9a15cd970efa20942 : SUCCESS in 1h 45m 52s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c0c2d2e6a70441c2b06c0ffdfdbcec6d : SUCCESS in 1h 25m 38s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c643a87be7a4484999dbf1a17a79097b : SUCCESS in 1h 01m 43s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/fcfefeac792b474bb93f3790b2444212 : SUCCESS in 21m 24s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/661cb128b97347e6b7c90c06523b9803 : SUCCESS in 15m 25s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ab053c6c008499c91ec666551e5240a : SUCCESS in 22m 36s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d5a2af2725c4b7484d93f7e127a8cd9 : SUCCESS in 15m 49s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59aa86ec14884715908aefaf6a35939e : SUCCESS in 17m 10s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/544fc9c3a2f6446ba1449ca7780401a5 : SUCCESS in 2h 48m 58s","accounts_in_message":[],"_revision_number":1},{"id":"a8b1bfaa185fefcb47d236ecaadc06138d76f5a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 01:54:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b635b66c1da6e263d4d4dbd1f59526dbfe27b56e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 03:58:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3cd7dec4aca4d8b9383af87180f8a03 : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2c3d1e89d0a43b9a52ea2e540c2a180 : SUCCESS in 3m 30s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6888e799b30c4310ab012a1cec66678d : SUCCESS in 4m 47s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f5da99bc0084bca9d67f589e65d5600 : SUCCESS in 17m 02s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/eba6d54e860a4b998715bcef834848de : SUCCESS in 17m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26bd3ef881e0418cac31fd1643b47ff1 : SUCCESS in 17m 04s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d70c911ef3d492ca6a43c1aae85e11c : SUCCESS in 13m 06s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7efc9eeb3fbd46c6b744e1bdec15b8ff : SUCCESS in 1h 22m 22s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bae5906e49714f3da17e1a9338892a49 : SUCCESS in 1h 09m 42s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dbd584871c1a422b80493fa94cd94cb4 : SUCCESS in 1h 12m 46s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b8e6af2f0ee54614ada358a0af9476bc : SUCCESS in 1h 23m 01s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7cde725fac9b4e30a139f0af29680ca7 : SUCCESS in 22m 29s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee74b161a51d475b97d3036d8222a56d : SUCCESS in 15m 44s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1054e4b0799b4780b03bf28a27471f3c : SUCCESS in 2h 01m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e4ec8309528ef68dc05bc7e52d4b03b8d565718c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 03:58:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4fafcc5b5a0e4c80fdf131844285bfdb5cf65b5f","revisions":{"4fafcc5b5a0e4c80fdf131844285bfdb5cf65b5f":{"kind":"REWORK","_number":1,"created":"2021-05-11 03:18:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/790548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/48/790548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/48/790548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/48/790548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/48/790548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/48/790548/1"}}},"commit":{"parents":[{"commit":"65395243f7ca82f003400be641e63faba5c976a8","subject":"Merge \"Escape hyphen in neutron_agent_sriov_numvfs regexp\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/65395243f7ca82f003400be641e63faba5c976a8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-11 03:16:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-11 03:17:44.000000000","tz":540},"subject":"Fix lint errors with the latest lint packages","message":"Fix lint errors with the latest lint packages\n\nThis change fixes the following lint errors discovered since we removed\npin of lint packages.\n\nmanifests/agents/ml2/networking_baremetal.pp:99:WARNING: No matching\nclass parameter for documentation of\nneutron::agents::ml2::networking_baremetal::user_domain_id\nmanifests/agents/ml2/networking_baremetal.pp:95:WARNING: No matching\nclass parameter for documentation of\nneutron::agents::ml2::networking_baremetal::project_domain_id\nmanifests/server.pp:275:WARNING: class included by absolute name\n(::$class)\n\nThe first two errors are caused by leftover of\n4b39bef363b8e47ba9c2d3fd161e76e3bea8b836\n\nChange-Id: Ibd4d25b1f4e658667a522db5dd450fa47867e92f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4fafcc5b5a0e4c80fdf131844285bfdb5cf65b5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4fafcc5b5a0e4c80fdf131844285bfdb5cf65b5f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
