)]}'
{"id":"openstack%2Fpuppet-neutron~790840","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fwallaby~I49998e3aa5c8d7b47c79764587594ee66c34bb24","project":"openstack/puppet-neutron","branch":"stable/wallaby","hashtags":[],"change_id":"I49998e3aa5c8d7b47c79764587594ee66c34bb24","subject":"Escape hyphen in neutron_agent_sriov_numvfs regexp","status":"MERGED","created":"2021-05-12 07:24:14.000000000","updated":"2021-05-17 19:05:08.000000000","submitted":"2021-05-17 19:05:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790840","cherry_pick_of_change":789834,"cherry_pick_of_patch_set":1,"meta_rev_id":"cfc4fbb8d1068311965703f517e549f377f633c4","_number":790840,"virtual_id_number":790840,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-17 19:05:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-05-13 00:39:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-05-17 15:47:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-05-17 15:47:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-12 10:43:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-13 00:39:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-05-17 15:47:19.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"21680fb2b7b0828ac354bc5c08f21a7486f96335","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-05-12 07:24:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ac0b8eae62d30bd6fa49de261aaef3b54b19fb53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 10:43:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6dd76cb0ef0747e987af6709067d0d1b : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0165f0a33abf43da95338ef3a16ce2c2 : SUCCESS in 3m 47s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/071fc9a44bb64337bc7bd58adef7f142 : SUCCESS in 18m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6eef4d63d2a64b7084cad40741d53476 : SUCCESS in 14m 27s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6dd26ce4ef842848393a64e725aada1 : SUCCESS in 14m 45s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6a686c9b66047c8a4393ee0905a4b0f : SUCCESS in 59m 03s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bd98f9013d44f87bf96ce71aa1c7cd3 : SUCCESS in 57m 35s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34a0f3c0be6e4b20b15a7961cda28f30 : FAILURE in 28m 06s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b340d52ac4946329d3883908cf631c3 : SUCCESS in 40m 10s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/36b750625db349e0988366edf64a22ff : SUCCESS in 1h 46m 47s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/77b1c134211c4bc5a677df5b89a1631c : SUCCESS in 1h 37m 15s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a7fc53507e14558b21234a48abca174 : SUCCESS in 1h 21m 22s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4d6b9a63bbf4802be6d9438aec66226 : SUCCESS in 1h 05m 16s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/64b43b7d88cd47cdb6b600623b2b8642 : SUCCESS in 22m 03s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/449e2b02dde34d9a902502eb7a19365d : SUCCESS in 17m 54s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/53d723a2795d42ce952216eed19d9440 : FAILURE in 3h 13m 38s","accounts_in_message":[],"_revision_number":1},{"id":"79815c8ebfbb48114670fbf7acb51fefd6f2c705","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-05-12 14:34:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eabbc4c9bb44dc4aba86e1fbba2d1e4515fc165b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 17:09:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14b4928105ca4fd2b5e310e2c5c0cd08 : SUCCESS in 4m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cebeee58ec414bcba8f4611ac0b9a114 : SUCCESS in 3m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c98aaf63d82c4dbf94ab645869294a30 : SUCCESS in 18m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d26fd8c8d62f4c0d879bb1415989fc30 : SUCCESS in 12m 41s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1739e44d4854d0c9a8ec7888da90c25 : SUCCESS in 13m 09s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c74053a5556240d68a86fe6ebd25a261 : SUCCESS in 53m 39s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/004d162e74704f02ad616e9c73a5e393 : SUCCESS in 53m 37s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6cc1774e7f841f28abedd00746f7a2c : FAILURE in 24m 11s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bab9e2eb0b340ff8f8bb0ec259cc188 : SUCCESS in 42m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a8e9a91d06f4cce92e6e924e5ddec7c : SUCCESS in 1h 06m 45s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ba0b51dda564abeb4fbe0b60041e031 : SUCCESS in 1h 06m 03s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/46b7f09fcb384350ab33b6c3a170cf7f : SUCCESS in 1h 21m 50s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/79352d73c80b4a4c814701ec254fa2e8 : SUCCESS in 59m 08s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d8f998ff9b24f4fb37f7c062c840508 : SUCCESS in 21m 44s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21401d9bc579441ca7da0e13f8cfa7b0 : SUCCESS in 16m 56s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6bb6af52958744909aacf353b3067ed0 : FAILURE in 2h 32m 43s","accounts_in_message":[],"_revision_number":1},{"id":"3da407c06b562cc00230d0ddbe7cd29f3a0d01c4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-13 00:39:50.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"99ac65ff5da2145f2df8232cdeb78a5da9a35c36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 03:01:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f0e63f34d9a45f5a05abf6dfc5f9931 : SUCCESS in 5m 52s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6e36f9fa1594cc983708273a15fb1b1 : SUCCESS in 5m 24s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e04bf7c5ae144e5b4fdea1a0f999877 : SUCCESS in 17m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6d42e5834f54aa5a9fd93d50da94a4a : SUCCESS in 14m 08s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29c9eae09ca548d48ab432f68d92c9e5 : SUCCESS in 15m 11s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ebe127068b24047b94b4f313324f262 : FAILURE in 40m 38s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ba870bbb57e45a79aaafdea5e28563e : FAILURE in 39m 35s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a26bcf295e4418d8229c149e57f1da6 : FAILURE in 27m 37s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7d17300ce4c4bb884ae0469a6779126 : FAILURE in 28m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d354743d269e45eaa00c05e00dbf52bf : SUCCESS in 1h 31m 16s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/81debe15a1b443f28a8a44ce5b24bd8b : SUCCESS in 1h 39m 31s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/15381671955f4c8294444a01259e81e2 : SUCCESS in 1h 14m 10s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/455df6c65abc4e0e8e228a90b57a6a95 : SUCCESS in 1h 01m 45s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/efbfa6d0647447ae9d21bca35dc3aff1 : SUCCESS in 26m 00s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0ddae387ed24ba7bd33000ff4723cd8 : FAILURE in 18m 19s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/91754e0d371849728817e6e3277c70c2 : SUCCESS in 2h 19m 46s","accounts_in_message":[],"_revision_number":1},{"id":"5a14033bee48ecff88019c0ffc1c3ac7c7258f3a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-17 15:47:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b5efa0e672a1365b0976ea0ff9e490d4adcf2f45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 15:48:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"312f24e5c84308f38c368cc34e076b10a0dc5636","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 19:05:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5176e95e6e449f6bf47a0973a8f772d : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e73e4e0a960f45e3b1520d34f7006976 : SUCCESS in 3m 55s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7472b9ed294c4f1ba473b7edf3ad3922 : SUCCESS in 15m 41s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ac267bea1664100bb59cb71833f2b7d : SUCCESS in 13m 06s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d8986fe8d3da4f0eb7571275fa27e307 : SUCCESS in 1h 26m 13s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/844fa33eeeb340c1afc890eca7927bfe : SUCCESS in 1h 15m 51s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d33b8c1882c94f7083eaecc76cb4703e : SUCCESS in 1h 27m 21s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6d030e8527d40dfbca897fd08afae9f : SUCCESS in 56m 25s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/64eceed0b52b49ef838d70ace8aca16b : SUCCESS in 20m 11s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60f94da82dae43db82230ceca526bc65 : FAILURE in 24m 28s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2a0c79e184264fb186af9e7da012b63a : SUCCESS in 3h 12m 39s","accounts_in_message":[],"_revision_number":1},{"id":"cfc4fbb8d1068311965703f517e549f377f633c4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 19:05:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c15b18bc6e9eb287bdfc272b50ab15489737103d","revisions":{"c15b18bc6e9eb287bdfc272b50ab15489737103d":{"kind":"REWORK","_number":1,"created":"2021-05-12 07:24:14.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/40/790840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/40/790840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/40/790840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/40/790840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/40/790840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/40/790840/1"}}},"commit":{"parents":[{"commit":"db151245184f2188e8e4ef8be6b5418eba4e926b","subject":"Fix L3 agent\u0027s extensions list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/db151245184f2188e8e4ef8be6b5418eba4e926b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2021-05-05 13:45:08.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2021-05-12 07:24:14.000000000","tz":0},"subject":"Escape hyphen in neutron_agent_sriov_numvfs regexp","message":"Escape hyphen in neutron_agent_sriov_numvfs regexp\n\nChange-Id: I49998e3aa5c8d7b47c79764587594ee66c34bb24\n(cherry picked from commit 5cff5cdfd5d2b5ce27152e49458c8fcd44e3888b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c15b18bc6e9eb287bdfc272b50ab15489737103d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c15b18bc6e9eb287bdfc272b50ab15489737103d"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
