)]}'
{"id":"openstack%2Fpuppet-neutron~791065","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fvictoria~I136a81cc857130b3befa5df204525bfa6c6f0206","project":"openstack/puppet-neutron","branch":"stable/victoria","topic":"py3-packages","hashtags":[],"change_id":"I136a81cc857130b3befa5df204525bfa6c6f0206","subject":"Fix name of networking-ansible/baremetal package","status":"MERGED","created":"2021-05-13 00:15:00.000000000","updated":"2021-05-24 15:04:31.000000000","submitted":"2021-05-24 15:04:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791065-py3-packages","cherry_pick_of_change":789872,"cherry_pick_of_patch_set":2,"meta_rev_id":"a42859ac0186add58228a8ec92775e760e0feda7","_number":791065,"virtual_id_number":791065,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-24 15:04:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-05-17 15:47:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-05-24 12:21:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-05-24 12:21:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-13 05:37:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-17 15:47:45.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-05-24 12:21:27.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"cae5ad7ec561525762dd7aaeb16befa69b8abf55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-13 00:15:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79d95f539ec9176ee69ade7e2a42afe16c02b4b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-13 02:27:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e392426baa165066a122e36542d8dcc1a6acc685","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-13 02:28:01.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"342c84f9e5a5c22fc984ba6d5f68fe0241c003ee","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-13 02:28:43.000000000","message":"Topic py3-packages-stable/victoria removed","accounts_in_message":[],"_revision_number":3},{"id":"93f9dbd2b62fd7f7a8279eebf0c1ecf3f1e33474","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-13 02:28:48.000000000","message":"Topic set to py3-packages","accounts_in_message":[],"_revision_number":3},{"id":"ba0f31b09a99c5eec6182e5a5eddb34f20517ebb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 05:37:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72d54782bd654d40b9e9ecda682ba4e9 : SUCCESS in 4m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08dce126ab664b7280cc7cf4eb4c700e : SUCCESS in 4m 14s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/66d278b7c86148a9bfd4bc0cc553b653 : SUCCESS in 18m 29s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5da876029e9e45d19872a87302aac41f : SUCCESS in 16m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/452f52c6dd1f40c192c17c89536231f5 : FAILURE in 16m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8681a8c4c50a46b4a6a508bd439daf09 : SUCCESS in 56m 23s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7ba0143cfb14ee68307f83adc3bcc90 : SUCCESS in 58m 51s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/600407dea28d403594e17930e88c8575 : SUCCESS in 48m 39s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e27af6c94fa64c7faa743c2493b3e6c9 : SUCCESS in 40m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d7323efcd3624d4fa7c6fe64030f23bd : SUCCESS in 1h 24m 25s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d579c33e851a4db7928a011188c4c186 : SUCCESS in 1h 41m 55s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f467382bd0b445c3a8a0a5a11fa1e575 : SUCCESS in 1h 39m 12s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/85d897586d4645a8ab2742819f06b973 : SUCCESS in 1h 03m 32s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/40702bc0cf994e93a26bf530de029245 : SUCCESS in 1h 15m 22s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e699195848594f0497ef9c6f5ec9049c : SUCCESS in 1h 45m 34s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2b98ce7395d415a9a4066d4870f3174 : SUCCESS in 1h 37m 46s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1772bc197b8a41e3944d505ace981e5a : SUCCESS in 54m 51s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/b32400bafc834a9aba4c050b44b6a375 : SUCCESS in 22m 35s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/18ac430e8dbf4f6c9b53ba9f80ab9640 : SUCCESS in 22m 18s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8b413fbcc1746cf8e8a9c57a538cd39 : SUCCESS in 16m 29s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/65918df00be747659e0d84544aa5cf73 : SUCCESS in 3h 03m 09s","accounts_in_message":[],"_revision_number":3},{"id":"992c9182f909c94bee707ba0b86481317c09ee28","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-17 15:47:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4ff4d0a628e98922884093f74c0befeabc32724a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-05-24 12:21:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c3aec5a434814aa8816cfa5729d67cafed81e2d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 12:22:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5a39c2190cf6c545a592e79bf36101ef35d7cd3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 15:04:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/915a36fbc1524259a558c1cd744c2038 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dcd197dcf3404384892459590cc8439b : SUCCESS in 3m 26s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/236cb1b576e249819aa9006ba7abb335 : SUCCESS in 15m 30s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aefab89017a1407384c2c9abc1f22ef7 : SUCCESS in 13m 41s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e07795fb57c843818fa6c0bd5f47acfc : SUCCESS in 1h 01m 33s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/49e89a8afcd74a91913b5d2e5cb19d3d : SUCCESS in 1h 10m 10s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a88ccd50788b418a954db00ac7c2dc03 : SUCCESS in 1h 06m 59s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ecdfb694bd634d75b1c82892d581d5e5 : SUCCESS in 48m 39s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/14556f5b7b034873a326b038150ae5eb : SUCCESS in 21m 00s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/800e7ea4bbc4462bb4570bd2a408ecbe : SUCCESS in 21m 10s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe50fbe241e14d8ca8f9783179413ce5 : SUCCESS in 18m 38s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8b683024efa94def8997d66c0462edf2 : SUCCESS in 2h 00m 43s","accounts_in_message":[],"_revision_number":3},{"id":"a42859ac0186add58228a8ec92775e760e0feda7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 15:04:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b85b99dfac4fb29c136bba52779d32cef16cab3b","revisions":{"682ac9897f23f27f4f79db35c191cf6b263ca20d":{"kind":"REWORK","_number":1,"created":"2021-05-13 00:15:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/791065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/65/791065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/791065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/791065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/791065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/65/791065/1"}}},"commit":{"parents":[{"commit":"9029fe04cc2f504930613000f2169befc3540255","subject":"Fix L3 agent\u0027s extensions list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9029fe04cc2f504930613000f2169befc3540255"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-30 02:42:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-13 00:14:39.000000000","tz":540},"subject":"Fix name of networking-ansible/baremetal package","message":"Fix name of networking-ansible/baremetal package\n\nWe should install python3-* instead of python2-* in CentOS8.\nThis change fixes the remaining usage of python2-* left for\nnetwokring-ansible and networking-baremetal.\n\nChange-Id: I136a81cc857130b3befa5df204525bfa6c6f0206\n(cherry picked from commit 34b5036ed8a1b7d7b50ec85484f496d861be57bf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/682ac9897f23f27f4f79db35c191cf6b263ca20d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/682ac9897f23f27f4f79db35c191cf6b263ca20d"}]},"branch":"refs/heads/stable/victoria"},"d0745538e1abc102a48850b2e277c4953ce116e9":{"kind":"REWORK","_number":2,"created":"2021-05-13 02:27:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/791065/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/65/791065/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/791065/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/791065/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/791065/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/65/791065/2"}}},"commit":{"parents":[{"commit":"9029fe04cc2f504930613000f2169befc3540255","subject":"Fix L3 agent\u0027s extensions list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9029fe04cc2f504930613000f2169befc3540255"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-30 02:42:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-13 02:27:11.000000000","tz":0},"subject":"Fix name of networking-ansible/baremetal package","message":"Fix name of networking-ansible/baremetal package\n\nWe should install python3-* instead of python2-* in CentOS8.\nThis change fixes the remaining usage of python2-* left for\nnetwokring-ansible and networking-baremetal.\n\nBackport note:\nThis includes 260aa9fe292646f553467cc6b63475261fed9b54 which fixed\nthe wrong package names in unit tests.\n\nChange-Id: I136a81cc857130b3befa5df204525bfa6c6f0206\n(cherry picked from commit 34b5036ed8a1b7d7b50ec85484f496d861be57bf)\n(cerrry picked from commit ca266112909fdd15f1c90c2778238bb66ff6c491)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d0745538e1abc102a48850b2e277c4953ce116e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d0745538e1abc102a48850b2e277c4953ce116e9"}]},"branch":"refs/heads/stable/victoria"},"b85b99dfac4fb29c136bba52779d32cef16cab3b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-05-13 02:28:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/791065/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/65/791065/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/791065/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/791065/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/791065/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/65/791065/3"}}},"commit":{"parents":[{"commit":"9029fe04cc2f504930613000f2169befc3540255","subject":"Fix L3 agent\u0027s extensions list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9029fe04cc2f504930613000f2169befc3540255"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-30 02:42:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-13 02:27:59.000000000","tz":0},"subject":"Fix name of networking-ansible/baremetal package","message":"Fix name of networking-ansible/baremetal package\n\nWe should install python3-* instead of python2-* in CentOS8.\nThis change fixes the remaining usage of python2-* left for\nnetwokring-ansible and networking-baremetal.\n\nBackport note:\nThis includes 260aa9fe292646f553467cc6b63475261fed9b54 which fixed\nthe wrong package names in unit tests.\n\nChange-Id: I136a81cc857130b3befa5df204525bfa6c6f0206\n(cherry picked from commit 34b5036ed8a1b7d7b50ec85484f496d861be57bf)\n(cherry picked from commit ca266112909fdd15f1c90c2778238bb66ff6c491)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b85b99dfac4fb29c136bba52779d32cef16cab3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b85b99dfac4fb29c136bba52779d32cef16cab3b"}]},"branch":"refs/heads/stable/victoria","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
