)]}'
{"id":"openstack%2Fpuppet-neutron~791066","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fussuri~I136a81cc857130b3befa5df204525bfa6c6f0206","project":"openstack/puppet-neutron","branch":"stable/ussuri","topic":"py3-packages","hashtags":[],"change_id":"I136a81cc857130b3befa5df204525bfa6c6f0206","subject":"Fix name of networking-ansible/baremetal package","status":"MERGED","created":"2021-05-13 00:18:16.000000000","updated":"2021-05-24 15:04:34.000000000","submitted":"2021-05-24 15:04:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791066-py3-packages","meta_rev_id":"64ac8b1d2e3d67884585f25ad20447a1092743e7","_number":791066,"virtual_id_number":791066,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-24 15:04:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-05-17 15:48:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-05-24 12:21:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-05-24 12:21:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-13 05:48:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-17 15:48:03.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-05-24 12:21:38.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"3c1a03630f53172dd93b0b9722fa3aedc3446a20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-13 00:18:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"521c53f7819842a89cc087001dfa1f8eae304323","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-13 02:31:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"61bdf07b77981ccae46d9cf2de79898a51743e8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 05:48:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2a7986f1f2c4e31a3d4b8fcb2cfef8c : SUCCESS in 3m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f93d1fdc2854483a8ec86c3cc5ec3a16 : SUCCESS in 3m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8d757ff5daa4d849379bfbe32bde6c8 : SUCCESS in 3m 40s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/9b8d3092bf52487d95f43243f7b15f6f : SUCCESS in 19m 37s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/95af17556ada403abedf28c1f13a4798 : SUCCESS in 20m 34s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/36a68dd8629b4ca4850c58650c7eeb5c : SUCCESS in 19m 32s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/9e45862205b3487b8cc260d763c7f796 : SUCCESS in 19m 56s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cfbf6962798944a9aa0bcea1e9919861 : SUCCESS in 12m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a28b41cda25a46838f3fe1b9917a4f6c : FAILURE in 16m 06s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/709865bdf5c642fd86a8bdd13a99924f : SUCCESS in 54m 20s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6ac66c5ec254f62bbcade83ff7531f8 : SUCCESS in 52m 56s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eefb5a73108d411f8cf14f496e58d640 : SUCCESS in 54m 06s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/9808b36976ab4770aa69059c965bde98 : SUCCESS in 40m 02s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fdf2caf53d5141f4ae44f4a82b5d6f30 : SUCCESS in 59m 15s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/72ad3e97039341fd8265f0677c08e469 : SUCCESS in 1h 11m 35s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6dc5857a14894b0f9c074579425acdce : SUCCESS in 1h 07m 16s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9de0286fdbec439cb69a3871a345b616 : SUCCESS in 49m 10s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/42cabcdec26748059e168f638b9e50ce : SUCCESS in 1h 02m 44s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/27d8e745ae254ba9b110c7f4130a5210 : SUCCESS in 1h 05m 33s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3bb8df042e744640b7bad0177b7cdca7 : SUCCESS in 1h 12m 35s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e0911f5dbb7f465a86a8f86b41ab5410 : SUCCESS in 49m 19s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/afdcf12a43dc46cf991199f0176aba66 : SUCCESS in 22m 11s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/4b49646090ac43efb02792c9cfd56287 : FAILURE in 22m 07s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64cd6a7b58834785b6a76e1480282537 : SUCCESS in 16m 31s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/01cf3b92c7c44231b14f00b4e09af395 : SUCCESS in 3h 10m 22s","accounts_in_message":[],"_revision_number":2},{"id":"7b4b773be79e226650e6f873155aac6f72c1d092","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-17 15:48:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"30c7fbcec1e9c06fad6ea28642faeab9e6f78821","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-05-24 12:21:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0769ac87d38924b60b0dced647358451440ca46c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 12:22:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5c56d1b0eac76f0905c968ab5b8fa06a56653f77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 15:04:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f17232bfd2e54b24883184a422e7b73e : SUCCESS in 3m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe95f713636c4386be41eb5a0db5ed2b : SUCCESS in 8m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d632bfa154642db86fea07f79783f2d : SUCCESS in 6m 11s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/64c247be32a94da3a418d0ed268bf327 : SUCCESS in 14m 21s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0ad1c6c0a284c3fb1091d65e9f38d75 : SUCCESS in 14m 20s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63692a242eab41d1bb69f826fdad6be2 : SUCCESS in 13m 25s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2aa5dcc8114d4123b43300afa64ea253 : SUCCESS in 58m 56s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e47aac42e97b427ca0aa851f5c71c9d7 : SUCCESS in 1h 09m 28s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7b9b30a6dbb74b3e9642126a984548b4 : SUCCESS in 1h 03m 07s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0fe924c08ef24003a0de4bb24344cf54 : SUCCESS in 46m 40s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/9e84cb951c8e4e089103a4e972442d71 : SUCCESS in 23m 38s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ec40f5e6a66749dd83461a7eb6aece37 : SUCCESS in 2h 39m 51s","accounts_in_message":[],"_revision_number":2},{"id":"64ac8b1d2e3d67884585f25ad20447a1092743e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 15:04:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3b9df8e4b043471d875b8c112e04634a02ca64f6","revisions":{"c3e2a0092ba35b7f304fc9f65305e748e7ea2ff0":{"kind":"REWORK","_number":1,"created":"2021-05-13 00:18:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/791066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/66/791066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/66/791066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/66/791066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/66/791066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/66/791066/1"}}},"commit":{"parents":[{"commit":"82324318ca5af337c4878588554965872d458ce4","subject":"Fix L3 agent\u0027s extensions list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/82324318ca5af337c4878588554965872d458ce4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-30 02:42:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-13 00:17:28.000000000","tz":540},"subject":"Fix name of networking-ansible/baremetal package","message":"Fix name of networking-ansible/baremetal package\n\nWe should install python3-* instead of python2-* in CentOS8.\nThis change fixes the remaining usage of python2-* left for\nnetwokring-ansible and networking-baremetal.\n\nConflicts:\n\tmanifests/plugins/ml2/networking_ansible.pp\n\tspec/classes/neutron_plugins_ml2_networking_ansible_spec.rb\n\tspec/classes/neutron_plugins_ml2_networking_baremetal_spec.rb\n\nChange-Id: I136a81cc857130b3befa5df204525bfa6c6f0206\n(cherry picked from commit 34b5036ed8a1b7d7b50ec85484f496d861be57bf)\n(cherry picked from commit 682ac9897f23f27f4f79db35c191cf6b263ca20d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c3e2a0092ba35b7f304fc9f65305e748e7ea2ff0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c3e2a0092ba35b7f304fc9f65305e748e7ea2ff0"}]},"branch":"refs/heads/stable/ussuri"},"3b9df8e4b043471d875b8c112e04634a02ca64f6":{"kind":"REWORK","_number":2,"created":"2021-05-13 02:31:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/791066/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/66/791066/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/66/791066/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/66/791066/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/66/791066/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/66/791066/2"}}},"commit":{"parents":[{"commit":"82324318ca5af337c4878588554965872d458ce4","subject":"Fix L3 agent\u0027s extensions list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/82324318ca5af337c4878588554965872d458ce4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-30 02:42:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-13 02:31:07.000000000","tz":540},"subject":"Fix name of networking-ansible/baremetal package","message":"Fix name of networking-ansible/baremetal package\n\nWe should install python3-* instead of python2-* in CentOS8.\nThis change fixes the remaining usage of python2-* left for\nnetwokring-ansible and networking-baremetal.\n\nBackport note:\nThis includes 260aa9fe292646f553467cc6b63475261fed9b54 which fixed\nthe wrong package names in unit tests.\n\nConflicts:\n\tmanifests/plugins/ml2/networking_ansible.pp\n\tspec/classes/neutron_plugins_ml2_networking_ansible_spec.rb\n\tspec/classes/neutron_plugins_ml2_networking_baremetal_spec.rb\n\nChange-Id: I136a81cc857130b3befa5df204525bfa6c6f0206\n(cherry picked from commit 34b5036ed8a1b7d7b50ec85484f496d861be57bf)\n(cherry picked from commit ca266112909fdd15f1c90c2778238bb66ff6c491)\n(cherry picked from commit b85b99dfac4fb29c136bba52779d32cef16cab3b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3b9df8e4b043471d875b8c112e04634a02ca64f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3b9df8e4b043471d875b8c112e04634a02ca64f6"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
