)]}'
{"id":"openstack%2Fpuppet-neutron~791067","triplet_id":"openstack%2Fpuppet-neutron~stable%2Ftrain~I136a81cc857130b3befa5df204525bfa6c6f0206","project":"openstack/puppet-neutron","branch":"stable/train","topic":"py3-packages","hashtags":[],"change_id":"I136a81cc857130b3befa5df204525bfa6c6f0206","subject":"Fix name of networking-ansible/baremetal package","status":"MERGED","created":"2021-05-13 00:19:43.000000000","updated":"2021-05-24 17:34:15.000000000","submitted":"2021-05-24 17:34:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791067-py3-packages","meta_rev_id":"6466648d40ae51e09b9ba72c514192b64d68dd03","_number":791067,"virtual_id_number":791067,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-24 17:34:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-05-17 15:48:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-05-24 12:21:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-05-24 12:21:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-13 01:32:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-17 15:48:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-05-24 12:21:49.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"0592a827d01cde8e5d514d3a911cffaaa19853e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-13 00:19:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1020b50bfdabfb28f04e1123dcf3f227508a3cad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 01:32:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca0ce5c6f0d443dbb9ef63c85a67b413 : SUCCESS in 5m 04s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a061c5fedd2c4373ab9f8b2d409265c3 : SUCCESS in 4m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a37ef914416e4f5c93cff42b63f8541c : SUCCESS in 5m 11s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/48dba683be3747878cde7b978c363a0e : FAILURE in 37m 46s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/eecc6c2513ac44e29e1bfa04240f4852 : FAILURE in 19m 42s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/9692a9e7f03646e0a02cb3237190b706 : FAILURE in 20m 16s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc8995aaf80a4459b521ce30d439573a : FAILURE in 15m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/598917d3b6ba4a619e4f0c9df64d196b : FAILURE in 18m 36s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/88b9ba2d714e4c8dba63f96f2c529fa3 : SUCCESS in 55m 09s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ab7a1095cc0347d6ab8c7fcb2ac32784 : SUCCESS in 1h 07m 54s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/73ebff040e7440e7a6d83bee4a7dbcfa : SUCCESS in 1h 02m 19s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/c28152554ca7456dafd3e03ca5e24c93 : SUCCESS in 48m 36s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/19b82094ad354defa6eacc81997736dd : SUCCESS in 45m 25s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aff94a2bc30a44b5b0d0703c1d2dfcfd : SUCCESS in 41m 57s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a56bcd8d8f7a49f7ac36b3b264cb1d87 : SUCCESS in 35m 18s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/161f7d9c674d44fa9575fdfe5db8e200 : SUCCESS in 33m 33s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d72aebd130d346d8891859f4813f83e8 : SUCCESS in 58m 01s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/46241730085c44e6b0e2f797ef495fc1 : SUCCESS in 1h 00m 38s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0126fa6df5e047c09e1403c20c6d3a95 : SUCCESS in 1h 00m 31s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6ca5aaedee0b47d79e7757c13378ef4a : SUCCESS in 43m 39s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4b096995e7cd494da4a9d82cde3624a2 : SUCCESS in 19m 17s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a150e4068a294acf8396afca51d3fdff : FAILURE in 8m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5217f81a539af58116e748b3071480488070d922","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-13 02:35:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"effc762c5d2668e80a1080ede2f8e798c77fb0ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 03:46:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bdbadfec19e44da2866c5e04790f4cdd : SUCCESS in 5m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1dadbb6b7284517b766c7d3f30448da : SUCCESS in 4m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f5397b9e8fc491e99307b6fea60bb31 : SUCCESS in 4m 54s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a4b5ac330f0f4cabb2d62939e1733630 : SUCCESS in 32m 36s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/821860b58d6542bcbe576f5e4adc3a26 : SUCCESS in 20m 22s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/6d07503d74134ae69c47c9d65c5ef5a2 : SUCCESS in 18m 37s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9c9128abb48416f90c17bb72d53210b : SUCCESS in 13m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c866035af0874b3fae3a97fa1158eef3 : FAILURE in 16m 44s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/4f1c9dcf3e1643478ce5cf4ad310fb80 : SUCCESS in 58m 00s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f55fbd64350a471aabd49af6497ad951 : SUCCESS in 1h 03m 27s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/0a0f782f81234a66bcad05c1675f5521 : SUCCESS in 1h 06m 32s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/18efcc531c1c4942ab070151daba89fe : SUCCESS in 49m 58s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/eda31591d44649d2867c2bc224556c99 : SUCCESS in 43m 25s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71eefdbab5f94a009e7cd2e2a58ebfce : SUCCESS in 42m 11s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47be657a7b944f829e21365bae54c079 : SUCCESS in 36m 03s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ecbc2eea362a4e2da1dfef4a9465383b : SUCCESS in 33m 30s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/61bc31eaee5b4e42ac094a1442894658 : SUCCESS in 48m 28s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/50cefb43a25744589e218a9b235a38b1 : SUCCESS in 56m 35s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/446bdc1fd5c64148a5b6057846c2f071 : SUCCESS in 52m 35s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ccd3977335204aa4b1c94536d22e41ec : SUCCESS in 41m 58s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d03f1da4751d482d88a0f94481b76727 : SUCCESS in 17m 22s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6098e59127504568aecdc5c53f778315 : FAILURE in 7m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"747010ee58f6441bf83272076f93bff204399ddf","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-17 15:48:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"57ba30088e503564a69849fc77879b74ff74c8dc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-05-24 12:21:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5facb7f24e85742f2c9e138737a275be7731c406","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 12:22:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a8ec00f89cf5bee99f0f174b436f1927d2db4eea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 15:04:36.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0ac7533afc9422ba5f006a484cae42d : SUCCESS in 4m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d92d45864d6f45c186fe32b5c171e0f9 : SUCCESS in 4m 48s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40599cfd7354467587864a4ea89f5e09 : SUCCESS in 4m 54s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/96f089fe3d9e4ef798a0fd5ab34e4fc4 : SUCCESS in 21m 51s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a977368046bb4724968d85131b27b347 : SUCCESS in 17m 32s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/7c7dbf23a2b3496fbb2d07ac513531d6 : SUCCESS in 44m 55s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/22f6dfdc278648308dfd159a9e008b21 : FAILURE in 47m 02s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3280204b2ba34e17958f3e17465303ed : SUCCESS in 55m 25s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/6df072072a7944019b6abb0b323db6d6 : SUCCESS in 41m 20s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8c38e93645034aa58f15afb384968d3e : SUCCESS in 18m 27s","accounts_in_message":[],"_revision_number":2},{"id":"63a4e1002cc3670a7bc89bb983f23ea39469edd2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-24 15:05:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f38d8e5c90056041465b3a11da526c116e19baf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 16:13:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e00322f0b6e34dad9b85066f25d1aeb2 : SUCCESS in 5m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51b7fd5f5d954ca9b9a81a0b57df001d : SUCCESS in 5m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/892bcae14623469589f511cdfbbfc7c1 : SUCCESS in 5m 12s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/689a605e0f554a88baf78157859176c0 : SUCCESS in 27m 03s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/6df4f9a6f02c44cd95b1810cdf002998 : SUCCESS in 15m 08s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/e35be257f5cd4578bd69efb7acc965e9 : SUCCESS in 13m 39s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5cf288de12b46f8a3f266e368c83349 : SUCCESS in 12m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de155e6d11504436b2817e82664920ef : FAILURE in 15m 21s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/592d198c645f44a78ccf600c5f20e095 : SUCCESS in 46m 00s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/dcfe23b8e02d4054883ee861d8b55a53 : SUCCESS in 53m 38s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a8d0e5cbba594bf6928ec8c196e3469f : SUCCESS in 1h 06m 26s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/9cf6737c77254d398a1f8ed08a1765f6 : SUCCESS in 47m 57s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e563d26735af4cc4bde691037f543675 : SUCCESS in 40m 30s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48aab0b8394e49a190c880e848dd6271 : SUCCESS in 47m 44s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/505310ff592d4fd795ed8ebe881f908d : SUCCESS in 34m 38s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/af00f4ad6ecc44728f5bc87a44f862d4 : SUCCESS in 40m 56s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a49260a78d66411abcb9e8ddb5489ec7 : SUCCESS in 53m 25s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e3e3bdcf0e0c46c2a37a47184f4f16c3 : SUCCESS in 53m 24s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/704175f5c95e4c26a122b830b11709a4 : SUCCESS in 48m 50s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4c5958f87d64414c90a12d5fc9f816f5 : SUCCESS in 37m 36s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2a9f06bfb3584cea8ec41c62eb58ac25 : SUCCESS in 18m 08s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5971174d7e34468785e161e158426440 : FAILURE in 9m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"88175fa8f5268538c5ae0f63789fc0990cae98e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 16:13:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ba3af1f8090658ed40205e546b82aee694edef81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 17:34:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c545af87c6c6434cbc90fed82d4ba717 : SUCCESS in 6m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8756bf0174f43b683bc9d005d07987e : SUCCESS in 5m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c67a23f226184f7bbfaddb8a63b97198 : SUCCESS in 6m 00s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3ea0f129cada4ac7a94049188de28ec7 : SUCCESS in 35m 34s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9626e881632148ea87dd8f51ab485df1 : SUCCESS in 18m 55s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/f4db6d64d84f4cbdb98088d7436f81e9 : SUCCESS in 1h 08m 09s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/730c6a4946a14cecbeb84332b753f445 : SUCCESS in 1h 11m 46s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4283c6c9e5364b50ab42a0987a47f348 : SUCCESS in 1h 15m 16s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/1a5e20115bb448d989157e14bb77ec0a : SUCCESS in 48m 29s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/bc9fc62241724bf6bd8f058b6edd8eeb : SUCCESS in 21m 26s","accounts_in_message":[],"_revision_number":2},{"id":"6466648d40ae51e09b9ba72c514192b64d68dd03","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 17:34:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9fc89a77ffa6c35eb2bf170762b1dcd7d9801e0c","revisions":{"a12f9a7a3c12ffe144934cdec1fc2b27aab1bdd7":{"kind":"REWORK","_number":1,"created":"2021-05-13 00:19:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/791067/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/67/791067/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/791067/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/791067/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/791067/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/67/791067/1"}}},"commit":{"parents":[{"commit":"c774dcb85ea936cec5fd7827b49eb9718764ee8c","subject":"Add support for the resource_provider_hypervisors parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c774dcb85ea936cec5fd7827b49eb9718764ee8c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-30 02:42:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-13 00:19:23.000000000","tz":540},"subject":"Fix name of networking-ansible/baremetal package","message":"Fix name of networking-ansible/baremetal package\n\nWe should install python3-* instead of python2-* in CentOS8.\nThis change fixes the remaining usage of python2-* left for\nnetwokring-ansible and networking-baremetal.\n\nConflicts:\n\tmanifests/plugins/ml2/networking_ansible.pp\n\tspec/classes/neutron_plugins_ml2_networking_ansible_spec.rb\n\tspec/classes/neutron_plugins_ml2_networking_baremetal_spec.rb\n\nChange-Id: I136a81cc857130b3befa5df204525bfa6c6f0206\n(cherry picked from commit 34b5036ed8a1b7d7b50ec85484f496d861be57bf)\n(cherry picked from commit 682ac9897f23f27f4f79db35c191cf6b263ca20d)\n(cherry picked from commit c3e2a0092ba35b7f304fc9f65305e748e7ea2ff0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a12f9a7a3c12ffe144934cdec1fc2b27aab1bdd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a12f9a7a3c12ffe144934cdec1fc2b27aab1bdd7"}]},"branch":"refs/heads/stable/train"},"9fc89a77ffa6c35eb2bf170762b1dcd7d9801e0c":{"kind":"REWORK","_number":2,"created":"2021-05-13 02:35:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/791067/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/67/791067/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/791067/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/791067/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/791067/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/67/791067/2"}}},"commit":{"parents":[{"commit":"c774dcb85ea936cec5fd7827b49eb9718764ee8c","subject":"Add support for the resource_provider_hypervisors parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c774dcb85ea936cec5fd7827b49eb9718764ee8c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-30 02:42:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-13 02:34:42.000000000","tz":540},"subject":"Fix name of networking-ansible/baremetal package","message":"Fix name of networking-ansible/baremetal package\n\nWe should install python3-* instead of python2-* in CentOS8.\nThis change fixes the remaining usage of python2-* left for\nnetwokring-ansible and networking-baremetal.\n\nBackport note:\nThis includes 260aa9fe292646f553467cc6b63475261fed9b54 which fixed\nthe wrong package names in unit tests.\n\nConflicts:\n\tmanifests/plugins/ml2/networking_ansible.pp\n\tspec/classes/neutron_plugins_ml2_networking_ansible_spec.rb\n\tspec/classes/neutron_plugins_ml2_networking_baremetal_spec.rb\n\nChange-Id: I136a81cc857130b3befa5df204525bfa6c6f0206\n(cherry picked from commit 34b5036ed8a1b7d7b50ec85484f496d861be57bf)\n(cherry picked from commit ca266112909fdd15f1c90c2778238bb66ff6c491)\n(cherry picked from commit b85b99dfac4fb29c136bba52779d32cef16cab3b)\n(cherry picked from commit 3b9df8e4b043471d875b8c112e04634a02ca64f6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9fc89a77ffa6c35eb2bf170762b1dcd7d9801e0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9fc89a77ffa6c35eb2bf170762b1dcd7d9801e0c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
