)]}'
{"id":"openstack%2Fpuppet-neutron~792564","triplet_id":"openstack%2Fpuppet-neutron~master~I056a6d5acc8312db2405795e2a52e60a0b902383","project":"openstack/puppet-neutron","branch":"master","topic":"pyvers-removal","hashtags":[],"change_id":"I056a6d5acc8312db2405795e2a52e60a0b902383","subject":"Get rid of the $pyvers variable","status":"MERGED","created":"2021-05-21 10:05:01.000000000","updated":"2021-06-18 10:28:22.000000000","submitted":"2021-06-18 10:28:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"792564-pyvers-removal","meta_rev_id":"d490795786d55ae76ff5a4dd162dcbbc6280f056","_number":792564,"virtual_id_number":792564,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-18 10:28:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-06-15 08:10:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-06-18 08:14:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-06-18 08:14:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-21 12:48:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-21 15:07:08.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2021-06-09 11:47:11.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-06-14 20:48:01.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-06-14 21:28:04.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-06-18 08:14:46.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"dc350cc1a01005692bf1a47c2f7b4585b97f91f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-05-21 10:05:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17eb49fa8b46b51aef8275191932269a90359de5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 12:48:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31b4eed509b949978e6cfb1503d4a065 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/339483d7b6b144e2a8bf56fdf43d3e27 : SUCCESS in 4m 05s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a68e64868a28411784a88015f499a449 : SUCCESS in 5m 10s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ea0f99d6e14044c5bf0f19c767683891 : FAILURE in 9m 02s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/d78fcf474dc64120bc7f6238e76caa90 : FAILURE in 10m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3757b63885e34ff58a262db164a9c085 : FAILURE in 8m 49s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de018800d95c43999cb7be690d3ee291 : FAILURE in 7m 17s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1b39966f4d24624b5251f6f9a0e1402 : FAILURE in 8m 07s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51c119ef4a574920899d09433d5f836c : FAILURE in 43m 29s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3943afe8a1114b3f9cfd9ac565524fd9 : FAILURE in 40m 13s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67e21f048c9e42328ff0037bfdb74593 : FAILURE in 30m 08s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25e74de39d984f0eab0982afe72f2917 : FAILURE in 29m 22s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4d7aa98efb343b4a48f9f528d30c657 : SUCCESS in 23m 37s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e0b36241552046c0bd306606ebef62a6 : SUCCESS in 1h 16m 52s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d428a3b7d164b58a29747d9e289cb93 : SUCCESS in 1h 14m 58s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0db1d65a020848ad9441a24d447b5202 : SUCCESS in 1h 09m 49s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e9d35bc645dc492eab15f3f7f6b045b1 : SUCCESS in 1h 19m 43s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e78567a4cf794aaf9bf3f1c43600c751 : SUCCESS in 21m 26s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8a0ea499c0c4484f88671e6b77f2b556 : SUCCESS in 1h 59m 52s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b4c068a1be24f85b889dbf8e5e46ccb : SUCCESS in 1h 31m 23s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ea30b87bdd2e4bb1afd4b395a2ea304f : SUCCESS in 1h 08m 39s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/92bf421d0c5347b1be6ccc2fddf73207 : SUCCESS in 50m 10s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c8d5e199f1da4207b497fd7fa56c9f93 : POST_FAILURE in 18m 56s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1201cb76fd74f1e96dc04c51d492828 : FAILURE in 19m 22s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/f7f047faf52d48daab5afc1964b6042b : SUCCESS in 21m 32s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ade9268a21c64b5d9eac0a520a44f3d6 : FAILURE in 19m 18s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd494a3ab81146b997d2c8342dc009d8 : FAILURE in 17m 38s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/aae580831e6c461296232e65be73a00b : SUCCESS in 2h 38m 06s","accounts_in_message":[],"_revision_number":1},{"id":"bcca215280b6909c28e906c8a56013b940b46541","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-21 15:07:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f11774087b5e4b157341df74af7b08ec35e688ac","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-06-09 11:32:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4631f7e3681d1f53e3da1cdb51cbbb9b0e7ab999","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-06-09 11:33:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f75ee89c59a8845713dc28d9b90125ec63d579c0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-09 11:47:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5f4802c438ae018c1ab6235d59a0954b4542fec8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 13:31:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef5fa02761cd45e99b6076839ada123b : SUCCESS in 3m 51s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a5f17aa757647c28c3c09618c71b4e8 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3327c20a57314aea9725707e44d777f6 : SUCCESS in 3m 38s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/153b609676624d8aaa21d862ad87d47b : FAILURE in 20m 42s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/926e8aa881d8435eb25e3b0d80d47de7 : FAILURE in 18m 52s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/996b31dc99064a62967132379c92079d : FAILURE in 14m 59s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0b908a8071b4be1b0d21d2b23cd493a : FAILURE in 15m 37s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c009688dfa664bd087f3c4badf90b051 : FAILURE in 17m 23s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01022b2261c9433bbe9e23815685326e : FAILURE in 42m 57s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd2aa6392c4149b9bfbcaff539bb3fac : FAILURE in 40m 02s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/711ceaa66c0c47daaa0f718f1ccd06b8 : FAILURE in 30m 42s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0aa4a81f5a8846c4893a66e34f5c0b73 : FAILURE in 31m 57s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f76191b94ccc4ecf95ff1e7aebb983fd : FAILURE in 17m 05s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/284f8ccb65be45bf8b5f85399262c50d : SUCCESS in 1h 14m 28s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/802868d437034120a2d41496930c0c9f : FAILURE in 15m 45s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9e67b0c91cf4e65bfdfe568a7b2e80e : SUCCESS in 1h 12m 09s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4bffcddca1de47109ebe993f75729d99 : FAILURE in 17m 47s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3225a35a91714306b2be989596ad4cb9 : FAILURE in 18m 24s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/982b63c7d53347b59deed70bdeaf7378 : FAILURE in 1h 28m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/25257f67a9a34ae796ddaea6ff58872c : SUCCESS in 1h 34m 36s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/968c10e1be7b4c569c55f870574f595f : SUCCESS in 1h 23m 43s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/99a00580c42546dd963a0d6e230e75e8 : SUCCESS in 1h 05m 04s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3a1275ac74b4f1b828f019d90cb87ca : SUCCESS in 25m 23s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68f6514b4fec4d41a228888e735cf767 : FAILURE in 17m 49s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6fd411431e54d8096b3d4cafeab4b76 : SUCCESS in 24m 10s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/709bde0a20dd4bfa838a16c43d20e0a2 : FAILURE in 18m 47s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c07fb4291584e2dbc5e6d3d44aeab4a : FAILURE in 19m 17s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/560ad305e10c4659be31ce4096f3951d : FAILURE in 1h 49m 49s","accounts_in_message":[],"_revision_number":2},{"id":"5c3002e57ab7f8e9bee4e0ae24978b89fa004974","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-06-09 15:46:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8edb1880581f179c606670edfc30782a3f962c6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 17:52:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/159f0cad8cc849d9a36f3887c2cb883c : SUCCESS in 4m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4867348c4a2448a7b7d0ad91b1f6c976 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4aaeada5fd874aa7aedf0a2198a2cb53 : SUCCESS in 4m 00s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5e00d7649f0149e8abea73971f76c9d6 : SUCCESS in 17m 32s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/c27c9a9fbfef4de782ae4890e72346be : SUCCESS in 18m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31ddb77abb6d4330816326159e6eb73c : SUCCESS in 14m 27s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20f5f563c7744d5c99ee8db5373247b9 : SUCCESS in 16m 47s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18f44fafd5664542afff8c79d494564c : SUCCESS in 15m 11s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59454c4798214dbab0424fa7e2b3dcc6 : FAILURE in 38m 57s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfcf34449adc4fc5bbb0d6f5c2849b25 : FAILURE in 41m 44s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a817a801accb4ea7b856ccbb60619cc0 : FAILURE in 29m 59s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/926592f1806c4c48a4bad71167a91dde : FAILURE in 29m 30s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/12daf04709444c4d92f4bc358ca1bb28 : SUCCESS in 24m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cdbfb0fee02e4e96b3e66d7e42a2a000 : SUCCESS in 1h 28m 10s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/be7fe60f9fc54f879003e2ccbb648be0 : TIMED_OUT in 2h 04m 57s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/997236248bc04da5a43411b8c3820249 : TIMED_OUT in 2h 04m 46s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3782957e571a43b89b9f813a0be94cc9 : SUCCESS in 1h 35m 38s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/142329fc98db41c0973e0d1819cf9410 : SUCCESS in 24m 22s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3fca0f65b50f431d9f3144d02249c637 : SUCCESS in 1h 44m 22s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f69fbfb8c43442ea1e2751f51da7790 : SUCCESS in 1h 21m 28s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a441bf1be10f4713abdf54e3a0e61304 : SUCCESS in 1h 19m 10s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8b82e53f62a04cfea83a4a1a88c91976 : SUCCESS in 58m 27s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/267e8717875c41cda6704fa49bcff220 : SUCCESS in 22m 27s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a3da2febf654ca29b11c9494d464e96 : FAILURE in 20m 26s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/916054b140714e1d8622d985a9b2b004 : SUCCESS in 27m 27s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ce6d4abb75a4b0094300727590ce8e3 : FAILURE in 18m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8dbe07ddffb483bb1482a79cf8f9fd3 : FAILURE in 19m 18s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/92879af3262f46ff8f6be92a0ab453c2 : FAILURE in 1h 57m 07s","accounts_in_message":[],"_revision_number":3},{"id":"e1a63bf2949c1bcb942d9535fcaff24c1ace8317","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-10 15:03:01.000000000","message":"Patch Set 3: Code-Review-1\n\nThe following two files should be updated as well.\n\nspec/classes/neutron_services_l2gw_spec.rb\nspec/classes/neutron_services_bgpvpn_spec.rb","accounts_in_message":[],"_revision_number":3},{"id":"b4237b4672eced1579ee88e7fcaf87001d859d9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-06-14 07:21:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dfa2388bf73f9efd7fb6bd457eece96a244d9d4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 09:45:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35fdc4d1baaa486b8eab66807afc8a91 : SUCCESS in 3m 46s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9001b0bdaa924bf4bb8da305a191b517 : SUCCESS in 3m 44s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5833eceb16346049e15588e5b5bf623 : SUCCESS in 3m 55s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/50c52a5d0c5f404499c929facb48b852 : SUCCESS in 15m 14s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/c5f928bece8b4de686b08836829cc595 : SUCCESS in 16m 09s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01bec277355f41b985e721d81af68f43 : SUCCESS in 14m 08s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecf6ecd13991442bbcf24e2801cb88f7 : SUCCESS in 14m 18s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/094d091438e341f5b772e340ce236b02 : SUCCESS in 15m 03s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfd78b8fa7db49c28680ea91fd07fd2e : FAILURE in 36m 45s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bc64b9f37be475984e41e057f85058d : FAILURE in 42m 10s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcfc61ef1ec841b49d3b25ec593d2f56 : FAILURE in 30m 07s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92ab831bf1144667b82c7eddf1c35d52 : FAILURE in 29m 10s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/77e34a02bbe640e0ba59ed5dfcea9bc7 : SUCCESS in 24m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f06238defba34cca88364c830127c41c : SUCCESS in 1h 23m 37s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2e30aadf7824eb0a9acd8b2c083a157 : SUCCESS in 1h 45m 01s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a3592184fa3e412f9869346499889ebb : SUCCESS in 1h 30m 02s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/514f81fa747b4c169e199dc2b8cb0f9d : SUCCESS in 1h 09m 46s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cbdafcd1ded24400a86e1d96e049a4da : SUCCESS in 21m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/73eebfdee6db4eafbc561308bac7c7a2 : SUCCESS in 1h 21m 56s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9f901ef0c53b4b3fba78fbce55259a4e : SUCCESS in 1h 50m 42s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/386631b95b6e4782a21f854bec059373 : SUCCESS in 1h 27m 28s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ae27bb8851e4150b33108f54d744456 : SUCCESS in 1h 08m 48s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb1926024dc0448fb74c172871a9ad07 : SUCCESS in 22m 26s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd4c899c3a8641a39406fbb4f32f0894 : FAILURE in 18m 57s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/78e3f3cd98004280b971f65f4d485b38 : SUCCESS in 24m 32s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e7a273d5e1148c1a697b567eadec689 : FAILURE in 17m 31s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd3290f5b88e4b718c48409b3957c326 : FAILURE in 16m 57s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2b8a827d5d514c95a8e9f9f4a1c2597b : SUCCESS in 2h 12m 27s","accounts_in_message":[],"_revision_number":4},{"id":"f4f3509051a8e9bf5b91db50e6f5878c1fd68138","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-14 20:48:01.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"84afbd5a530bef5e16e8eeca9bf1d3f2e2188a94","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-14 21:28:04.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"2c7a77d7782937cfd050d6b30fe7f63fb68b73af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-15 08:10:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"08fbbf163322f1fcdca7f302a0dc1ea57e68a65b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-15 08:10:52.000000000","message":"Patch Set 5: Code-Review+2\n\nUpdated spec/classes/neutron_plugins_ml2_bagpipe_spec.rb to remove remaining logic which expects pyver.","accounts_in_message":[],"_revision_number":5},{"id":"eea0f28c81dcbd3905065c817789caf944facd8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 10:17:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdb4a4dd42604db7ab095ab03f78ab9f : SUCCESS in 4m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d10ea2dc6b047699abc6a6cb0a889e2 : SUCCESS in 4m 05s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c62149d90165413faa43faf0a78c68ea : SUCCESS in 4m 12s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/fa0ff6e7ab964490b10eee32dea0d92b : SUCCESS in 15m 38s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/03da3e9867ed4ee4aa94bcea1023ce7b : SUCCESS in 23m 28s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8ba4e52e28a41139d73141ca773286f : SUCCESS in 17m 46s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bdee472131647c9b2cc649f1cccaf55 : SUCCESS in 14m 35s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6af3d96468554dfa9cf9a668aba243d8 : SUCCESS in 14m 14s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0686091574c843d1bf0c34d99eaf89bc : FAILURE in 46m 03s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc18090b248d43799dad11de7bc1ec75 : FAILURE in 41m 00s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7d9ee2730404f9c94e3b8688ccd0601 : FAILURE in 31m 15s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03ce4159c00242debb87eea2a0ae2f4c : FAILURE in 32m 40s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6d6bdc86be4c4270bd4ab2ab70138976 : SUCCESS in 25m 49s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/048af6d797214d5894e160537c9100be : SUCCESS in 1h 11m 31s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/90817b21aeef4ab18b8bf8011ec24e1f : SUCCESS in 1h 25m 49s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/935b85d7f3c04d11af9e60f9f75f308d : SUCCESS in 1h 19m 36s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d2d0c99ed824474aaeaf8668cafc74f9 : SUCCESS in 1h 12m 08s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/56b06d52e54c403381151d77e0b5d15e : SUCCESS in 29m 27s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf877573d78a45f495d2197ce1be8501 : FAILURE in 21m 43s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bcf9deff8af7412f990b4da952ab2d4e : FAILURE in 16m 29s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/17bfa457c1624af89673a20bc2f1df6d : SUCCESS in 1h 18m 41s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/19261a12eace43a8b19012fa3bd361c2 : SUCCESS in 1h 29m 17s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/85b3d37c1e0d4441ae38c0cc1d695ecb : SUCCESS in 25m 26s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71269c6abaf447c98cc59440e06473e1 : FAILURE in 18m 10s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf3823e949584d98bc54e0788bd619c5 : SUCCESS in 24m 52s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/588a5c399e8b4907a3a43908a7b96150 : FAILURE in 16m 34s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53b1749ff9b841e6a83b7557f19ef755 : FAILURE in 17m 02s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b15f57a5104740819b4eab5c58c56462 : SUCCESS in 2h 03m 45s","accounts_in_message":[],"_revision_number":5},{"id":"75a9e472e5f034285a53ae3490aac7b3481e36c2","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-06-15 10:57:41.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"2768a5b3dcc3d0faaaa43383c05b792d96d1bb55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 13:04:07.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55eaa1aee9f1410d8ba7b25979612d87 : POST_FAILURE in 6m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0da39ddbdfa4bfa95012ca03c73cf77 : SUCCESS in 3m 33s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56772dab8f744b04a6ce1d3cb7537fda : SUCCESS in 4m 49s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f6a6630ceef54b95af4722514c433298 : SUCCESS in 16m 02s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a83c4d35dbd496496fb9222a4f65e3d : SUCCESS in 19m 45s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/208002ab9eaf433da55183d0e74f236d : SUCCESS in 14m 03s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a475f392f294125bbe6f31c0ef0c063 : SUCCESS in 13m 27s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf23588e828447f290048d98f85d8108 : SUCCESS in 16m 12s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf4bad0249b84ebf964aaa76dbd03781 : FAILURE in 44m 58s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f6e1ee8dc8a4f2cb259cfe44350601c : FAILURE in 45m 00s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4476253d7c0b48679eae6580b4c706e7 : FAILURE in 28m 40s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8bb37b1e48541ebaa736954e931c0bf : FAILURE in 34m 39s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a94233e9d91540209963f88eb4913a88 : SUCCESS in 30m 02s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3d7e5966891b473c84f78423f3ad557f : SUCCESS in 1h 14m 47s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c868681bef74a549115eaa71c482d52 : SUCCESS in 1h 41m 01s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/18ec446834b34c7e891eda14c78689f7 : SUCCESS in 1h 18m 42s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1253de5c66b5497ebbd63cf3b980d6e2 : SUCCESS in 1h 03m 49s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/18efced19d8544cc963bc46042516e5a : SUCCESS in 22m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/40bd93983ce346a3a793ff491ecdf6d5 : SUCCESS in 1h 10m 53s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5fba7c68d530475fad134cb3cc480356 : SUCCESS in 1h 35m 42s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7d558429db82404aac153ea32efd23ab : SUCCESS in 1h 16m 20s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8d5336b6d8624e9d8b39fe63852c4b29 : SUCCESS in 55m 37s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6ecda9a184284e449002d28d6a5da8d3 : SUCCESS in 26m 13s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fb97265ee834fc8b13f81b8bfd5548f : POST_FAILURE in 23m 49s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/67887ca7ec554a08a9bad98d858d206d : SUCCESS in 22m 51s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ab674758546406c95a10de44e80b455 : FAILURE in 20m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32ac6dab62c146578122ced1974f9bec : FAILURE in 18m 01s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/70ab7cfb851240c58b24c24cb9bd9757 : SUCCESS in 2h 04m 33s","accounts_in_message":[],"_revision_number":5},{"id":"bf1ac215094040cd22752df66964791de1a02d34","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-06-15 13:38:11.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"66a89c4e9966abae4b26b32dd473787c90a76f93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 15:44:51.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee92244230854ce0a153769621b91e0c : SUCCESS in 4m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52e903581fbd42f3822fea975cf9f1ab : SUCCESS in 4m 15s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecbe05d81d884c18b603d90c39660cd4 : SUCCESS in 4m 00s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/41574dfbb9ac4c9e8ada8165f3fefbde : SUCCESS in 20m 48s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/66c9be4e9d75410c971292f39709cc83 : SUCCESS in 19m 05s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/563a572a230048c5afc59db49e31412b : SUCCESS in 15m 52s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8b72a4db54b420ba56cd48fd9729b04 : SUCCESS in 15m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e71f5cbd512e46db8cf82052e9a29fad : SUCCESS in 15m 43s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80a025e0fae74227a5256da5d5f04e16 : FAILURE in 35m 25s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b29470ae5b474e509f434a580bee1a23 : FAILURE in 35m 53s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/660a7fd24b9f45a19264cb8852beb1fd : FAILURE in 34m 05s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b92c73683ab34143b93363e70ad7ad5a : FAILURE in 28m 49s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d1df32fb0d974e62b30f46c40d960e28 : FAILURE in 15m 42s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/50575ea50ed44d7f84ea9c7a691b129c : FAILURE in 15m 27s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/583d96d5b7d64067b073ed1f633c1b4b : SUCCESS in 1h 48m 25s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2be4caa91bb84af5af9a5144f84069cc : FAILURE in 17m 11s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4f86cffe9024c598aeaf61835085ff6 : SUCCESS in 1h 08m 21s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d4843e7353642038789bc58183627bc : SUCCESS in 27m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/85f71b2d340b48c7b2d27f08946e8ae0 : SUCCESS in 1h 15m 11s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/21c52adfd07f41baa4ae5f81e3b65ab0 : SUCCESS in 1h 36m 55s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8b8b5ee78114e5886421b745ba87720 : SUCCESS in 1h 28m 57s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4bff32ae1d384c75b2a36ca0b2979eee : SUCCESS in 1h 03m 52s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8786afdec3d74972b9581359000671f7 : SUCCESS in 24m 13s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/598fec77d6b7452b804033c40b043348 : FAILURE in 20m 08s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/5e34dcf7bcd1474f8dd4639d3e3a19bf : SUCCESS in 28m 46s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe5f96d87be14c5ba5c471e520e82fb8 : FAILURE in 17m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/116703fa09bd46cfbc3b6b54cd5688e6 : FAILURE in 16m 53s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b2053382e0b84167823c4a9fbbe7f323 : SUCCESS in 2h 03m 57s","accounts_in_message":[],"_revision_number":5},{"id":"0aec85f7b5b5dfb92d7b34e2d82518dc7e3ed0f8","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-06-17 09:51:10.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"434f00249253dc95759a50fd6aa1fa9bc55df216","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 11:52:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbe6528d05114fa3bd948f1c4a004e09 : SUCCESS in 4m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6649f612547f4d1ba43740315acbfe38 : SUCCESS in 3m 53s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f4a8483a1b14a02be4c21e9dd9bce86 : SUCCESS in 4m 01s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ec1d6723d444c23a8efdda8eeecdd6d : SUCCESS in 18m 46s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/25b22ea1109b4175a3ce5e2dcc144c66 : SUCCESS in 19m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9866a9c10144083a7c199d9a8550dc5 : SUCCESS in 15m 54s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29f25d164bf54fc59fa6fb9657b65d62 : SUCCESS in 14m 48s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a922d76e1d08440b8130809393173ed2 : SUCCESS in 15m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5d93d4653014e1c9b486f27afa72348 : FAILURE in 44m 27s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5350cb72afa349cfa4cc27d5b9c5d64a : FAILURE in 42m 15s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9e3172d4705447ebc2cac3fd8041999 : FAILURE in 29m 51s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ddd397969a648628aca49e560c7e0f6 : FAILURE in 37m 42s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9de65db7414a4891b81efb2ee0fb764f : SUCCESS in 26m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1890aee5963740179d6468bbdd86980f : SUCCESS in 1h 13m 37s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/628360367b2d4813a7265955d01002d3 : SUCCESS in 1h 29m 53s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/413cea0e8f2b41fda604d81e35b9e20a : SUCCESS in 1h 23m 14s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fb3940441bfc46a3929e3eda14ed4ee9 : SUCCESS in 1h 21m 39s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7c4a91055b774fd3a44d6edf5b05c26d : SUCCESS in 26m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/52fac0eb100c49cfab74707cd1efd448 : SUCCESS in 1h 14m 11s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b65439c787040ce951ebd580841bfa9 : SUCCESS in 1h 35m 51s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/51d2599b70664159ad28d88d17d45587 : SUCCESS in 1h 26m 43s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2aafec7a033248fb8ad2dd269b8409e2 : SUCCESS in 1h 01m 49s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/074606539ef640dca3b14c8542807270 : SUCCESS in 25m 01s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4e8c197560846f0bca2d991571ed554 : FAILURE in 19m 07s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/77ada63f967c49488955dcc803f674e5 : SUCCESS in 27m 42s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9ab5afee0364ebebea20e6e1912fbe3 : FAILURE in 18m 47s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b67f629db474802bba6ed55503718f3 : FAILURE in 18m 07s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bc52aad197dd4f7a9e185a48c9a739d0 : SUCCESS in 1h 54m 26s","accounts_in_message":[],"_revision_number":5},{"id":"f356d7a49cd21116f61e730987887dc0c6738222","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-06-18 08:14:46.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1c69eff98bbe6cdc772a1ad2fda6d32d09bac23f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 08:16:12.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ecb22a88c4269049707546f3b2b586d3163dc8eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 10:28:20.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3295dcc2ee340f3ba48f2e58c315cba : SUCCESS in 4m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fca4e5fa22184c6ebe705f3edf4d1462 : SUCCESS in 4m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dac64b2e0f274e1186b9df454971942a : SUCCESS in 4m 52s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5a3d3876ba254833a6a966aad2873647 : SUCCESS in 19m 08s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/a06992667a7b456e9dc347aec43d001f : SUCCESS in 19m 39s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4f0ef0d659a49189c5170b39cbd5b30 : SUCCESS in 15m 12s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da7466dec8e04a7495ae3a48e7626a6b : SUCCESS in 15m 50s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fd934c4bf16b42a3a961b0dd28f23f24 : SUCCESS in 1h 17m 09s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0bfbdcb62f254bada2d0822d1f42c1ae : SUCCESS in 1h 37m 08s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dea2750530ea44498f121feb2659447d : SUCCESS in 1h 34m 58s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9959191ea6cc4400bc4eee85c67b0562 : SUCCESS in 1h 08m 36s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b666a1b010594f0da283ab4d8b09d9d9 : SUCCESS in 25m 25s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b001f22825004633be781379c2113e28 : FAILURE in 17m 13s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9e63c8fb217c4ec9a64272ab4cb5fc83 : SUCCESS in 2h 06m 46s","accounts_in_message":[],"_revision_number":5},{"id":"d490795786d55ae76ff5a4dd162dcbbc6280f056","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 10:28:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b52c14963e82aedf8e8d02bca1950808a76a796c","revisions":{"8f48bffcb79f054ced1c42b39d11f9edcef52172":{"kind":"REWORK","_number":1,"created":"2021-05-21 10:05:01.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/64/792564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/64/792564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/64/792564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/64/792564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/64/792564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/64/792564/1"}}},"commit":{"parents":[{"commit":"8522ba85566feca76ed0bea2ce71d0e5d554c0e0","subject":"Merge \"Drop old workaround to install keystoneclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/8522ba85566feca76ed0bea2ce71d0e5d554c0e0"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-05-21 10:04:46.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-05-21 10:04:46.000000000","tz":120},"subject":"Get rid of the $pyvers variable","message":"Get rid of the $pyvers variable\n\nSince everyone has switched to Python3, it\u0027s time for the removal of the\n$pyvers variable.\n\nChange-Id: I056a6d5acc8312db2405795e2a52e60a0b902383\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/8f48bffcb79f054ced1c42b39d11f9edcef52172"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/8f48bffcb79f054ced1c42b39d11f9edcef52172"}]},"branch":"refs/heads/master"},"71ca0619e8cc31038799a3147727aa5c9b653b33":{"kind":"REWORK","_number":2,"created":"2021-06-09 11:33:12.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/64/792564/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/64/792564/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/64/792564/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/64/792564/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/64/792564/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/64/792564/2"}}},"commit":{"parents":[{"commit":"8522ba85566feca76ed0bea2ce71d0e5d554c0e0","subject":"Merge \"Drop old workaround to install keystoneclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/8522ba85566feca76ed0bea2ce71d0e5d554c0e0"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-05-21 10:04:46.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-06-09 11:33:01.000000000","tz":120},"subject":"Get rid of the $pyvers variable","message":"Get rid of the $pyvers variable\n\nSince everyone has switched to Python3, it\u0027s time for the removal of the\n$pyvers variable.\n\nChange-Id: I056a6d5acc8312db2405795e2a52e60a0b902383\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/71ca0619e8cc31038799a3147727aa5c9b653b33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/71ca0619e8cc31038799a3147727aa5c9b653b33"}]},"branch":"refs/heads/master"},"c1283b8a834e57970c32a31664354cd7b2fdc0c4":{"kind":"REWORK","_number":3,"created":"2021-06-09 15:46:01.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/64/792564/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/64/792564/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/64/792564/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/64/792564/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/64/792564/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/64/792564/3"}}},"commit":{"parents":[{"commit":"8522ba85566feca76ed0bea2ce71d0e5d554c0e0","subject":"Merge \"Drop old workaround to install keystoneclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/8522ba85566feca76ed0bea2ce71d0e5d554c0e0"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-05-21 10:04:46.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-06-09 15:45:53.000000000","tz":120},"subject":"Get rid of the $pyvers variable","message":"Get rid of the $pyvers variable\n\nSince everyone has switched to Python3, it\u0027s time for the removal of the\n$pyvers variable.\n\nChange-Id: I056a6d5acc8312db2405795e2a52e60a0b902383\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c1283b8a834e57970c32a31664354cd7b2fdc0c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c1283b8a834e57970c32a31664354cd7b2fdc0c4"}]},"branch":"refs/heads/master"},"c09272c3b99a2d8cd235c9addcaec0a9c7ef99f2":{"kind":"REWORK","_number":4,"created":"2021-06-14 07:21:27.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/64/792564/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/64/792564/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/64/792564/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/64/792564/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/64/792564/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/64/792564/4"}}},"commit":{"parents":[{"commit":"8522ba85566feca76ed0bea2ce71d0e5d554c0e0","subject":"Merge \"Drop old workaround to install keystoneclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/8522ba85566feca76ed0bea2ce71d0e5d554c0e0"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-05-21 10:04:46.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-06-14 07:21:11.000000000","tz":120},"subject":"Get rid of the $pyvers variable","message":"Get rid of the $pyvers variable\n\nSince everyone has switched to Python3, it\u0027s time for the removal of the\n$pyvers variable.\n\nChange-Id: I056a6d5acc8312db2405795e2a52e60a0b902383\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c09272c3b99a2d8cd235c9addcaec0a9c7ef99f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c09272c3b99a2d8cd235c9addcaec0a9c7ef99f2"}]},"branch":"refs/heads/master"},"b52c14963e82aedf8e8d02bca1950808a76a796c":{"kind":"REWORK","_number":5,"created":"2021-06-15 08:10:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/64/792564/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/64/792564/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/64/792564/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/64/792564/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/64/792564/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/64/792564/5"}}},"commit":{"parents":[{"commit":"8522ba85566feca76ed0bea2ce71d0e5d554c0e0","subject":"Merge \"Drop old workaround to install keystoneclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/8522ba85566feca76ed0bea2ce71d0e5d554c0e0"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-05-21 10:04:46.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-15 08:10:07.000000000","tz":540},"subject":"Get rid of the $pyvers variable","message":"Get rid of the $pyvers variable\n\nSince everyone has switched to Python3, it\u0027s time for the removal of the\n$pyvers variable.\n\nChange-Id: I056a6d5acc8312db2405795e2a52e60a0b902383\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b52c14963e82aedf8e8d02bca1950808a76a796c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b52c14963e82aedf8e8d02bca1950808a76a796c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
