)]}'
{"id":"openstack%2Fpuppet-neutron~796359","triplet_id":"openstack%2Fpuppet-neutron~master~I545f467d50b5eebbae26a6fd9e9fb7b35f8b84f8","project":"openstack/puppet-neutron","branch":"master","hashtags":[],"change_id":"I545f467d50b5eebbae26a6fd9e9fb7b35f8b84f8","subject":"Update networking-baremetal parameters","status":"MERGED","created":"2021-06-15 03:30:29.000000000","updated":"2021-07-23 05:49:24.000000000","submitted":"2021-07-22 22:44:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"796359","meta_rev_id":"1f16050d59e65982069ae6244f080b4f9e1c507b","_number":796359,"virtual_id_number":796359,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-22 22:44:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":1,"date":"2021-06-28 06:48:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2021-07-01 04:46:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":2,"date":"2021-07-21 04:55:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-06-28 20:40:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":1,"date":"2021-07-21 04:55:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"date":"2021-06-28 20:40:59.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}],"CC":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-15 05:36:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-15 08:08:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-06-28 20:40:54.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-07-01 04:46:40.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2021-07-23 05:49:24.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"CC"}],"messages":[{"id":"c15fec62611116baf00dabe55c90822e66ce9763","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-06-15 03:30:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bbb70210b9bab282a145968456bfe715afcec88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 05:36:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56793b6ddf86451d81d7e7a913b0a089 : SUCCESS in 4m 10s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b00e59fcd7c3470ba3615c5be0891853 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ff88cc10e854465aecd334795b737a0 : SUCCESS in 3m 40s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef3610e6468d4cd98fbda603f177e742 : SUCCESS in 15m 55s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/16a2bdf9663c4e40ad391831731b91a1 : SUCCESS in 15m 45s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d176704c91624ab0943221e844d76be7 : SUCCESS in 12m 43s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/576604fafee2419096c1a9123f0f28c0 : SUCCESS in 12m 16s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78f7c33d8a9d4649a4e6cf142083310e : SUCCESS in 12m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a70488fc0214869abf2c0ba5830f430 : FAILURE in 37m 15s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e76ac21079d4782949967935b30b8b8 : FAILURE in 42m 02s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c57fc4507fc4cac898c34a8cb5acb09 : FAILURE in 43m 38s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ad8a5162ff64400a3c93d7aeb70fa8f : FAILURE in 28m 49s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dadc46054e434a30a1114626e4734310 : SUCCESS in 24m 00s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b28d1dcdc87a47e880fc09eae57e76c2 : SUCCESS in 1h 31m 44s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c45d04f2a8d4463aa47a5c8582079371 : SUCCESS in 1h 29m 06s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/71276a7c170c4c3aac9219741fdc5460 : SUCCESS in 1h 04m 03s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba9290189a934cd5832db60fb71bde1b : SUCCESS in 58m 03s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a909c0d293d544e9aa9b6b1027c6efd2 : SUCCESS in 21m 16s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/351176655a274307ae7d9ede937f99f0 : SUCCESS in 1h 18m 36s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0baeaf6e3ab494ab490f3fb1d5d5dac : SUCCESS in 1h 40m 25s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d62e4b394114614afb3d899a3abcce4 : SUCCESS in 1h 11m 43s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d3d69e472f584dda95073a9fd2e89130 : SUCCESS in 1h 08m 04s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ee9e279e97f140289122b039a3ef26a7 : SUCCESS in 22m 55s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42903ce54b1f4e55a0ef8ed7c4cf8686 : FAILURE in 15m 23s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/49bf27ef52f34f42b58851ddbb3d1d19 : SUCCESS in 22m 11s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6945b441ca174113994da20aa8664b50 : FAILURE in 18m 35s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6f2b33068da42f09c9312e28e116372 : FAILURE in 19m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a67d85ce483471284952fa82349e400 : SUCCESS in 3m 50s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/254c82aa7da646d28cf1cb72b61cae5b : SUCCESS in 2h 04m 06s","accounts_in_message":[],"_revision_number":1},{"id":"ac2f3ef68329c02070f8b54ab2c10beceab80667","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-15 08:08:15.000000000","message":"Patch Set 1: Code-Review-1\n\n(6 comments)\n\nThank you for submitting this patch ! Please find my comments online.\nMost of the are trivial formatting fixes but the usage of deprecated parameter is the most important point.","accounts_in_message":[],"_revision_number":1},{"id":"93283986a6418cd88ffdc1a794f4b024cb8fd76c","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-06-15 08:20:06.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"295904598bfad7076b61d5631561f5cb746b7a77","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-06-15 08:21:45.000000000","message":"Patch Set 1:\n\n\u003e Most of the are trivial formatting fixes but the usage of deprecated parameter is the most important point.\n\nironic_url is useless atm and does not work. We should not have any issue in ignoring it.","accounts_in_message":[],"_revision_number":1},{"id":"9996471f4be935c2d050b3de7ac77d81e9448cac","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-15 08:32:06.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Most of the are trivial formatting fixes but the usage of deprecated parameter is the most important point.\n\u003e \n\u003e ironic_url is useless atm and does not work. We should not have any issue in ignoring it.\n\nI\u0027ve checked the latest networking-baremetal but there still exists the logic to accept ironic_url.\n\nhttps://github.com/openstack/networking-baremetal/blob/master/networking_baremetal/ironic_client.py#L26-L27\n\nIf this doesn\u0027t work then that is the bug in networking-baremetal IMO...","accounts_in_message":[],"_revision_number":1},{"id":"815d17d89f0e9750be653438af893453f63b4159","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-15 08:35:17.000000000","message":"Patch Set 1:\n\nAlso looking at the commit which deprecated the parameter, it seems we need to rename/update retry_interval, max_retries and auth_strategy as well ...\n\nhttps://review.opendev.org/c/openstack/networking-baremetal/+/683971","accounts_in_message":[],"_revision_number":1},{"id":"e8c62f462188435caff3d8d1ad9de69a9cabbdc2","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-06-15 08:42:34.000000000","message":"Patch Set 1:\n\nauth_type is already there. The change I proposed is to make it work with standalone ironic, which works only with endpoint_override and auth_type. I did not intend to take care of all deprecated parameters.","accounts_in_message":[],"_revision_number":1},{"id":"ff5edd7823a2f1042ba4170aff8baea1f8f15e62","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-15 09:00:06.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e auth_type is already there. The change I proposed is to make it work with standalone ironic, which works only with endpoint_override and auth_type. I did not intend to take care of all deprecated parameters.\n\nHmm.... From puppet\u0027s perspective it is important to make \"consistent\" change here to avoid leaving outdated parameters partially, and I prefer covering all parameters instead of only some portions affecting specific use case.","accounts_in_message":[],"_revision_number":1},{"id":"dd6a0a17651193d409870628fcfd05d987fb0a26","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-15 09:56:03.000000000","message":"Patch Set 1:\n\nIf you are using auth_type\u003dnone then another option is to use the endpoint parameter which seems to be more native way to set endpoint ?\n https://github.com/openstack/keystoneauth/blob/master/keystoneauth1/loading/_plugins/noauth.py#L37-L38","accounts_in_message":[],"_revision_number":1},{"id":"48a7c3c2e23fd798ebe91a05183d7b8224abf63e","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-06-15 11:09:43.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"86f11241211ac693a21a131606735dba951f0ce2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-28 05:40:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e7f86405a2a6656ad87ee2ec58fe063c8aac5476","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-06-28 06:48:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4f3fce5e0be5674608c61c406afc6018e2bc459a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 07:36:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c97d207bb42429ea0782958c1922390 : SUCCESS in 5m 17s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16307542fce44c52893d95fdd97d1ad3 : SUCCESS in 5m 19s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d585b9ea9214dd1a2d2e484f8796779 : SUCCESS in 5m 08s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e0ae9766a95c4b229f41f9717cb6fe45 : SUCCESS in 18m 07s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/07dcad5172b54b7baff1df96a7fa0ec8 : SUCCESS in 18m 23s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fdac949594f4a9aa914b5a4a44ebe6e : SUCCESS in 16m 14s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2a0d81d6cb949a7bc998d5f34484ead : SUCCESS in 15m 24s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/402a37b7079c4d76af656ef74056eae7 : SUCCESS in 15m 24s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4c53d916f5c409a822db10326c86d44 : FAILURE in 43m 58s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eed11b7c4d5c46aba6966487b497f698 : FAILURE in 43m 31s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c55e49dd55d466792099721cf40d436 : FAILURE in 30m 09s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a260459b84fa4e0da187527e48ff52c8 : FAILURE in 32m 33s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d8f921078f1644e9a266be97113e2822 : SUCCESS in 27m 27s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2fec29fd2d2d4ddb9db6e0f74f6b0c53 : SUCCESS in 1h 22m 05s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/74df9bfa99af4f0ab9b81d468e61bd0a : SUCCESS in 1h 30m 11s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/811a27a5ed5e48aca5b344ca2faa4b4c : SUCCESS in 1h 33m 57s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6964ad8bbfa1430c8e1eddf4c9575caa : SUCCESS in 1h 08m 25s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc5e2c4831d64f688bdac143cf67ac75 : SUCCESS in 24m 16s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f59dc9acc484ba688dab4256cb3898b : SUCCESS in 1h 17m 05s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ce4704f7fd0467998c2caa8807a4b08 : SUCCESS in 1h 29m 21s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/22ac0b31718b400cb635fdb7307f8e91 : SUCCESS in 1h 20m 34s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6ae8e95c0734bd3a80e883a087c05f8 : SUCCESS in 1h 06m 14s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e4b8885c9724435c859770e26bdb84cb : SUCCESS in 24m 15s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35c8fb8bc5da45609919883c28978367 : FAILURE in 17m 21s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/6e6d7da2ee184ba89dfd2b8e82ffb595 : SUCCESS in 23m 55s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b825107755e348faba412e394e447344 : FAILURE in 17m 06s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f831be3ca194a28a0236d2473a2a516 : FAILURE in 17m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d4d05ed07c24731ac6e8f73d543b5f3 : SUCCESS in 5m 14s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3e147ed09ebb4824abd19248aea0200b : SUCCESS in 1h 52m 07s","accounts_in_message":[],"_revision_number":2},{"id":"6f9d173d533a906f648156140883eedeec4e385a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-28 20:40:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b85d93c35d543d2673209b110e66a804f761c7f0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-28 20:40:59.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"328ff88b7e48af3ba275e7a8f2922633f047a185","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2021-07-01 04:46:40.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"85939aae928c800b96d6e5e30c66293161f5456b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-21 04:55:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"075f1869d64e0ddf4cbb297f39988460aa733ea7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 04:56:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2d65cf04cf4d5eb90978abf919522742a3f82276","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-22 18:33:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4ac9c12a36bbdf02551f2561a717512da353800c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 20:54:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/282dfd1ecdc24a689d0249e87cc57e3c : SUCCESS in 4m 38s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/daf213dcc582456fa10f5a2d4c247d31 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/febaf80eb5b245a883e1cf6a21226ec0 : SUCCESS in 4m 10s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/595bd59baa094e4badf69803719808ac : SUCCESS in 18m 35s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/c092e1ced85e4867b25f1d0fa817dbca : SUCCESS in 20m 45s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/701b054e2f994999befce837302c307c : SUCCESS in 16m 59s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/116fd90ddbea495291f07c9256efaf58 : SUCCESS in 14m 18s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d131ab193bb464b866917c5a71f329f : SUCCESS in 15m 43s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/204f7e6673634bbba73a4e0c63c215e9 : FAILURE in 45m 32s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83cc1306861d4b79b21b9cbaa5ba060e : FAILURE in 43m 04s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0ced3ac45954a1b959175712df9bf24 : FAILURE in 29m 52s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a93735dea85943eebdfb8d7e65124f32 : FAILURE in 31m 41s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e19efda2801a4727b8cc06fe6466480f : SUCCESS in 25m 51s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc432e2e9b4749d4803596baf06abaff : SUCCESS in 1h 11m 32s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0dee7cc0eb1146f5a3a13b27d90fa38c : SUCCESS in 1h 40m 30s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c3388a12503a4a39b24c559586032dbf : SUCCESS in 1h 13m 53s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0bfc28f748c64aa0a45674441f3db43b : SUCCESS in 1h 05m 47s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5d80f548f2ec417194ee7913af167536 : SUCCESS in 24m 39s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8c6ee5134e024095a0e8deea83e8bd18 : SUCCESS in 1h 17m 28s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b2a84f9b0445465c95d78858c9dc72b1 : SUCCESS in 1h 26m 47s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f5aa87dcc08841048f11059d34a09636 : SUCCESS in 1h 34m 45s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/61e162d13edd4980b1309ef6940436d2 : SUCCESS in 1h 12m 20s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/bea44a3c3f1d44e2abd20b16df1b239b : SUCCESS in 24m 24s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44a28b8075c24614b55e33901f038f24 : FAILURE in 19m 17s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/cdddc98305ea4f2aa9c53b63a9bef91c : SUCCESS in 27m 43s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a833dd2edc84756ad80221dce12b808 : FAILURE in 18m 04s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7782cc5f8ff54e31a7e46f6620e6280f : FAILURE in 18m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/289593fd518e4aa3add41422becb10d9 : SUCCESS in 4m 49s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b996e696ae6d4225ba14c57f1d50887c : SUCCESS in 2h 08m 21s","accounts_in_message":[],"_revision_number":2},{"id":"de4e7cef08f10edcc18e9d7f582460b491a71401","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 20:55:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c9f1ba5f9a28b23ca295d067dff95b1b52f006df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 22:44:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9db04e037a824304ac674e9fd534caf6 : SUCCESS in 4m 06s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3b9027fae3a4c0f9d6a0fc5795a54b5 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b84521b063b440aca338ffff081c30d5 : SUCCESS in 4m 10s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6451593ca9a84c4db6ee355ae81e89e0 : SUCCESS in 16m 47s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/7403bfec3d7a4d93a7b683f2b0032e1a : SUCCESS in 16m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bea5a9daafd4f5ca6a6e81cc2bf57ed : SUCCESS in 12m 44s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44ecef9ca3a543f49f9e9d8e0b95bc99 : SUCCESS in 15m 42s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/eaff4172a97c4bf3b4f7775cc1f23457 : SUCCESS in 1h 10m 46s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/643be74424ca407aa2c0ec77000a3bd0 : SUCCESS in 1h 17m 33s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/45d50d3700674191bd0b391d811bd898 : SUCCESS in 1h 18m 31s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/30707ce998e14db2a7e937b327c7000f : SUCCESS in 1h 10m 58s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/275de73da6c0476a9f6b7be503d52252 : SUCCESS in 24m 06s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f15823a0105d4e99a3602440aa0df134 : FAILURE in 19m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb2ab8d38e9144c48072ab587eea0cad : SUCCESS in 4m 52s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cb31cc7cdab948e7977766b139c8d73e : SUCCESS in 1h 47m 40s","accounts_in_message":[],"_revision_number":2},{"id":"f0b1326a453e134f47f1f5f33e4634c9a8fe2602","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 22:44:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"52bdd494f1f39a370005d651a1d33d1400295a00","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 22:45:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f46c9667eaf4f0f9f6953a21d137a8b : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2},{"id":"c78cc2d844b9b143af988a6085df7b9d724b563f","tag":"autogenerated:gerrit:revert","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-23 05:47:57.000000000","message":"Created a revert of this change as I4e75e4ec3c34360be04d144f9885b957a39b1f9f","accounts_in_message":[],"_revision_number":2},{"id":"1f16050d59e65982069ae6244f080b4f9e1c507b","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-23 05:49:24.000000000","message":"Patch Set 2:\n\nfolks I just hit the revert button because this caused a new gate blocker filed at https://bugs.launchpad.net/tripleo/+bug/1937333 \n\nplease check the bug and see if you would rather do something different than https://review.opendev.org/c/openstack/puppet-neutron/+/801879","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"09429d0ebc6bff6aed5808faf7c4430e72d2dafb","revisions":{"52386b317a4e50de170b3b52dc5436f90d21149e":{"kind":"REWORK","_number":1,"created":"2021-06-15 03:30:29.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/59/796359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/59/796359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/59/796359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/59/796359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/59/796359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/59/796359/1"}}},"commit":{"parents":[{"commit":"ea79d50ef5df27457c78220a0594fad2d5848f9b","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ea79d50ef5df27457c78220a0594fad2d5848f9b"}]}],"author":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-06-15 03:14:54.000000000","tz":330},"committer":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-06-15 03:30:26.000000000","tz":330},"subject":"Add ironic endpoint_override parameter","message":"Add ironic endpoint_override parameter\n\nAdds endpoint_override parameter which is now used with\nkeystoneauth. Deprecates ironic_url parameter.\n\nChange-Id: I545f467d50b5eebbae26a6fd9e9fb7b35f8b84f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/52386b317a4e50de170b3b52dc5436f90d21149e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/52386b317a4e50de170b3b52dc5436f90d21149e"}]},"branch":"refs/heads/master"},"09429d0ebc6bff6aed5808faf7c4430e72d2dafb":{"kind":"REWORK","_number":2,"created":"2021-06-28 05:40:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/796359/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/59/796359/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/59/796359/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/59/796359/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/59/796359/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/59/796359/2"}}},"commit":{"parents":[{"commit":"ea79d50ef5df27457c78220a0594fad2d5848f9b","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ea79d50ef5df27457c78220a0594fad2d5848f9b"}]}],"author":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-06-15 03:14:54.000000000","tz":330},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-28 05:40:25.000000000","tz":540},"subject":"Update networking-baremetal parameters","message":"Update networking-baremetal parameters\n\nAuth parameters in networking-baremetal has been updated by [1].\nThis change updates parameters in puppet-neutron according to the\nchange.\n\n[1] a4e75f839605f6e2aba6b5f658004e9ffabc7156\nChange-Id: I545f467d50b5eebbae26a6fd9e9fb7b35f8b84f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/09429d0ebc6bff6aed5808faf7c4430e72d2dafb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/09429d0ebc6bff6aed5808faf7c4430e72d2dafb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
