)]}'
{"id":"openstack%2Fpuppet-neutron~798223","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fstein~I2b443a6113e94b25c5cbc1f70431dada89ba69af","project":"openstack/puppet-neutron","branch":"stable/stein","topic":"stein-tripleo","hashtags":["puppet-tripleo-job"],"change_id":"I2b443a6113e94b25c5cbc1f70431dada89ba69af","subject":"Stein only: Remove tripleo job","status":"MERGED","created":"2021-06-27 14:53:19.000000000","updated":"2021-06-28 17:01:12.000000000","submitted":"2021-06-28 17:01:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798223-stein-tripleo","meta_rev_id":"5c20066d8b76a82aca702844093142f74198ca65","_number":798223,"virtual_id_number":798223,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-28 17:01:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2021-06-28 15:51:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2021-06-28 08:33:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2021-06-28 15:51:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-27 16:01:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-28 08:33:19.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-06-28 15:51:39.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"}],"messages":[{"id":"859a5cb3980b66527ec669c7b7cbc08e47471296","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-27 14:53:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"229ebd6945cef064c43f783daad369aad36a6de8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-27 16:01:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/047cf52f444e49de84d7a38d551a5d25 : SUCCESS in 7m 00s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d86604aa769b4d8daeef3ff2846e607b : SUCCESS in 6m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae77eb7520514307bfd88cc8fa9b0ba8 : SUCCESS in 6m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7607b2f6668f4d148682eb521a3391b5 : SUCCESS in 7m 03s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/fc39d26bee8f44869a5d3bd5dbcbe960 : SUCCESS in 34m 06s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/6f8279af07ec470eb1efffee85994376 : SUCCESS in 25m 12s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5345f27a1bfc46c5b3e506c89973cbb2 : SUCCESS in 14m 54s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9bc0fe0636544f0b91a3e8c136b5a129 : FAILURE in 16m 47s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/a0f8983147f342e0a6685fc1634cf309 : SUCCESS in 51m 32s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2b30985ee3bf4659b59ee5f05dae4cda : SUCCESS in 1h 01m 31s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/bdeda48efafe408da8d18b22b94a5645 : SUCCESS in 1h 02m 20s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/167e1257c6b648d2aadadc2f19c04f91 : SUCCESS in 44m 10s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7fe5d1cf1ada4535b8ab1501f6053f4c : FAILURE in 24m 01s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/239c675545ff47849c53fb04072b2b4c : FAILURE in 29m 32s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba92e46749874923ac04a9da9dd62c88 : FAILURE in 22m 09s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/9a4c3ebb76ab4fb8bc3358c185c18e13 : FAILURE in 22m 58s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/051d4560c1cf4265b9f31cd3d73e413f : SUCCESS in 20m 36s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/895985fc33a44fcfb625fd1f7fc20bb1 : FAILURE in 10m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b19729773439ce5e12ee78af84479fd15ada2fea","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-28 01:11:46.000000000","message":"Hashtag added: puppet-tripleo-job","accounts_in_message":[],"_revision_number":1},{"id":"e99d270a273356cc2bde431cab5f9dc59522b21d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-06-28 08:33:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1a781d96954daa6ed4e8fa2bd12e5624f34305f5","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-06-28 15:51:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fa62db885a769359f27151eb15d47ec891492836","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 15:52:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13fa0cf2cbed2e2fb14fbea3387ac12e48f0090b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 17:01:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/438c5672b3f243a39f87d5224dbb15e4 : SUCCESS in 6m 14s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed130ae589ad4eef9cea1f39688d7881 : SUCCESS in 6m 41s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/040e70262ce64ddf90af5edd59057aae : SUCCESS in 6m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0b5cb6304a44386b11d56c8cba1a6b2 : SUCCESS in 6m 49s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/764d71468c2446889c47e18416a1a287 : SUCCESS in 24m 15s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2fe01191e46440188ec6b2f42c5255d8 : SUCCESS in 33m 21s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4019751f1a04011b862c693a5b9166f : SUCCESS in 16m 00s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/69a2572042f4423a9a7ec3dc70f51ec4 : SUCCESS in 51m 17s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9af88017de104fccb3306f24098467af : SUCCESS in 55m 55s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/73d9d191a2ea4cfbb27d91983e4ea016 : SUCCESS in 1h 03m 51s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ab77569f33cc40de83a71732e134f91c : SUCCESS in 19m 36s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21b0c4157c7e4445bad81d655f3fc364 : FAILURE in 9m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5c20066d8b76a82aca702844093142f74198ca65","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 17:01:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"471738d1c6230c67726311f33505dd8c072888f7","revisions":{"471738d1c6230c67726311f33505dd8c072888f7":{"kind":"REWORK","_number":1,"created":"2021-06-27 14:53:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/798223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/23/798223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/23/798223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/23/798223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/23/798223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/23/798223/1"}}},"commit":{"parents":[{"commit":"4d5fd04844b58c84be7428a5bfcaab1b2f4d7eab","subject":"Add ovs/igmp_snooping_enable support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4d5fd04844b58c84be7428a5bfcaab1b2f4d7eab"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-27 14:53:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-27 14:53:10.000000000","tz":540},"subject":"Stein only: Remove tripleo job","message":"Stein only: Remove tripleo job\n\n... because TripleO is moving its stable/stein to EOL.\n\nChange-Id: I2b443a6113e94b25c5cbc1f70431dada89ba69af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/471738d1c6230c67726311f33505dd8c072888f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/471738d1c6230c67726311f33505dd8c072888f7"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
