)]}'
{"id":"openstack%2Fpuppet-neutron~801933","triplet_id":"openstack%2Fpuppet-neutron~master~I6513eb610a9dfd64f30f4d5cb465cb8b0ff2d897","project":"openstack/puppet-neutron","branch":"master","topic":"bug/1937333","hashtags":[],"change_id":"I6513eb610a9dfd64f30f4d5cb465cb8b0ff2d897","subject":"Fix wrong method name","status":"MERGED","created":"2021-07-23 05:53:08.000000000","updated":"2021-07-23 16:01:40.000000000","submitted":"2021-07-23 16:01:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801933-bug/1937333","meta_rev_id":"f2db325ac5362172b2a6e597c28e0d2959ca95ed","_number":801933,"virtual_id_number":801933,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-23 16:01:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2021-07-23 05:56:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2021-07-23 05:57:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2021-07-23 05:57:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-23 05:56:14.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-07-23 07:54:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-23 08:05:46.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"CC"}],"messages":[{"id":"5cbf93e3117d0127c549f1a9c71d8fa6b9157e22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-23 05:53:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60ef6ee265f4ea81504357d18abe76792a016c3d","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-23 05:56:14.000000000","message":"Patch Set 1: Code-Review+2\n\nthanks takashi","accounts_in_message":[],"_revision_number":1},{"id":"75c831432b30635d3c65a0bd5b5694e20027d0d9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-23 05:57:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThis is blocking TripleO gate I\u0027d do fast approving.","accounts_in_message":[],"_revision_number":1},{"id":"352eb6b084ab3310d4f78bc932fd5e34efa3282e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 07:54:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1da7271d18f437d8678ccaa8ba6f8f9 : SUCCESS in 4m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da676a9de9074cbdb100a1b27369a6e9 : SUCCESS in 4m 06s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/753d94c509f54cf886a10bf50a2afdcd : SUCCESS in 4m 08s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0ee2ad04f9394a85a8158a5705366110 : SUCCESS in 21m 09s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed2ec03af79f43359d21e107ffbeef2c : SUCCESS in 19m 41s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d806a25f3d5147a3a10c5e932edf04fd : SUCCESS in 17m 48s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/093cbb52b8a54ced90aa87b0605f28ed : SUCCESS in 16m 11s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f48bc3d1366425fa295a718e54c2fa8 : SUCCESS in 13m 37s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cc6fb19571c43cfaeb220ee123c7db4 : FAILURE in 44m 43s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/680a3e85bd5d4bb9b671882bdfd6b051 : FAILURE in 39m 57s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92c1a38b61c0492093cd511ba07d8120 : FAILURE in 30m 29s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1796bbcd62f24df6b7ae96226286ecd9 : FAILURE in 34m 28s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/39b83c14b5e044e699bbad3ba6449c31 : SUCCESS in 32m 42s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e56267b5938c48e39db9a4b6208998a0 : SUCCESS in 1h 13m 12s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fa2b582862a64d5fa1ae3a5886c009b9 : SUCCESS in 1h 36m 53s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e402c65e59884f769f76f7ac668e8a4c : SUCCESS in 1h 35m 22s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/28dc5d0f87a8467590c3165776be3be5 : SUCCESS in 1h 27m 21s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/041a71de0e194029b768657d9f347b14 : SUCCESS in 27m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/321dc9fadc9f4884b8d9dfb02e29f664 : SUCCESS in 1h 30m 50s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ffd8d26db3b44bbeb4f6899d226c749f : SUCCESS in 1h 28m 45s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b843d5b314e4a7a83fc090dcd03cda6 : FAILURE in 58m 36s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e04273943703480cba17335418df17d0 : SUCCESS in 1h 11m 07s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/be6e49d1e6b74c899d134d965d8b09d6 : SUCCESS in 25m 08s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1bc3858b7684afe9aa57716409e93d6 : FAILURE in 19m 10s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/7aff36d148114a9aaa6107ad35b7cf12 : SUCCESS in 24m 16s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fb6a361680a4ae99f603f0f92a5bdd5 : FAILURE in 20m 45s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d957477ea06741d6beeb1040b6771bc6 : FAILURE in 20m 44s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/254169c1823942f3b7558f11e14d3747 : SUCCESS in 1h 57m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3e8b9ee1188bb73548d262ed19a3a62a80373a9a","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-07-23 08:05:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f931b74526df7dc6d629cc7c2819719bae11bdee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 10:02:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71393ade26744fc2950419b05366d6a0 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64db394b6c5c4aadb0980ca075374148 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/611d39418bcd427fb888611f171e6793 : SUCCESS in 5m 02s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/dc80733990664b109ec5eef01e0eab0c : SUCCESS in 16m 07s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/9f320020e5824c27aee682c1c83bd352 : SUCCESS in 16m 53s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4546593488d345348242033459b0baf2 : POST_FAILURE in 19m 56s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5da3e93e3ea34d4bb554aaa81a8c39e7 : SUCCESS in 14m 44s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f70d9a6ea90b4a679dca3357982e8954 : SUCCESS in 16m 09s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e9bc46ea905425f90956344d6362bce : FAILURE in 52m 14s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/583891a8b1a84071b40c391c483e6880 : FAILURE in 39m 13s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1861793f3c246fbab281206fb083809 : FAILURE in 28m 40s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/798f4539f20f4144961b6e9a108071f6 : FAILURE in 27m 44s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9cf826bb09564a8fa45b3cad22fde6d6 : SUCCESS in 26m 19s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ebf4662169848e1bb7c66fd28dd1b6d : SUCCESS in 1h 04m 41s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d26f266897cb4e289e84c798e5d389d0 : SUCCESS in 1h 11m 02s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/91c9b4e241104223b3300e1a29680adf : SUCCESS in 1h 14m 05s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/145581bc87034c429011a49a251cc2ec : SUCCESS in 1h 14m 33s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/83ac852f845e440c8cebc84521a39d43 : SUCCESS in 27m 38s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b536cd9b2fc74d35b0f534b0c6c55ee8 : SUCCESS in 1h 12m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e143861554c342cf97e05007f7df131e : SUCCESS in 1h 18m 41s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/94d89b85a791490b884cd5d9f5ed6746 : SUCCESS in 1h 14m 40s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf4b34caf4d1474699e0680fbf53f43f : SUCCESS in 1h 04m 59s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f7d7abc63794ba3b9608e9b4bdc0135 : SUCCESS in 26m 17s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e131017544c04b30ba108ab39c664806 : FAILURE in 19m 44s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/03dcb6f8bd5a4b92abefd6ffbbeaef5d : SUCCESS in 26m 25s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7b223acb1c942599218c1329213bf5a : FAILURE in 18m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa3a5977ad1f47c498476eb3b83d819b : FAILURE in 21m 02s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/da0c63eac5484db78469fd1a2211c891 : SUCCESS in 1h 49m 22s","accounts_in_message":[],"_revision_number":1},{"id":"5a0eef40cd55034a153012c35c10fbfd49ab7912","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-23 10:19:08.000000000","message":"Patch Set 1:\n\nrecheck\n\n https://zuul.opendev.org/t/openstack/build/4546593488d345348242033459b0baf2\n\nfatal: unable to access \u0027https://github.com/nanliu/puppet-staging/\u0027: Failed to connect to github.com port 443: Connection timed out","accounts_in_message":[],"_revision_number":1},{"id":"a1b25b7c2c39db0ce01436ad39d9532869245588","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 12:23:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97c25e5648964f39b25656aaf9012c47 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e2fa24dc31b4b23ac4f4866cb67c61a : SUCCESS in 3m 55s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba44e1a369b3422ea9f28a7ad353aec3 : SUCCESS in 4m 34s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/811d8ab8194d4bc1834e25d7444c3aee : SUCCESS in 18m 55s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/6e2c2a8945e74260ad044f8abc91e2bd : SUCCESS in 19m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/021a6afd3afb4f64aa1d0861379ca28a : SUCCESS in 14m 48s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb3b5d7654414bf2a4973b905e985296 : SUCCESS in 15m 46s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2e9a97cab1843b4bed353b99331c2f0 : SUCCESS in 14m 19s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31f220e0f0fc448fb94f408425d7f703 : FAILURE in 35m 09s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc79c68840dd4ced94c70970a9581e53 : FAILURE in 42m 50s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d67f716b69ff4fdfa117be8969eb2042 : FAILURE in 33m 03s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51d23102db5c4e7ba2e0c8c11099cacc : FAILURE in 30m 02s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f799abad41d04363a83ca233da20861c : FAILURE in 31m 48s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/833e7dc5c5f644e9947636bac37969a1 : FAILURE in 39m 39s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2f0eb831945b4fca9cc6fa5e60198a44 : SUCCESS in 1h 17m 06s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c170b6e21404ebe85d828d5e5422ca8 : SUCCESS in 1h 17m 47s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b90d0189e8d148118c76bd2da9ec0366 : SUCCESS in 1h 22m 34s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb0399bc7a554dcfaec389b0430a28ee : SUCCESS in 21m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/48dcc3bff84146eda6742760c7a1b9b8 : SUCCESS in 1h 12m 02s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2cd40534e17a420d8bfd64ff6c1b7812 : SUCCESS in 1h 17m 25s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9af639531cf5485984ceab34d6c256db : SUCCESS in 1h 20m 20s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/10fcd6628a4940d9bcdc2f3f588d210a : SUCCESS in 1h 06m 35s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f938f61040d4e51bf81fe1b96f8c0b5 : SUCCESS in 20m 51s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fac6e09c9e2407e849044b0b3c1f2ed : FAILURE in 18m 20s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a7460d2fd85438eb0da20a01a8757b9 : SUCCESS in 25m 23s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fdb2f9a75e3454593e90ccef46224b3 : FAILURE in 17m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ff599c6fc1f43a980d2bbce31fb5b0c : FAILURE in 17m 43s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4e1248015f094acabe445124f031e741 : SUCCESS in 1h 58m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e31aa841c52c90f82df8c31f3bd809ad43c71b6e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-23 12:25:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"74be92a52532902c64e34f894edf64949952000e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 14:02:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1614d9ec23f401184e98f1c92cb6224 : SUCCESS in 6m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92f50da42619497b91ed59190e4f172b : SUCCESS in 4m 27s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7174d784d2fc4a91abbbf0f63d663045 : SUCCESS in 4m 18s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3b2fed2b1d6a4f0eb8f9606472b1377f : SUCCESS in 17m 16s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/9070a53b3d8747b88cd6814bc9884e39 : SUCCESS in 19m 09s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f7edef7bce24f7da7f160ee5778ea35 : SUCCESS in 13m 57s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/462cfa6d664343eb957a0cb4e1089709 : SUCCESS in 16m 00s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc4f91baef404638a0c2a4b7f5c2ac3b : SUCCESS in 13m 56s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78cc45813def40d5b06ac6e03cc87b78 : FAILURE in 43m 38s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b9314aae61b47249eb01e3bc5513368 : FAILURE in 39m 17s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be44977f762b44339cf03644887db94e : FAILURE in 32m 27s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ca00d1270bb41b0a643a7e5934b9799 : FAILURE in 29m 50s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e36197b586f4fd2bb59e6f751713c71 : SUCCESS in 26m 39s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/159a91ddd44449e9bf514b2c05cadefb : SUCCESS in 59m 03s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/52aac551b74a4edd8f6ae597cd70fa8b : SUCCESS in 1h 21m 54s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0a8676cc5f146ab83290783cdeffa7c : SUCCESS in 1h 15m 05s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc58ea9440104a3c9ad80f90fae7ba7a : SUCCESS in 1h 05m 24s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6ae457b775314cc985c08dfd975584e7 : SUCCESS in 22m 16s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2a5083856964f479a7181855a7cbc98 : SUCCESS in 1h 10m 31s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bdbb6cbd4be640dbb8c936e8369a4d9e : SUCCESS in 1h 19m 28s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a73bb7688a6f412c8d9a2a64d46b368c : SUCCESS in 1h 28m 19s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/eaee14a379914fbfaf4414cf8946d84c : SUCCESS in 1h 08m 27s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/70d1ae12ce0846439eb9607a45c95a34 : SUCCESS in 28m 13s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f30723f86da04978b7c5258080bbfcb8 : FAILURE in 18m 41s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/286c4755158a407f99c2d8e42fbd0841 : SUCCESS in 21m 23s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9bb99b21ff7540bd8926b1870b73e0f9 : FAILURE in 17m 18s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe2fc3ab861f45448ab979dac8eb6539 : FAILURE in 19m 37s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d70754c6d155498286fa81233f96e55d : SUCCESS in 1h 34m 40s","accounts_in_message":[],"_revision_number":1},{"id":"7c66a691a6474c7a47ed9713ca9e30c47293bb19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 14:03:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa9caf9428262225546b16ec973b75a4c30f496e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 16:01:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a59709ca662e4a0eaf5fa1b4df903943 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6214d6c2480c4e03baf86dbaa6100e0c : SUCCESS in 4m 06s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea38e7668d7342aeae24cb0e61581d10 : SUCCESS in 4m 02s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3bdf3b8b572446e1b43823eb1d148a18 : SUCCESS in 20m 39s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/46744e803a094bd99b45e54dd3f8ce67 : SUCCESS in 19m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/112e7bbb090e4e10b32bf7d06790d836 : SUCCESS in 16m 29s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbc85257ce684686b0c3c84c00fd17de : SUCCESS in 14m 28s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/735eb8dfa5f9443fb2136220d7bdd102 : SUCCESS in 1h 22m 12s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/55a1d6fc925b45c688db2482195fac73 : SUCCESS in 1h 53m 38s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/26725369f694412aabef88bd1fc9d682 : SUCCESS in 1h 30m 46s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e192e33bff494d61a3255c31b3e8312c : SUCCESS in 1h 03m 49s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/99d4c99ac16b4215b26f47e700acf98d : SUCCESS in 24m 21s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfa58590b6e64afe94052ba72c3bd0ba : FAILURE in 17m 29s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8b66323a2f2e411bb797aae7e8d093a1 : SUCCESS in 1h 50m 43s","accounts_in_message":[],"_revision_number":1},{"id":"f2db325ac5362172b2a6e597c28e0d2959ca95ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 16:01:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ab03d57aafe7acbfffd515e0eb953e0dd0070773","revisions":{"ab03d57aafe7acbfffd515e0eb953e0dd0070773":{"kind":"REWORK","_number":1,"created":"2021-07-23 05:53:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/801933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/33/801933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/33/801933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/33/801933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/33/801933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/33/801933/1"}}},"commit":{"parents":[{"commit":"e54561145994f8b1e72e8d18cec24763ae9eda8c","subject":"Merge \"Update networking-baremetal parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e54561145994f8b1e72e8d18cec24763ae9eda8c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-23 05:51:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-23 05:52:56.000000000","tz":540},"subject":"Fix wrong method name","message":"Fix wrong method name\n\nThis change fixes the wrong method name introduced by commit\n09429d0ebc6bff6aed5808faf7c4430e72d2dafb .\n\nCloses-Bug: #1937333\nChange-Id: I6513eb610a9dfd64f30f4d5cb465cb8b0ff2d897\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ab03d57aafe7acbfffd515e0eb953e0dd0070773"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ab03d57aafe7acbfffd515e0eb953e0dd0070773"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
