)]}'
{"id":"openstack%2Fpuppet-neutron~806487","triplet_id":"openstack%2Fpuppet-neutron~master~I573d1f9abe201bdd17bf16602218252cc9b7f132","project":"openstack/puppet-neutron","branch":"master","topic":"report_interval","hashtags":[],"change_id":"I573d1f9abe201bdd17bf16602218252cc9b7f132","subject":"Support more agent specific report_interval","status":"MERGED","created":"2021-08-29 05:56:59.000000000","updated":"2021-09-20 09:09:39.000000000","submitted":"2021-09-20 09:07:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806487-report_interval","meta_rev_id":"03069210ad07c847ab5d4ed67c4a9d573d7820c4","_number":806487,"virtual_id_number":806487,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-20 09:07:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-09-02 16:01:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-09-20 06:07:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-09-20 06:07:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-29 08:17:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-02 16:01:16.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-09-20 06:07:12.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"f7fa229973dd1d429f3d2ff555e14ab09a1db766","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-29 05:56:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76b4415968d212c237ab84422430126c2d33f3d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 08:17:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b10307c84b774ea989a139b25aad0b30 : FAILURE in 4m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27642453c5ca4bf7a4a81381f77e149e : FAILURE in 3m 46s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d35ee3c9d6c461b94ae70f85f7bc1ef : FAILURE in 4m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/379d8f45e0e34cf892d8185640a28ca0 : FAILURE in 19m 55s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/195a31070bd44b93b214223d66cc964f : FAILURE in 20m 53s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b6e7892c4d544c889cbaf8f77769163 : FAILURE in 13m 15s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f38535f2a774c2fb25acbe93ea263c1 : FAILURE in 16m 27s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31e727c15e5147f5a749997b03ee1e8a : FAILURE in 14m 13s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf235dd23881467c883dc08afff3468b : FAILURE in 11m 28s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6dfb94ad43b41bbadcb39183a578dba : FAILURE in 13m 41s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ecf58d236e244b1b7f2f274d878f91c : FAILURE in 10m 52s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e38d8d7e4514bcdbf08a7487e14b57b : FAILURE in 8m 50s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/758abce3f9ef4739becab7325bd72c16 : SUCCESS in 31m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/09c17eaa823d4c66bc1c49550cf0085d : FAILURE in 11m 19s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/37dd64ca3cb94b01ac85104a651e0325 : FAILURE in 11m 26s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1458cc53c1f84391807775b67fd3b70d : FAILURE in 15m 15s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/81a0b1a86e614500a5ebd7ad286ec7d3 : FAILURE in 11m 13s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17e4ab851b6e458f881d043088506696 : FAILURE in 8m 26s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a92b1647833b4d34b82f784962c9842a : FAILURE in 14m 32s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/164adcff9cf34924ab03b08406ad4998 : FAILURE in 8m 48s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c2fb42007bc4e30bb2a63cf326c3d56 : FAILURE in 11m 30s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/da96763412af456d8c2fd3a611a7595e : SUCCESS in 2h 18m 32s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a49311b497254904b91e20896fd5562d : FAILURE in 37m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71b3ad7b75af4e39824c554304c590f7 : SUCCESS in 3m 48s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ee8e904246494e5794affdc30cacf95a : FAILURE in 1h 00m 46s","accounts_in_message":[],"_revision_number":1},{"id":"9cacae384c85a8a45d1f206215283362954a4531","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-29 08:34:13.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b80275b92b9dfe0ed8cd541df53046174ad509e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-29 08:34:59.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"bad5aaa87e34dfcc791d86fa5c187702383a40cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 11:14:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56ce2ce22a0a40778cd8f386f5cb2f3f : FAILURE in 4m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a24a31b509f8470089d3f4de9d88bec1 : SUCCESS in 3m 52s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/838632f04836492b8bb2c2e228c8daeb : SUCCESS in 4m 32s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/22ae9a70e1be4a8faf24a498db241f85 : FAILURE in 17m 37s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/70795c3e6b0241c9ac9828c2bed5da32 : FAILURE in 17m 43s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4419988f68648fb9917723bd2813525 : FAILURE in 13m 39s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9811f6441f7c4b94b7a0cf70adc1b238 : FAILURE in 14m 46s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7797b99fa4284fb8be6e52ef64c9ae86 : FAILURE in 14m 15s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa017965839249d08bc47f11ec75c5fc : SUCCESS in 55m 12s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8dc266cc66644a619de57d253f930a5d : SUCCESS in 51m 49s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b996f53070ed4ee691999196302389eb : FAILURE in 28m 43s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38cef13f90914fe29ae6921c1bae6b55 : SUCCESS in 38m 57s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/43c7b4e27c0e431ebd8858812368c290 : SUCCESS in 25m 58s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0bfc91734e634998bbf20fb00ecb1252 : SUCCESS in 1h 12m 02s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2dd9b9e3b3e44e38695ce66356903c8 : SUCCESS in 1h 26m 11s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/00914571b1e446608828d68e9509407b : SUCCESS in 1h 16m 27s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7fbd3b41c1014a968b7adb53963ca95d : SUCCESS in 1h 02m 18s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e460a5e16ecc4601a673b9d62aa2aacb : SUCCESS in 16m 51s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5e50b675fc74885a5ceba04f2450b3d : SUCCESS in 24m 10s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ff0c5253aa0420ab1bae0b9e0a4108b : SUCCESS in 15m 49s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4f74e700ac3484e9b48e3f1dc90739a : SUCCESS in 16m 12s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f43b69001d8f4458ae6136733edcb899 : SUCCESS in 2h 33m 11s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e845462919a04d3ba336230abead4297 : SUCCESS in 51m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e646a4d854ca44a69fb55c8ff94eb2be : SUCCESS in 4m 16s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/49ed6ac51f3149e5a8801c12d4cd4c47 : FAILURE in 1h 05m 50s","accounts_in_message":[],"_revision_number":3},{"id":"57903d3d3cf1361c279e91312f3f9b42e4a90b59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-29 13:45:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"52582627619af9aadd1824bd6c571fd1ed406b92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 16:06:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bffe79a3922048feb2610b97ad86e7e8 : SUCCESS in 4m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b22f149519bb4497b1353ddfb6fd39ff : SUCCESS in 5m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a17a363106ec47dfa8bf407b28f0e5b9 : SUCCESS in 4m 02s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6cde308697de4b17acfec9e25d8ba1fd : FAILURE in 18m 58s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/27aea6e0b3824917973e3155a22bd2f4 : FAILURE in 15m 15s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4efc76ed104b435aa8fe23798415ea55 : FAILURE in 12m 06s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95842e5568f64c599f36178d4ed26fa6 : FAILURE in 17m 01s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/437efc8fb214496bb9358167c71e88f9 : FAILURE in 13m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56d6ddf54b984522bfdd5d0a97b04326 : SUCCESS in 44m 47s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/400844d57c024835ba6addf998e14eb1 : SUCCESS in 53m 51s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3d57b7d4b274d8ab99f398ea1df16f7 : FAILURE in 30m 03s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ebd1ae8ff7c4c04b1469f9b4f39e887 : SUCCESS in 36m 01s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ecacdfd856fe4bda91882673494bdc26 : SUCCESS in 21m 45s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9b7a74aa2d2c46bebca16a49cac38419 : SUCCESS in 1h 08m 26s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d7cd34bc07a643e39f7de9ae5330babf : SUCCESS in 1h 14m 55s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/296852dedc96442fa8909192094dc447 : SUCCESS in 1h 15m 39s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/11110f2a21744f8f88cc5b3ab70be5a7 : SUCCESS in 58m 45s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9aa3f0d80a7476b98513b01a257184b : SUCCESS in 17m 52s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1fcb4cafca75438abe27fee461da4f4d : SUCCESS in 21m 03s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99e9f77cc484436495df2e948cc166cd : SUCCESS in 19m 37s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82483ffeffc540b0a4750df380db36c2 : SUCCESS in 17m 45s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d7a006f8cd374b849494d92a9c337554 : SUCCESS in 2h 17m 51s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ef57529754c74ae0a0abee703241bfd1 : SUCCESS in 42m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9c23151b406481a8428f07888b9b91c : SUCCESS in 4m 26s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3b3878d8401146609e8a68a037e9d3f5 : SUCCESS in 1h 04m 49s","accounts_in_message":[],"_revision_number":4},{"id":"427eae53590d149ee8aea69a62c6fb65e6a81570","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-30 10:12:58.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"bc0e9d4de8135f38604b6590129f5c53304ee8c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 12:38:58.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25e6b6a5f6cc45018d5be4764dbfb2a0 : SUCCESS in 4m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b58a0e6970b4182a186f990f2f18a41 : SUCCESS in 5m 40s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9fcca29e89c4ae9a56de10b9539b1f5 : SUCCESS in 5m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3177f417fd2c45849eaedb6e97733be2 : FAILURE in 18m 52s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/871b8cfd6c854adebf1bf24feaf5f53a : FAILURE in 18m 34s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/001a6bb3ff4741adb28e1d5cf4a4295b : FAILURE in 17m 42s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ec94f234d7f4524be90c59127761201 : FAILURE in 15m 48s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ae07b646e564d0380af291b038cb8b2 : FAILURE in 15m 16s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c7c6f4a80fc43cb87d965399c92c7e6 : SUCCESS in 52m 56s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a2281d056294dbabffbaec697d08f6d : SUCCESS in 58m 14s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ad937e2793d4465ba8a73706a804bdd : FAILURE in 36m 21s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11e2cffb57bb4f5faf296a664740e223 : SUCCESS in 43m 14s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a237e51f954640d1af8da3132288cc41 : SUCCESS in 31m 24s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8931f23b5a0941c2b760ba38535be625 : SUCCESS in 1h 21m 10s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/365ce61877dc4545a6782d1a75ede333 : SUCCESS in 1h 19m 31s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e096c3f0b4a40a5bf434b4294e1cd8c : SUCCESS in 1h 20m 55s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/829f5c5c4f6648e6931845ab4059e7ae : SUCCESS in 59m 24s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a643d69671f40589ba3cea31229c50d : SUCCESS in 28m 09s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/e3e89ba9aaf1408e9d22df61fdc4ae5f : SUCCESS in 33m 14s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a9ae7f971c84667bae249e7d1f8867c : SUCCESS in 20m 42s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2325ba9340ca4b7890854069f89648c9 : SUCCESS in 18m 34s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/940d7f8ee2da426e80c2060a05ebb0b1 : SUCCESS in 2h 22m 24s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a9fb491b98c94df195df198fad844442 : SUCCESS in 45m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1838cd58c3a45e8beacaf70fb6f4f7e : SUCCESS in 5m 42s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1baca8f4a2cd4ebfa0d8af8c36704257 : SUCCESS in 1h 03m 58s","accounts_in_message":[],"_revision_number":4},{"id":"303928f82155089aa516220adc842e40b217a00c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-30 12:44:08.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"12360729f65d8a72ba617e2677fa8c65ac405714","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 15:09:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ba5f631851b4f1699c0a2d736ee07a4 : SUCCESS in 4m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/711dbfcca4924d76a71ea7c968fe5954 : SUCCESS in 4m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03d0f85beb7f4138aa86f87ebc9985fa : SUCCESS in 4m 57s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/469e9b6dccfe4af98b1c15618909e8a4 : SUCCESS in 17m 37s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/968b37ecb6344dd79e42bf5fb1690208 : SUCCESS in 18m 13s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc1685c132c24824a5431c0236713ba1 : SUCCESS in 13m 21s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dd0bb8358de4093bd1993b0a97941fb : SUCCESS in 14m 11s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bb7c4073881423087e15243680ca912 : SUCCESS in 15m 12s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/469b47bf08fd45af90fd858db592371f : SUCCESS in 54m 03s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7a2f4d0df0449548b587915e45744ba : SUCCESS in 56m 17s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8feb4da391e4e3c8c1d200d01c43e1d : FAILURE in 27m 24s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6e70c92824540b993ffe65b7174e665 : SUCCESS in 35m 39s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/81fa500801d74b1ca0a03de80b4bd15b : SUCCESS in 23m 05s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/702804b4f10e4c3985cf342e4a01c0c3 : SUCCESS in 1h 16m 24s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cea248fa960a447fb0759d4516c637ae : SUCCESS in 1h 27m 57s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b3c947dfa93f428282cdb47b42ffab71 : SUCCESS in 1h 16m 28s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/31b0ba33585d485aa74a841b96e202e2 : SUCCESS in 1h 00m 12s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce9af1d563334862aaa2ba8dc662b773 : SUCCESS in 16m 08s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb3aa57c6e164fd28a5a1efe481f4a46 : SUCCESS in 21m 05s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2a562f755b54d22a8a0551e1aa7b73c : SUCCESS in 17m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0891e47595674f21bd24abd2ac81255b : SUCCESS in 16m 52s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/496caa4672f642528cf1e047cabbaa2a : SUCCESS in 2h 23m 12s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a9711bf9d5d741b586d52dbf9a25d81a : SUCCESS in 45m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4aee53d1fe774411b87088cd9ceeb73b : SUCCESS in 4m 00s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2a8de9fb8ded46e09f5b87a0bcf00784 : SUCCESS in 1h 06m 31s","accounts_in_message":[],"_revision_number":5},{"id":"1e870d1fae22e1e8d43a6edc137c55bbd614b295","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-02 16:01:16.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"dfdb83a4da0603732514f0bbafcfda4b949307ec","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-09-20 06:07:12.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f43bdd29cc6d28eb1126a4a8316782e6aa8fbac9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 06:14:52.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a790acc29169f07489ccd5fac2c8f7997a677195","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 09:07:52.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9047398a5a3b4780a4f0156a25b7ce43 : SUCCESS in 4m 32s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd1b9b91890d4500916a767865b47108 : SUCCESS in 4m 33s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a33331bdcb924ad599440c1dee2466fd : SUCCESS in 4m 10s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/42629fa6166942d4a9d827be3b4c5667 : SUCCESS in 20m 13s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/9483f96e139a4ed78a7708f71f7aa92e : SUCCESS in 17m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4eb9d4aa701a4865b19db08441533adb : SUCCESS in 21m 50s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a68b234aa56944f4b410b3d80c080939 : SUCCESS in 16m 41s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a124d47e75ee4ed88d1f12922fdc69b8 : SUCCESS in 1h 11m 18s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/da32187ab76b4acaa01b7207ed42f1c3 : SUCCESS in 1h 41m 59s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a0efdec3c3c48a095925e7fbee0f819 : SUCCESS in 1h 20m 01s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5369996078314c5694a983dec9620d41 : SUCCESS in 1h 10m 35s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d01b2a123ec4bce8a37284b3413188e : SUCCESS in 26m 41s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/855809e98c704fbaadf556b32e062977 : SUCCESS in 20m 53s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/66965913bf55459aa0d22faf022a12cd : SUCCESS in 2h 46m 14s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/166721d01a184c039252f09e94e261bb : SUCCESS in 48m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bac97f56ffe4ddfbce9f0588b4b4dd9 : SUCCESS in 4m 43s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e65fc7ed1535472e9fe8b7254279a01a : SUCCESS in 1h 12m 04s","accounts_in_message":[],"_revision_number":5},{"id":"d87ae93dcbfd509c26421121809b6fe4af7a40f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 09:07:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"03069210ad07c847ab5d4ed67c4a9d573d7820c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 09:09:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5213d6fcd0da475b8977a40fa27bd638 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3307b54610f7da905f7e7c748c821e31308b9bbf","revisions":{"af8b4cfc99046ebd6feeb21cd337913557cbc579":{"kind":"REWORK","_number":1,"created":"2021-08-29 05:56:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/806487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/87/806487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/87/806487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/87/806487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/87/806487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/87/806487/1"}}},"commit":{"parents":[{"commit":"dae33092a120655dfe4b9a550ec29a8c2dba73a5","subject":"Add report_inteval parameter in ml2 baremetal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dae33092a120655dfe4b9a550ec29a8c2dba73a5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 05:55:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 05:55:32.000000000","tz":540},"subject":"Support more agent specific report_interval","message":"Support more agent specific report_interval\n\nThis change introduces the report_interval parameters to more agent\nclasses, so that users can override report interval of a specific\nagent more flexibly.\n\nChange-Id: I573d1f9abe201bdd17bf16602218252cc9b7f132\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/af8b4cfc99046ebd6feeb21cd337913557cbc579"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/af8b4cfc99046ebd6feeb21cd337913557cbc579"}]},"branch":"refs/heads/master"},"85f278f183abae957ff73c9cdcd4ff64a9f7bebb":{"kind":"REWORK","_number":2,"created":"2021-08-29 08:34:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/806487/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/87/806487/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/87/806487/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/87/806487/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/87/806487/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/87/806487/2"}}},"commit":{"parents":[{"commit":"dae33092a120655dfe4b9a550ec29a8c2dba73a5","subject":"Add report_inteval parameter in ml2 baremetal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dae33092a120655dfe4b9a550ec29a8c2dba73a5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 05:55:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 08:34:09.000000000","tz":0},"subject":"Support more agent specific report_interval","message":"Support more agent specific report_interval\n\nThis change introduces the report_interval parameters to more agent\nclasses, so that users can override report interval of a specific\nagent more flexibly.\n\nChange-Id: I573d1f9abe201bdd17bf16602218252cc9b7f132\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/85f278f183abae957ff73c9cdcd4ff64a9f7bebb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/85f278f183abae957ff73c9cdcd4ff64a9f7bebb"}]},"branch":"refs/heads/master"},"a406b5adb5f9ff559a778ff3b29904c5d9bf188b":{"kind":"REWORK","_number":3,"created":"2021-08-29 08:34:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/806487/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/87/806487/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/87/806487/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/87/806487/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/87/806487/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/87/806487/3"}}},"commit":{"parents":[{"commit":"dae33092a120655dfe4b9a550ec29a8c2dba73a5","subject":"Add report_inteval parameter in ml2 baremetal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dae33092a120655dfe4b9a550ec29a8c2dba73a5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 05:55:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 08:34:56.000000000","tz":0},"subject":"Support more agent specific report_interval","message":"Support more agent specific report_interval\n\nThis change introduces the report_interval parameters to more agent\nclasses, so that users can override report interval of a specific\nagent more flexibly.\n\nChange-Id: I573d1f9abe201bdd17bf16602218252cc9b7f132\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a406b5adb5f9ff559a778ff3b29904c5d9bf188b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a406b5adb5f9ff559a778ff3b29904c5d9bf188b"}]},"branch":"refs/heads/master"},"6ccd04f67b281b5fd794dbc82dcf658ebf5553df":{"kind":"REWORK","_number":4,"created":"2021-08-29 13:45:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/806487/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/87/806487/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/87/806487/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/87/806487/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/87/806487/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/87/806487/4"}}},"commit":{"parents":[{"commit":"dae33092a120655dfe4b9a550ec29a8c2dba73a5","subject":"Add report_inteval parameter in ml2 baremetal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dae33092a120655dfe4b9a550ec29a8c2dba73a5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 05:55:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 13:45:11.000000000","tz":540},"subject":"Support more agent specific report_interval","message":"Support more agent specific report_interval\n\nThis change introduces the report_interval parameters to more agent\nclasses, so that users can override report interval of a specific\nagent more flexibly.\n\nChange-Id: I573d1f9abe201bdd17bf16602218252cc9b7f132\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6ccd04f67b281b5fd794dbc82dcf658ebf5553df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6ccd04f67b281b5fd794dbc82dcf658ebf5553df"}]},"branch":"refs/heads/master"},"3307b54610f7da905f7e7c748c821e31308b9bbf":{"kind":"REWORK","_number":5,"created":"2021-08-30 12:44:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/806487/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/87/806487/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/87/806487/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/87/806487/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/87/806487/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/87/806487/5"}}},"commit":{"parents":[{"commit":"dae33092a120655dfe4b9a550ec29a8c2dba73a5","subject":"Add report_inteval parameter in ml2 baremetal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dae33092a120655dfe4b9a550ec29a8c2dba73a5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-29 05:55:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-30 12:44:04.000000000","tz":0},"subject":"Support more agent specific report_interval","message":"Support more agent specific report_interval\n\nThis change introduces the report_interval parameters to more agent\nclasses, so that users can override report interval of a specific\nagent more flexibly.\n\nChange-Id: I573d1f9abe201bdd17bf16602218252cc9b7f132\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3307b54610f7da905f7e7c748c821e31308b9bbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3307b54610f7da905f7e7c748c821e31308b9bbf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
