)]}'
{"id":"openstack%2Fpuppet-neutron~807445","triplet_id":"openstack%2Fpuppet-neutron~master~Ief0ba5da927484a4e691a1e1796b1e9e4879d8df","project":"openstack/puppet-neutron","branch":"master","topic":"remove_empty_pol_file","hashtags":[],"change_id":"Ief0ba5da927484a4e691a1e1796b1e9e4879d8df","subject":"Allow purging policy files","status":"MERGED","created":"2021-09-04 13:15:54.000000000","updated":"2021-09-10 15:08:40.000000000","submitted":"2021-09-10 14:35:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807445-remove_empty_pol_file","meta_rev_id":"c06b0e59477812a7936ed053dc2c7b4975e9af82","_number":807445,"virtual_id_number":807445,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-10 14:35:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-09-09 16:12:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-09-10 06:35:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-09-10 06:35:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-04 15:39:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-09 16:12:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-09-10 06:35:12.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"69407fedec822335447b6b5dce79c199b2bd9eaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-04 13:15:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bea369790ca13de16c0c15cfd90ef61b9a9fa530","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-04 15:39:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9101156a1c58487c81ce5fa3f24ce6f9 : SUCCESS in 4m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e04f24f5f2e448e6a7ad7a1136aa6227 : SUCCESS in 3m 57s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbfcfb41542344a692a836dc268f34d6 : SUCCESS in 4m 25s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/25e0bba061fd40fc814987e38ee8a087 : SUCCESS in 22m 05s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/e017a0d492c8424ca3578d7f27dd8dd9 : SUCCESS in 18m 51s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b83743ffa354e5fb66770fc7cfbf3a6 : SUCCESS in 16m 50s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c0d3ec44f8e458b9ac6f37cf3fbbfa1 : SUCCESS in 17m 16s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b6cc4acb88340ccabb6a075a628e54b : SUCCESS in 16m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c80c4933bae441da3dca94751699490 : SUCCESS in 54m 55s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee3bbd59aaad4445ac16509848fefc3d : SUCCESS in 57m 57s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29a22c79e71f4e46b5e071290acb7ce9 : FAILURE in 32m 23s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4769db44b21f451b9fbb11c713b1c096 : SUCCESS in 41m 29s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ab3a26afb19c4486a8f785db6e27037d : SUCCESS in 23m 05s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2aac18cd0e74b8ea488b9ef566cacbf : SUCCESS in 1h 20m 02s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3dc2f05963b248e287bc66cd5a182114 : SUCCESS in 1h 32m 45s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5eba388ab53a4dba840e7d8c94c5bca6 : SUCCESS in 1h 26m 00s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3803e2466d564380b87ec8d39dd00d21 : SUCCESS in 1h 10m 29s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a23b299406ca45f68a53543805843e7f : SUCCESS in 19m 19s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/4b2f073be3e7451f8934c4a2043f9c9f : SUCCESS in 24m 52s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/883a5dd90fd24f4493aee31cb6b0587c : SUCCESS in 18m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8ca7eb8e6984fa982a524ccd7ef51d9 : SUCCESS in 17m 28s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f23243a603614b13a54a5d0c3937f986 : SUCCESS in 2h 22m 11s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/0d5e066352bc48698366bd3fc0c5ee85 : SUCCESS in 44m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8077d71bfb9c41dbb9e2b45b214d32fb : SUCCESS in 4m 42s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e57d52874b52431898f72eb9d1d3463a : SUCCESS in 1h 06m 07s","accounts_in_message":[],"_revision_number":1},{"id":"8c1a8e0a0c7e54be6ab0de42f964804c523d05f5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-09 16:12:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ed4ebe617b92ac7155229fb0b775f92afe60e97f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-09-10 06:35:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a04283a1f9ec8125e5a9bf70ae5c707d285480c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 06:36:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"952028cf8f65701e55b8233136529a0d2eff9b42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 09:19:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cefbcdefbe744e2999203c321d0d88a9 : SUCCESS in 5m 23s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dc2c0ccad7640e5a42915bdc0afb3d5 : SUCCESS in 5m 55s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/637f4683e6b942649dc272ced2ab2a0e : SUCCESS in 4m 47s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e0dc6bfa6be24b67bb91535ef66125b8 : SUCCESS in 26m 09s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb8e718f726646919f8a49fd49f18039 : SUCCESS in 18m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/507f42f67e524d029a1bd2296180e958 : SUCCESS in 19m 29s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/933fa65ae5bd42b29d6d1b728e27b5da : SUCCESS in 17m 24s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b194f4852254c989010ddcd08958242 : SUCCESS in 1h 16m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2b5aa2283f124e8ea9a382a25c445f51 : SUCCESS in 1h 25m 55s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/19aaeb487965447498baa9efcdc8053e : FAILURE in 1h 02m 13s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f80166a84bc47ada77708235ac9ef4c : SUCCESS in 1h 08m 37s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/e8c83d43ba8d481fb715232f3d39479b : SUCCESS in 24m 21s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31ca3703c27e4e4a90178fc465527dae : SUCCESS in 20m 05s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/fc9686aa75f3449bb564362276a7e2cd : SUCCESS in 2h 38m 02s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f02d2c3bc71f40499e0542c646054ac4 : SUCCESS in 56m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69e155517e1f4cabbfda8d88c609f790 : SUCCESS in 5m 08s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7601d98103ef4250a32c48703f2031a6 : SUCCESS in 1h 08m 47s","accounts_in_message":[],"_revision_number":1},{"id":"4c3a10d238e764aa24379b7b51643e070eaaa194","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-10 09:21:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4e86a212c974fea4395a3b014f81a1037687896c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 11:49:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea66578593774475964a78c9b62c5546 : SUCCESS in 5m 03s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1ad9ed97f264d7696c305af7a4e04a5 : SUCCESS in 5m 56s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b92012d87dd3484d828ec2f423625e35 : SUCCESS in 4m 24s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ea34f96f9278417fb10878d5192fde12 : SUCCESS in 18m 44s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0227416a2274688b220291088ec53ff : SUCCESS in 18m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1749791a4eb649d1b12abe8acdb69128 : SUCCESS in 15m 06s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e42dd049b9544ac697a6beb647cedfad : SUCCESS in 15m 08s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a83737bbf6f4a438e42e5b58e0cacd0 : SUCCESS in 15m 00s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27af72373ae440489ce39a3b312a8a7d : SUCCESS in 49m 45s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c9dfa41184c41149a1c7ad59a51b229 : FAILURE in 52m 41s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f9ecad2dd884704a486eee9ea7991fd : FAILURE in 30m 47s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8e72719e1b8450e80b4afe2f4088f2d : SUCCESS in 40m 47s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9f316231a6b64f39913e0ee145fc0336 : SUCCESS in 27m 36s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4214c473e6dd40179cd2c4e300e05fc6 : SUCCESS in 58m 15s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd81de22b68c4a7fae339575e6b1ea28 : SUCCESS in 1h 20m 34s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/98e93050fdd0414f95efcb3a6b1e491c : SUCCESS in 1h 13m 37s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0e8a937d99c4f1ebd11a7d2e265dd32 : SUCCESS in 58m 39s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67cc640f591b47969dc8273dc2bbb767 : SUCCESS in 19m 51s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf8d169a6504418983af191d2976a592 : SUCCESS in 24m 48s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b24b9dfcca9407f9d2948153f5f9ec1 : SUCCESS in 19m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1429af5e79b04f8a80a62046c6dc76b5 : SUCCESS in 19m 47s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1d9f16ae439c4077b841103b8ae162ee : SUCCESS in 2h 26m 05s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/371b29e78f5c450baf205a4a31bfa61c : SUCCESS in 45m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6dcf52aef4694d93b91af5f41ebe8d1b : SUCCESS in 5m 24s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3295d3bc405649d68b4fbf81b52a719f : SUCCESS in 1h 08m 01s","accounts_in_message":[],"_revision_number":1},{"id":"cd5df295f21469f9c02ff5ffc930554f050cbec0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 11:50:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f4ba3be10f9ae1f24cd69a175f7966ab42754cf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 14:35:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d0c03f5703b42fcba4a1f9475530cd8 : SUCCESS in 4m 29s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/572dcbe922bc43acbbdd42a2319dfdbe : SUCCESS in 4m 34s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d0fc969a76f4626aa8f7db0befdfcf0 : SUCCESS in 4m 26s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/067cd56cde1a49c697e3cb76752a8165 : SUCCESS in 22m 28s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3fa9caa741e145bd85cbe3295d9fa5cf : SUCCESS in 16m 23s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66819beb45be4b208a4717c5d622ec78 : SUCCESS in 14m 44s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f9953497cea44cc825ca43b0f3ddf09 : SUCCESS in 15m 07s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3f4d064df796431f9c988f4f4aac5814 : SUCCESS in 1h 35m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c0b5fdcfb6444ba0b7c4d6f121da0837 : SUCCESS in 1h 18m 15s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8af2acd419d74a40b4f910f210c216dc : SUCCESS in 1h 16m 12s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/04f2942424c2485a9022e78db2d4d5de : SUCCESS in 1h 01m 52s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2435672f0bf54f038e1c4253b2e08c8f : SUCCESS in 24m 52s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09e2d565b1444a0db9987001f296b521 : SUCCESS in 17m 53s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/98520063ab77445f90089b3ab71b9630 : SUCCESS in 2h 41m 35s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/fef5fe1136c347739dc9d614a9a09948 : SUCCESS in 50m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3dae2e8a70544f27801b1aad44f5e1e1 : SUCCESS in 4m 51s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1a252c1b53634c6c80667f84d3f71ba3 : SUCCESS in 1h 31m 01s","accounts_in_message":[],"_revision_number":1},{"id":"c76c3f5e848caee6dfe831d714c5f1289727f170","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 14:35:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c06b0e59477812a7936ed053dc2c7b4975e9af82","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 15:08:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/178a50f9eb5341baa398de118d5ea38c : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b276901eed0594efb8845a3340b5aa33dc87053b","revisions":{"b276901eed0594efb8845a3340b5aa33dc87053b":{"kind":"REWORK","_number":1,"created":"2021-09-04 13:15:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/807445/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/45/807445/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/45/807445/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/45/807445/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/45/807445/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/45/807445/1"}}},"commit":{"parents":[{"commit":"dae33092a120655dfe4b9a550ec29a8c2dba73a5","subject":"Add report_inteval parameter in ml2 baremetal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dae33092a120655dfe4b9a550ec29a8c2dba73a5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-04 13:15:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-04 13:15:44.000000000","tz":540},"subject":"Allow purging policy files","message":"Allow purging policy files\n\nThis change introduces the new purge_config parameter to the policy\nclass so that any policy rules not managed by puppet manifests can be\ncleared.\n\nCo-Authored-By: Martin Schuppert \u003cmschuppert@redhat.com\u003e\nDepends-On: https://review.opendev.org/802305\nChange-Id: Ief0ba5da927484a4e691a1e1796b1e9e4879d8df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b276901eed0594efb8845a3340b5aa33dc87053b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b276901eed0594efb8845a3340b5aa33dc87053b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
