)]}'
{"id":"openstack%2Fpuppet-neutron~807740","triplet_id":"openstack%2Fpuppet-neutron~master~Ib9b255c5dd63bb7d7d9d7fd9347e35287940caaa","project":"openstack/puppet-neutron","branch":"master","topic":"quota-cleanup-xena","hashtags":[],"change_id":"Ib9b255c5dd63bb7d7d9d7fd9347e35287940caaa","subject":"Remove deprecated quota parameters","status":"MERGED","created":"2021-09-07 14:57:55.000000000","updated":"2021-09-10 10:27:59.000000000","submitted":"2021-09-10 10:26:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807740-quota-cleanup-xena","meta_rev_id":"85e879ca59314b5b12a79581930f9bac19849033","_number":807740,"virtual_id_number":807740,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-10 10:26:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-09-09 16:12:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-09-10 06:35:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-09-10 06:35:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-07 17:27:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-09 16:12:18.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-09-10 06:35:43.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"5b5753b4d4667015e52cd38f817a1357221f833c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-07 14:57:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2b444920288d7d94598400faf8f0efc9e140983","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 17:27:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edfa8e9e466347ca97b79679fb7180dd : SUCCESS in 6m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bfc9e35d1f54ffb9b86259b06cf2af2 : SUCCESS in 4m 44s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a6bcf5f6c75433bb3c153ac96832b99 : SUCCESS in 5m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7bb44c3e1a824454a2d400b4c19e4031 : SUCCESS in 19m 49s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9e8120e4ea34136929de1de1fde3619 : SUCCESS in 20m 41s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e66e55f8ab3461ab859c3e0332dd947 : SUCCESS in 14m 47s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38c1b598233642388a901558f896f46f : SUCCESS in 15m 09s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c791912157a4423cb79b50b41579caf8 : SUCCESS in 15m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4543f2e019f9498cbe85d2fb29937e22 : SUCCESS in 51m 10s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1df41eb5498a46559325e94d70e81953 : SUCCESS in 51m 49s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fa22031651f44dcb10b2a311d8674fd : FAILURE in 30m 25s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65ddbc66a5704a91a39d0bcd3a1212bd : SUCCESS in 38m 23s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/561bb6d992ef42028838803279b8a7c1 : SUCCESS in 23m 18s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ede7d3d92f9548b1a9ce1a9198432299 : SUCCESS in 1h 09m 44s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0dbfb2d695554e03bdc8af2a92870656 : SUCCESS in 1h 18m 07s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ab07fd6928e4e9fb8c143a57f609482 : SUCCESS in 1h 19m 34s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2986ddd1394f44fe9082a6570979c436 : SUCCESS in 1h 01m 53s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbe8227ede5a416c86cdacf05b88b16d : SUCCESS in 17m 50s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/d9051892b3ad4770b1d88738697dc708 : SUCCESS in 23m 04s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df7539585d4b4cf2b98df1349056512c : SUCCESS in 18m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b67d1d42dfbd4b7288bf8148d18206fc : SUCCESS in 19m 24s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e851daf09040409b9aa262a5f8a96c77 : SUCCESS in 2h 27m 36s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/80204dcb53214b988e7d301c021ca519 : SUCCESS in 48m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3ac1daf5afb45e090706fe0c60e686a : SUCCESS in 5m 28s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/627af4eaa17e4805a08006d98acd8f11 : SUCCESS in 1h 07m 34s","accounts_in_message":[],"_revision_number":1},{"id":"0f0246b8d0c90d166897fec934f5cef6f08848a0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-09 16:12:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"79143f650ac521a30d13f8d6ce8e2ed63b1b8404","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-09-10 06:35:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"052cc93e895737cf8487d73f45ece535df0b8cac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 06:36:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dff3853d836fcf0d04dc88391a1127cebc6aad57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 10:26:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c30383550f14c92837aa1aafd582309 : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67c33a12eb9a407b9150905059dd583e : SUCCESS in 4m 30s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e2732f46d2e4183a7cbe14360e31e19 : SUCCESS in 4m 36s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ed60b4675f14106b6b60fb4e32100bc : SUCCESS in 16m 35s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/255d8eca0f434515b278a38e2861bd83 : SUCCESS in 16m 55s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fdf459a76269431aa110905d42a88511 : SUCCESS in 15m 49s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/740d5b25ae1b4517b5e198d3a9039dd5 : SUCCESS in 16m 52s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca3f918a4dd041acb14216e23e3132d6 : SUCCESS in 1h 13m 57s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0029d5fbe31446fb85a7a9679fb5e369 : SUCCESS in 1h 32m 54s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/eb285c64233f4a5799f04db632eff30c : SUCCESS in 1h 19m 28s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/281037e5b1e343d98b895d9a474191bb : SUCCESS in 1h 04m 41s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc62c792d1f947a79699e1eb6f25e0f1 : SUCCESS in 24m 25s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a770433c38f44188b7cdb0c416282b06 : SUCCESS in 16m 43s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8ad0df933eed41a3b5deec43a6c3640d : SUCCESS in 2h 36m 26s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ebda460e8206463daa46fcf5f490f8a8 : SUCCESS in 56m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80543e15f63446ddbee7c0e14b518da7 : SUCCESS in 4m 50s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b05d2b286b294fa4ad86ed01f2e460c7 : SUCCESS in 1h 07m 58s","accounts_in_message":[],"_revision_number":1},{"id":"6d630643d70906a175132b76d5d85af8f4cfda77","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 10:26:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"85e879ca59314b5b12a79581930f9bac19849033","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 10:27:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7167633a7b324943b2da907ba9faab51 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"58f1d6a5255d06b9493af7efd389eabe07c08607","revisions":{"58f1d6a5255d06b9493af7efd389eabe07c08607":{"kind":"REWORK","_number":1,"created":"2021-09-07 14:57:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/807740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/40/807740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/40/807740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/40/807740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/40/807740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/40/807740/1"}}},"commit":{"parents":[{"commit":"dae33092a120655dfe4b9a550ec29a8c2dba73a5","subject":"Add report_inteval parameter in ml2 baremetal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dae33092a120655dfe4b9a550ec29a8c2dba73a5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-07 14:56:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-07 14:57:23.000000000","tz":540},"subject":"Remove deprecated quota parameters","message":"Remove deprecated quota parameters\n\nThe quota parameters for LBaaS[1] and FWaaS[2] were deprecated during\nthe previous cycle, so can be removed now.\n\n[1] 4ca9ac106243433eebca1492218d72718617d1d0\n[2] 1e2c71d36db1126387ae93dc4116313b0b998932\n\nChange-Id: Ib9b255c5dd63bb7d7d9d7fd9347e35287940caaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/58f1d6a5255d06b9493af7efd389eabe07c08607"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/58f1d6a5255d06b9493af7efd389eabe07c08607"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
