)]}'
{"id":"openstack%2Fpuppet-neutron~807744","triplet_id":"openstack%2Fpuppet-neutron~master~I5dcdffd06b7dab598bb62a5cfcbc7833472196c3","project":"openstack/puppet-neutron","branch":"master","topic":"quota-params","hashtags":[],"change_id":"I5dcdffd06b7dab598bb62a5cfcbc7833472196c3","subject":"Deprecate the unused quota_vip parameter","status":"MERGED","created":"2021-09-07 15:13:47.000000000","updated":"2021-09-10 20:33:46.000000000","submitted":"2021-09-10 20:32:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807744-quota-params","meta_rev_id":"09949e8b22a5b772691a145aa2644fe5a2858bbd","_number":807744,"virtual_id_number":807744,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-10 20:32:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-09-09 16:12:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-09-10 06:36:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-09-10 06:36:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-07 18:10:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-09 16:12:27.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-09-10 06:36:16.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"a00bdb5f4b58f5befc46d7f77bfb79f391a8061a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-07 15:13:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff5e26bb99156a353a5b2c1b3ec7fa5ae2f6a9e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-07 15:18:46.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e6d17ce61cd03e10f64e871cf9e648957d62fc44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 18:10:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6334c0544a914b7486cba0d3d320463f : SUCCESS in 5m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a68d0d474fc44a8b9dd2b88114586ca4 : SUCCESS in 5m 08s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d45025b6bfb743fe9f43b890b7f28979 : SUCCESS in 4m 57s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3cb381f49f1c47098e8b3602869af319 : SUCCESS in 19m 30s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/0cd5a840344c468e86e3991fafc28f2b : SUCCESS in 20m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a07b13f03aa44f73a2ba55c72b359063 : SUCCESS in 15m 15s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abc7f131a3db44e2b4ada151d6c67e6d : SUCCESS in 16m 40s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6eae2f915e1641f5a6669b8bcf18d446 : SUCCESS in 19m 48s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8a4353dacd3429e91c8916de80630cd : SUCCESS in 51m 44s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f0814a5743e46238ca5eef223c8e372 : SUCCESS in 53m 56s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea869731a4e84f9aaa3cc1ab4fd5f66b : FAILURE in 30m 37s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/378e64f8b32c4639b8fb8dbca8bec1fe : SUCCESS in 40m 05s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c22a6fb1214544e18448a6e8741792f7 : SUCCESS in 22m 52s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3b6df7ff38ab49fc8500c497dfeacb2d : SUCCESS in 1h 17m 04s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/35cead3bed704b77b668addd499b7ea9 : SUCCESS in 1h 40m 56s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ecac89f07b93403b9d097565a2d3b242 : SUCCESS in 1h 25m 02s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/09ed9dacb7504eb28dc20d66420ed3c7 : SUCCESS in 1h 50m 15s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e52a3c05d8648968e672c521599d99a : SUCCESS in 17m 00s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/59913566cabe4483bbc85fa92ea3c704 : SUCCESS in 33m 56s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae78608341c24ee090f4ef6c634d779c : SUCCESS in 18m 38s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0cbcd34f5159402f90c8b63599b56bd7 : SUCCESS in 29m 12s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f70030828c9b41438e6c959d164dd6d0 : SUCCESS in 2h 46m 56s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/5baf2618661d45a089f861c81f7cbb3d : SUCCESS in 54m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88f231d91fce483181a27dfdc17d0d72 : SUCCESS in 4m 42s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6ab0cfdd537d4b7d869953c81bb42aaa : SUCCESS in 1h 19m 23s","accounts_in_message":[],"_revision_number":2},{"id":"f6797bd98d1378c5e8658a296fe18e83c776360c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-09 16:12:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3c7a557d2a25ff67753949b29d0b1e25cf4305eb","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-09-10 06:36:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1598c76f8d391fe9588ca75d13237022ba69e468","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 06:37:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"866d76c20c7790083ecdbbcf29ca6f572dbfbe6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 10:32:49.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b98bb05ace14de197dfecc6a3e065ce : SUCCESS in 4m 21s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53e92fde04a749b8937a397851e41bbc : SUCCESS in 4m 01s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7394f4f7dfe4292a767c2e584ce8107 : SUCCESS in 4m 16s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/64618c2a0bc3415ba701250322f3e9ea : SUCCESS in 16m 53s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/35d2df9b127643439bd9c5a611455803 : SUCCESS in 16m 05s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd0de8eb36054fc196d28534ebe4cd32 : SUCCESS in 17m 37s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9bad01e963894f3ebc0d9ba5328266d6 : SUCCESS in 14m 32s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/78c78b6a6d4a4420a753832d1aeb5124 : FAILURE in 1h 19m 34s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb28483f5b9540ccaba93d2c22bc427b : SUCCESS in 1h 32m 31s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e364de2b26ae41839655089dfdc78e2c : SUCCESS in 1h 27m 03s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c228e91081994f90b805b2baa2cf54e3 : SUCCESS in 59m 35s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/6835943b04dd42cd93bbc31c26765755 : SUCCESS in 25m 11s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abd73cc212aa4592a8b61637cc35926f : SUCCESS in 20m 13s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/11f86b94a3a44185a5dfedf8fed8d25e : SUCCESS in 2h 21m 19s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/9dcc6625a1e542d097a620d271d8f25c : SUCCESS in 45m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87779cb02c064ee58f0de2fdd1680168 : SUCCESS in 4m 40s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/aea8231a6e02406eb28f7881ea37077b : SUCCESS in 1h 08m 35s","accounts_in_message":[],"_revision_number":2},{"id":"0bce37cac4a1e8bccd89340f851cf7409f0dde4a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-10 10:59:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ecdb60aff88c482d1d2bcfccdfbf2443e6fcca3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 14:09:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fbe503cb9d1455eaf4693dd955c5587 : SUCCESS in 5m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd8c227838bc4794a0b45e4f136a572a : SUCCESS in 5m 20s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e7463103cc643978f7688ea87484ff1 : SUCCESS in 5m 24s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9bd294cebe18445bbcfd660a6028f755 : SUCCESS in 17m 15s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/9e965a98206249a9bffbb8c5b8d819ed : SUCCESS in 20m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf92b537261047f69cf42fba9155a17f : SUCCESS in 14m 32s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37dedb6c87514ca1b89cded575fe2de1 : SUCCESS in 16m 34s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/706d7963ad4847e497b52bfa11f3d8a9 : SUCCESS in 13m 18s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1735a380525d4c359dc4da8ced633f7f : FAILURE in 29m 45s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9644a4309de348da9dff19b66a3228fa : SUCCESS in 1h 10m 41s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09b5ee4b84f14c3482c2447e83f9df21 : FAILURE in 22m 26s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75e244ce16834413996616599f91b3f9 : SUCCESS in 35m 23s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/15ab33de43924cbc9c79dfa6abf00db9 : SUCCESS in 24m 57s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/250677c81fd84675874e748189d862a0 : SUCCESS in 1h 00m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/669ea2cfd0a649a5862b397fc80328ec : FAILURE in 1h 03m 10s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e4025bd36eaf4e21b6f24b9e210c6edd : SUCCESS in 1h 07m 12s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/610de8650d354d929166a3c259745054 : FAILURE in 59m 09s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce821ba4600f47d1bb66027a3a562f29 : SUCCESS in 17m 37s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/e799b12ffd1f4bf98b58c5de8dfd8d0b : SUCCESS in 22m 13s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ee4110e6522466382918b6af2dc2e60 : SUCCESS in 19m 59s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f03fe16347a4ad0ab189f391f821823 : SUCCESS in 19m 40s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6a1235e0627c4a94ab3cc3319461e1e5 : SUCCESS in 3h 08m 34s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/8fd7ba20f2b440efba433059ad2962d8 : SUCCESS in 56m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98f0797ccf5d4f068232093976ae3e4c : SUCCESS in 5m 34s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c4986a42d15f40df84f58fb608f17f1d : SUCCESS in 1h 21m 14s","accounts_in_message":[],"_revision_number":2},{"id":"e7c22fac0a74fb42d4acbc3e4ae7e9eab3e702c1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-10 14:19:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"facea62ba6978921d8d4003171e3ffe98328d4f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 17:27:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98c1e0818e144f2abe4bb254ea866642 : SUCCESS in 5m 52s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5066fb81c7b946ecb529b645eb36ed21 : SUCCESS in 5m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f729d344a7f4ff2a46d0ab77efa2a8f : SUCCESS in 4m 40s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d63b2fa926c845c1832fe1c0c85e816e : SUCCESS in 19m 02s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/d2edd0a9e0af4bb2b29a066ae6e5e726 : SUCCESS in 20m 50s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b44bd1b546ea4bea88673f9ce40842d9 : SUCCESS in 19m 39s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54c7c68ab6b2444dad507c7f0f45b118 : SUCCESS in 16m 14s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d2797d533b540b1b22e428a8e39763c : SUCCESS in 21m 36s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8354498a2da34467b5f20f29bc920188 : SUCCESS in 50m 49s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91fe7accb2204ff78956f00bc015657c : SUCCESS in 52m 06s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/316f4d1cfff84a649539b78c87a9d8b8 : FAILURE in 37m 04s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/876efb4bfb604e348299ac053ae27c31 : SUCCESS in 45m 50s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/86db5ac927a04948ba633b3d744fe9bd : SUCCESS in 25m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/94ce2e028d70472a956a1481629dcb81 : SUCCESS in 1h 13m 16s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0990059add4a4451bdb8ba4502c45436 : SUCCESS in 1h 26m 39s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2812e68c8328404badace8a5ed2f8cb4 : SUCCESS in 1h 28m 06s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b85bc9b011d40838758ec126cc017c9 : SUCCESS in 59m 33s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/288c50f794904a79b4b3ff6d8f02b417 : SUCCESS in 20m 25s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/d5f139cd95c1467f82c43701f83caf9f : SUCCESS in 24m 40s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90166574529548539906e1a130684697 : SUCCESS in 20m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb59723f67014ae8a9e490f6f904436d : SUCCESS in 20m 11s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9a1d26c5167742e1af32d6a122ef9cdf : SUCCESS in 3h 02m 54s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/de81c5d32eef46a8930678afd1194523 : SUCCESS in 51m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1a40d3ba6614814ac015414fea071a8 : SUCCESS in 6m 09s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4d7447902efe4624ab0ea59f581c31fe : SUCCESS in 1h 13m 59s","accounts_in_message":[],"_revision_number":2},{"id":"dbd5d4c8ed39f0ef37f463f6eb88a8278e214fbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 17:30:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1fd20c83607dba69c9a42faf99a0053bc1f2b1d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 20:32:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df1ec76869f14273b853f799152b02ca : SUCCESS in 4m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/469491a745884762adb231b4824b358d : SUCCESS in 4m 21s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/595d37566fc04cbfbc2d593e53db6568 : SUCCESS in 4m 15s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/fee379aff0dc404db6808892ec3afa28 : SUCCESS in 18m 43s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c0911034805b43ef8b9784750369051a : SUCCESS in 19m 43s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4c78063e9c1438d966921540ecad1a7 : SUCCESS in 15m 09s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c53c234416c449dbba8a926d5b60f314 : SUCCESS in 14m 38s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/12aa12b0429146f5a3d77145d503e376 : SUCCESS in 1h 21m 44s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c9f45532c4704136be4c569423271af1 : SUCCESS in 1h 26m 29s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/589c4e8d4e854f5c87c416e47774051e : SUCCESS in 1h 18m 32s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/352e7fd2788440009a96cf4572ae9dc9 : SUCCESS in 1h 14m 35s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f728b1242e29409b94e4e1d443daf0ca : SUCCESS in 27m 14s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4533e2ed114450695d718363a7a55e3 : SUCCESS in 18m 42s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/4a2e3a4ddc7e40eb8fea955f0b2a5ca2 : SUCCESS in 2h 53m 01s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e5a47dcfceaa4b11a18fe66dfdea4328 : SUCCESS in 48m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9255b15ab7f4aa0b9e983e02712195d : SUCCESS in 5m 17s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/66e13f25c84948b894d9061b97888bed : SUCCESS in 1h 32m 11s","accounts_in_message":[],"_revision_number":2},{"id":"02e9e9a99809f4cc1597aaeb45fef7beecdad34f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 20:32:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"09949e8b22a5b772691a145aa2644fe5a2858bbd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 20:33:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4df8f80ee5db48a383164e8364c6b034 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f67ea44f7469f76be328579083a895c2df9bbeaf","revisions":{"c528c4a90ec13f50aa62f273db762844eb79af07":{"kind":"REWORK","_number":1,"created":"2021-09-07 15:13:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/807744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/44/807744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/44/807744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/44/807744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/44/807744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/44/807744/1"}}},"commit":{"parents":[{"commit":"9fc674d1191c571a82e21ec2b9dc82f26d26a440","subject":"Deprecate the useless quota_packet_filter parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9fc674d1191c571a82e21ec2b9dc82f26d26a440"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-07 15:12:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-07 15:13:06.000000000","tz":540},"subject":"Deprecate the unused quota_vip parameter","message":"Deprecate the unused quota_vip parameter\n\nThis parameter was used in Neutron LBaaS v1 but the implementation was\nalready removed in 2016[1], and the parameter is no longer used.\n\n[1] https://review.opendev.org/c/openstack/neutron-lbaas/+/286381\n\nChange-Id: I5dcdffd06b7dab598bb62a5cfcbc7833472196c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c528c4a90ec13f50aa62f273db762844eb79af07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c528c4a90ec13f50aa62f273db762844eb79af07"}]},"branch":"refs/heads/master"},"f67ea44f7469f76be328579083a895c2df9bbeaf":{"kind":"NO_CHANGE","_number":2,"created":"2021-09-07 15:18:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/807744/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/44/807744/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/44/807744/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/44/807744/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/44/807744/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/44/807744/2"}}},"commit":{"parents":[{"commit":"528290ea2bb6c402d615bb092fc849633d421494","subject":"Deprecate the useless quota_packet_filter parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/528290ea2bb6c402d615bb092fc849633d421494"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-07 15:12:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-07 15:18:34.000000000","tz":540},"subject":"Deprecate the unused quota_vip parameter","message":"Deprecate the unused quota_vip parameter\n\nThis parameter was used in Neutron LBaaS v1 but the implementation was\nalready removed in 2016[1], and the parameter is no longer used.\n\n[1] https://review.opendev.org/c/openstack/neutron-lbaas/+/286381\n\nChange-Id: I5dcdffd06b7dab598bb62a5cfcbc7833472196c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f67ea44f7469f76be328579083a895c2df9bbeaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f67ea44f7469f76be328579083a895c2df9bbeaf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
