)]}'
{"id":"openstack%2Fpuppet-neutron~814554","triplet_id":"openstack%2Fpuppet-neutron~master~Ie8b347a4705045dfa0363284ec41e9e519ec19f9","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-08 13:09:53.000000000","reason":"Change was submitted"},"25241":{"account":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"last_update":"2021-11-08 13:09:53.000000000","reason":"Change was submitted"},"12171":{"account":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"last_update":"2021-11-08 13:09:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie8b347a4705045dfa0363284ec41e9e519ec19f9","subject":"Change authentication method for Mellanox sdn controller","status":"MERGED","created":"2021-10-19 11:12:30.000000000","updated":"2021-11-08 13:11:40.000000000","submitted":"2021-11-08 13:09:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"814554","meta_rev_id":"9ebdf4bc8470d1e1de82e8da3665e021776cfc9b","_number":814554,"virtual_id_number":814554,"owner":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-08 13:09:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-11-08 07:10:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-11-08 09:57:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-11-08 09:57:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-19 11:35:35.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-10-19 12:59:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-21 06:55:55.000000000","updated_by":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"reviewer":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"state":"REVIEWER"},{"updated":"2021-11-08 09:57:36.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"a0d144ffec12b9922254c2a59937a220f224156c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"date":"2021-10-19 11:12:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e0245764032875060b0de87db26e7210ae4c994","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-19 11:35:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"895b2ae27d3775b1f07959569b3e53185d38be1a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-19 11:42:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"911db2a98f605059e5c9098f0c1d4b3defd4aa82","author":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"date":"2021-10-19 12:00:51.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"9292d2045a571fb31316c64914578558662b6569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 12:59:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cbe63463110411da5cd2bbd0eacbfaf : FAILURE in 3m 55s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6f4668251cd48f0b1c0695fdbc06b8c : SUCCESS in 4m 16s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db93b94e4c9a4239bc3a6930917ff113 : SUCCESS in 3m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/fdc0874aef8f4951abc7c91bc2386958 : SUCCESS in 17m 22s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba3e6b31d2f04c59a08684a121a96300 : SUCCESS in 16m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01942771b3e8412eb82db8a6cc040e70 : SUCCESS in 14m 46s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f5544052ee44189a2de76213faf6154 : SUCCESS in 13m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b23bf0cb3bc34fb4a4c3190f736eefdf : SUCCESS in 13m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d06c86925c9e4be0a45a7e8b38eca7ff : FAILURE in 10m 09s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46de53c0d7df405ebaffe2158d2f5add : FAILURE in 7m 32s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a08afbe4a7ca46f9b7e039687fad72a6 : FAILURE in 7m 09s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07c108cadee946e29ecd16bd939ea898 : FAILURE in 10m 00s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/aba644195a1c458b83487da77fb5fd7c : SUCCESS in 23m 36s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c8275fe4a49d40a8aa77a997ba62cf67 : SUCCESS in 1h 10m 39s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9bab231f2a014f0c8139bb826f0c7b40 : SUCCESS in 1h 21m 43s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4652527c08a34cf4a9bbb3e468efece8 : SUCCESS in 1h 06m 49s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/18cda67f402b49dd8dab8c988058373b : SUCCESS in 1h 00m 37s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78ff82c6a27d4c5aa56c640dd6c21c1c : SUCCESS in 14m 00s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/43911ac90cc94e859636a3a61ffdfa65 : SUCCESS in 22m 48s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c597bc40766f4f6e8d48657ad6e20e42 : SUCCESS in 16m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f51e4f1c464421d8127879b105ab334 : SUCCESS in 15m 04s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2869558dba1940bd92dcce0811ffbede : SUCCESS in 1h 45m 26s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f8f14baf051a4707bea0e84c2bf4b4d9 : SUCCESS in 39m 39s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d06126fb7658466ea41bc604354b50ea : FAILURE in 46m 06s","accounts_in_message":[],"_revision_number":1},{"id":"4f5d8d8fbf0ef48641ed943ec8919d37a0c4fc06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"date":"2021-10-27 13:49:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ec80f6d7f733f229b5e71c22ff361e750c6261d3","author":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"date":"2021-10-27 13:51:01.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"3188b19b1582691792f16746672c473236eabe3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 16:43:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/603258272247413db86bf67c17ea8614 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee0c91c747414deb8ddd731b222b3ab3 : SUCCESS in 4m 05s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0809c9fba2a4bce8890e9441ae5cbe3 : SUCCESS in 5m 35s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4d593a118fe479385bfde51bcf37099 : SUCCESS in 19m 01s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/22ff2b97d3c54272b26ed67f409d898f : SUCCESS in 18m 06s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73e1e593883f4d06850f907649836db9 : SUCCESS in 15m 08s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b7d61feb30441fa8c5336a97e5772da : SUCCESS in 19m 26s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d229bd89b0e471194e1d7ded3a2fa9d : SUCCESS in 16m 59s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5acfcd4e5a24259a8ca2f11014f9a9f : FAILURE in 9m 55s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbf684d2361c453eb8b57c68ffce3f84 : FAILURE in 11m 27s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6cd985aeea14355bbfa92638172d03e : FAILURE in 11m 53s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65f8bed9da51425eb4a7392961aca07c : FAILURE in 8m 00s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ba495c2265c480790032ace52c961fc : SUCCESS in 22m 57s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/adf687b05cae4810b83094d08cd749bb : SUCCESS in 1h 16m 58s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/329ecf47b1da47fdb981d81c8e9490e9 : SUCCESS in 1h 35m 37s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fb0067c0a13b4b61900a49a24c8be7c7 : SUCCESS in 1h 24m 53s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/669c3de56df64bc9a00bba4876e613a1 : SUCCESS in 1h 05m 52s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76db9f8a1f0142329a367ac995d80c43 : FAILURE in 21m 21s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/6cd049746af84905beb37f0272e241a6 : SUCCESS in 22m 26s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eeaaf0419cf841179d737e3e71178f67 : SUCCESS in 20m 50s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04931047e19444e891a0320ebbc32087 : SUCCESS in 17m 08s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0a822adf734346cc93a8681b84856b1d : SUCCESS in 2h 44m 23s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/52d5d4ff18e34c4ab25821ebaef00679 : SUCCESS in 55m 59s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0e33b597154c40e5ae12ceeb0f92570d : SUCCESS in 1h 22m 26s","accounts_in_message":[],"_revision_number":2},{"id":"f6b886280193ceea3e3e3c8da037bf3ecd8676cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"date":"2021-10-28 05:57:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"954b4ae199512741bc2b9488deb1b2549b351a5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"date":"2021-10-28 06:27:44.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"db0b1ba30c8e4d0529bd7d0d6f5ee70c11f6fb83","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2021-10-28 06:34:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"963d1f023de6b71da21c1d3a255aa9251fd9245a","author":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"date":"2021-10-28 06:34:51.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5f042b67dde6a1e05f622ed17f965b56ef7176d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"date":"2021-10-28 06:43:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5f3b279aab08b5b503f0d81921e55dcfa94190f1","author":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"date":"2021-10-28 06:44:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"eaf6387434f3813908f84791c8cdf1630c16ce88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 09:22:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20de32d15b0d4646b51b7104e7308764 : SUCCESS in 4m 01s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c740269e108a49f79629559cd7f49e42 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2020d7462b4e4f74a6008cf500185414 : SUCCESS in 3m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a4fc4372f814010b3e5521c0ca6ee1a : SUCCESS in 17m 26s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/de10fc76089c4ae383e385598702b4c4 : SUCCESS in 17m 29s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2f912af3050489283dea67388dab3fa : SUCCESS in 14m 08s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6c6bd812f14475a8bc1a7b862986668 : SUCCESS in 14m 33s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1257513f7e604331986807d5b574e01a : SUCCESS in 15m 27s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adc31b1829274fed88b543c2afadfa81 : FAILURE in 7m 55s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b053d906d0ff4a51b98dd3dca308f9d8 : FAILURE in 8m 06s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61da7e823946429b910b7f6329597bf0 : FAILURE in 7m 20s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4293ffafcf4342d2832d53e358bec78a : FAILURE in 7m 22s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d583bc64f814a4f83b59f507698eb69 : SUCCESS in 23m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/453ba9e202e4462aa4d46204205da7f2 : SUCCESS in 1h 20m 30s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f65290a4973045f5b1da07a1a5a0d1f1 : SUCCESS in 1h 32m 23s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c5c07f4e9192453eb080bb2e7ce08238 : SUCCESS in 1h 50m 44s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/911f4e3da1514bfcbad7327fa9c4ef29 : SUCCESS in 1h 04m 31s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47f8b5b6fec141a0a0e1125e9806d473 : SUCCESS in 16m 31s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c03ff39a9b4a4c4b83f5056452e22cd8 : SUCCESS in 24m 20s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35d5251e30b64f18b111f3632d00bd0f : SUCCESS in 17m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/daaef7bef572412e8d0f4486e8f90f51 : SUCCESS in 17m 37s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ab3711e431864161bfb810a0f528189b : SUCCESS in 2h 37m 38s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a32d78ae8f7d4f25a2be177617a98990 : SUCCESS in 52m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c92dfe8b3dd34637985758b911685a32 : SUCCESS in 4m 06s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/07951690dc1049cbaefbd3685d8080e4 : SUCCESS in 1h 22m 53s","accounts_in_message":[],"_revision_number":5},{"id":"6e4ed8776240ed07484d1a8049edb8f791fb32cb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-28 11:23:50.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"e0fa5d7ee4d5a166e92ec279da3936419a5d711b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-28 11:24:03.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"90a25e6848bb4832fa45321125b0f2616e427b5c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-28 11:32:20.000000000","message":"Patch Set 6: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"0cbb60f1da2016dca201082b0f2bc398d39db1c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 14:19:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8c3c8bd9e314588a7dd3a05a78f187e : SUCCESS in 4m 33s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4affb4897fb14b99ba44761b09ab4cd5 : SUCCESS in 4m 28s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3163ed0c17345c09bce64a993570933 : SUCCESS in 5m 11s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6e1bd47ee45f4c9e9c6615ec057d4a1c : SUCCESS in 18m 19s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/68cb0ae242c840c9a07d63e8d4f8d9ed : SUCCESS in 19m 27s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64adf394206e4a9dbc3f3c2385d10c18 : SUCCESS in 15m 34s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3678a045fff549ccb52ba38eadbb6f16 : SUCCESS in 16m 17s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47187beeb48a483cb974c5aed2c9fa11 : SUCCESS in 20m 15s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e614aba74aa4e78a03b8276c8520924 : FAILURE in 10m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46855c4480a34167b9f4fbc125aa703d : FAILURE in 11m 26s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0706dd8719b444ebdc32a54502044be : FAILURE in 11m 35s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc79ef9804a74aec98ce64e58e6c3bf2 : FAILURE in 9m 09s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7db4de754d9d4c62aff8b06fc6e3394c : SUCCESS in 27m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e166f842123f4f73a329d09d91ecb087 : SUCCESS in 1h 26m 32s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cbe9976dccd643d0b341f0bfedce9c5a : SUCCESS in 1h 30m 26s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/43b5e92a575242a0be6bde38a7b85b6d : SUCCESS in 1h 26m 02s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d96a9b95f89649df89b29262fdc218db : SUCCESS in 1h 02m 12s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/548d44c1e6a94086b020f03ebcb07245 : FAILURE in 23m 51s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/041bf52628a94fec9bd4830a2640166a : SUCCESS in 27m 58s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5eb8f401016b4b09bc77ec39553f4d7a : FAILURE in 20m 12s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4b5a47fa44e472ebb40d9f8cc123a62 : FAILURE in 22m 59s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b23f457263c94a549fbec2ffe3257483 : SUCCESS in 2h 46m 38s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1ac5da29073a46179af75a503e2d9762 : SUCCESS in 53m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7921a9252eec4f418b01137c079bb795 : SUCCESS in 5m 29s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/30b91548dbef406c874285f80faa7915 : SUCCESS in 1h 18m 26s","accounts_in_message":[],"_revision_number":6},{"id":"c655cff7b472f21f2c1998934c52dad5497699c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-04 12:06:43.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"bd79272a20adc880130907605792aaa6c7844d5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 15:10:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09d3b408533d447db336cd7a7706a8cb : SUCCESS in 4m 42s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ed95edcf7be49a0bb7db618c3d090ae : SUCCESS in 4m 07s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4b88e5e9ed84dd88cb66ccfbca1a9e6 : SUCCESS in 6m 10s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/57c382dc53394ce19080837efb5e5a21 : SUCCESS in 18m 26s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/8596f4cf23d248b99abf6f104a1b7fb1 : SUCCESS in 17m 27s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51b8599da9c644d2948c3c4eb56d9206 : SUCCESS in 13m 51s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eda5a99bc7a845fe8d22db6e7b721c30 : SUCCESS in 13m 51s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40eb8b23cb344c2a82450b0caa6dc761 : SUCCESS in 15m 41s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/703a763398e04f168c7d86e3349832d1 : FAILURE in 1h 11m 38s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a0a9d90784a4a57a7e61f40d53275ed : FAILURE in 1h 11m 23s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98f92e0c762c451a93ec88139702cfca : FAILURE in 36m 03s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8dda0da962544ad182a210926798d54c : FAILURE in 25m 51s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/91f0d136e5584dc9a3ac5d3c8b18420f : SUCCESS in 23m 44s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a88c40daac144631b60e839ae409ea27 : SUCCESS in 1h 16m 00s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8337f744002640f7a0aa4110b7b878e0 : SUCCESS in 1h 22m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cba69348184d479098bc96ac1e05adba : SUCCESS in 1h 19m 25s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e013d64cbf1746c5bb856508a5d2870c : SUCCESS in 1h 01m 49s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35a3afcedb794d67bf8c931ff4658adf : FAILURE in 20m 20s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2bbd06fe867c44059ab576d2febf2060 : SUCCESS in 32m 22s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3e539d422ea473a929c681d1ddbff86 : FAILURE in 18m 23s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/732caf94ebf247629c5e6befca3c6bb5 : FAILURE in 21m 06s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/bf046f4d97da4e988c315a70902c6b8a : SUCCESS in 3h 00m 46s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/358016bf09ab4f1e89b2e9e9508fd70a : SUCCESS in 46m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b94a9b29f674d4d82140bdb17f748bb : SUCCESS in 4m 26s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/728ea98e1c894a219c3980323d9728a7 : SUCCESS in 1h 25m 18s","accounts_in_message":[],"_revision_number":7},{"id":"30d162d58a3278d7f2c5f185d2cf4502e8a99d15","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-08 07:10:54.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"a16ef966330255c55496741f18f312104ef69a91","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-08 09:57:36.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"bf2bd3f7f1f1994a979e65cf8809ba55ef7c29bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 09:59:34.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"083b1df4ab522b4b503a6fc1defe7d033a06bc8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 13:09:53.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e809db914909413fa9ed175ad62d8a7c : SUCCESS in 4m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98aa4cf646a94410a8c5f1ba05b35802 : SUCCESS in 4m 30s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f71f82a10ae142629ae8c3fa9b18259b : SUCCESS in 4m 39s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4f2b635e29734702bde4fb9683c9e215 : SUCCESS in 18m 30s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f5bbff8e8bf04d388f9d7eec0e5332d6 : SUCCESS in 17m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ccde14164c2142889601de237cf0df7e : SUCCESS in 14m 37s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4595eb4db0214598b3da59b1f929a82f : SUCCESS in 14m 28s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa3f618a70d04ad081e6e61619d12c98 : SUCCESS in 1h 21m 59s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f73e4b97a03e4f5696792d58aedff952 : SUCCESS in 1h 52m 58s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/39f9ae73e03d4959a7ca36756477a173 : SUCCESS in 1h 31m 31s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3be64b8ac885421291b7053aa51aac59 : SUCCESS in 1h 07m 38s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/30e5593bcb25465180334702f48e455a : SUCCESS in 22m 55s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3459e85d918243af84abda1445d300f2 : SUCCESS in 16m 47s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/94c8af34e80f4126a24da43c6804a1e3 : SUCCESS in 3h 04m 24s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/282ddcbe08134698be63329833424a7a : SUCCESS in 50m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e65d3885405c4f81937934f1a859f0c3 : SUCCESS in 4m 15s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6aac1cb20d5c4f709f76f3ccc85dcadd : SUCCESS in 1h 31m 23s","accounts_in_message":[],"_revision_number":7},{"id":"50b611960141f543b75535b19415f4e3fb767c8c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 13:09:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"9ebdf4bc8470d1e1de82e8da3665e021776cfc9b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 13:11:40.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc5d349d706e442e842c31c8851247a6 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"8ce1ea3d936856354beff4a8df18d05fd5e2c6c7","revisions":{"2bbc8e91ba7bc630b226c78ae6de3c8b3400489b":{"kind":"REWORK","_number":1,"created":"2021-10-19 11:12:30.000000000","uploader":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"ref":"refs/changes/54/814554/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/54/814554/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/1"}}},"commit":{"parents":[{"commit":"dc2bdfd72c29f828c08e276acf8b82b05873284c","subject":"Merge \"ovn: Do not expect the separate plugin package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dc2bdfd72c29f828c08e276acf8b82b05873284c"}]}],"author":{"name":"waleedm","email":"waleedm@nvidia.com","date":"2021-10-19 11:02:14.000000000","tz":0},"committer":{"name":"waleedm","email":"waleedm@nvidia.com","date":"2021-10-19 11:12:06.000000000","tz":0},"subject":"Change sdn authentication from basic(username,password) to token authentication","message":"Change sdn authentication from basic(username,password) to token authentication\n\nChange-Id: Ie8b347a4705045dfa0363284ec41e9e519ec19f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2bbc8e91ba7bc630b226c78ae6de3c8b3400489b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2bbc8e91ba7bc630b226c78ae6de3c8b3400489b"}]},"branch":"refs/heads/master"},"67397e144f64fb95284eecbd45f35c68c0d8f947":{"kind":"REWORK","_number":2,"created":"2021-10-27 13:49:58.000000000","uploader":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"ref":"refs/changes/54/814554/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/54/814554/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/2"}}},"commit":{"parents":[{"commit":"dc2bdfd72c29f828c08e276acf8b82b05873284c","subject":"Merge \"ovn: Do not expect the separate plugin package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dc2bdfd72c29f828c08e276acf8b82b05873284c"}]}],"author":{"name":"waleedm","email":"waleedm@nvidia.com","date":"2021-10-19 11:02:14.000000000","tz":0},"committer":{"name":"waleedm","email":"waleedm@nvidia.com","date":"2021-10-27 13:49:35.000000000","tz":0},"subject":"Change sdn authentication from basic(username,password) to token authentication","message":"Change sdn authentication from basic(username,password) to token authentication\n\nChange-Id: Ie8b347a4705045dfa0363284ec41e9e519ec19f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/67397e144f64fb95284eecbd45f35c68c0d8f947"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/67397e144f64fb95284eecbd45f35c68c0d8f947"}]},"branch":"refs/heads/master"},"754e073d49c8c285c31e40fc1aecf945aaada74d":{"kind":"REWORK","_number":3,"created":"2021-10-28 05:57:23.000000000","uploader":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"ref":"refs/changes/54/814554/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/54/814554/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/3"}}},"commit":{"parents":[{"commit":"dc2bdfd72c29f828c08e276acf8b82b05873284c","subject":"Merge \"ovn: Do not expect the separate plugin package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dc2bdfd72c29f828c08e276acf8b82b05873284c"}]}],"author":{"name":"waleedm","email":"waleedm@nvidia.com","date":"2021-10-19 11:02:14.000000000","tz":0},"committer":{"name":"waleedm","email":"waleedm@nvidia.com","date":"2021-10-28 05:53:50.000000000","tz":0},"subject":"Change authentication method for Mellanox sdn controller","message":"Change authentication method for Mellanox sdn controller\n\nChange sdn authentication from basic(username,password) to token\nauthentication, deprecate username and password and add new\nparameter token\n\nChange-Id: Ie8b347a4705045dfa0363284ec41e9e519ec19f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/754e073d49c8c285c31e40fc1aecf945aaada74d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/754e073d49c8c285c31e40fc1aecf945aaada74d"}]},"branch":"refs/heads/master"},"d45e3b36b7c35505ac80e99af5aeebb425d2551b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-10-28 06:27:44.000000000","uploader":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"ref":"refs/changes/54/814554/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/54/814554/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/4"}}},"commit":{"parents":[{"commit":"dc2bdfd72c29f828c08e276acf8b82b05873284c","subject":"Merge \"ovn: Do not expect the separate plugin package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dc2bdfd72c29f828c08e276acf8b82b05873284c"}]}],"author":{"name":"waleedm","email":"waleedm@nvidia.com","date":"2021-10-19 11:02:14.000000000","tz":0},"committer":{"name":"waleed mousa","email":"waleedm@nvidia.com","date":"2021-10-28 06:27:44.000000000","tz":0},"subject":"Change authentication method for Mellanox sdn controller","message":"Change authentication method for Mellanox sdn controller\n\nChange sdn authentication from basic(username, password) to token\nauthentication, deprecate username and password and add a new\nparameter token\n\nChange-Id: Ie8b347a4705045dfa0363284ec41e9e519ec19f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d45e3b36b7c35505ac80e99af5aeebb425d2551b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d45e3b36b7c35505ac80e99af5aeebb425d2551b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6d8b50161227325b540016cf47eeb074a47d3589":{"kind":"REWORK","_number":5,"created":"2021-10-28 06:43:03.000000000","uploader":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"ref":"refs/changes/54/814554/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/54/814554/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/5"}}},"commit":{"parents":[{"commit":"dc2bdfd72c29f828c08e276acf8b82b05873284c","subject":"Merge \"ovn: Do not expect the separate plugin package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dc2bdfd72c29f828c08e276acf8b82b05873284c"}]}],"author":{"name":"waleedm","email":"waleedm@nvidia.com","date":"2021-10-19 11:02:14.000000000","tz":0},"committer":{"name":"waleedm","email":"waleedm@nvidia.com","date":"2021-10-28 06:42:53.000000000","tz":0},"subject":"Change authentication method for Mellanox sdn controller","message":"Change authentication method for Mellanox sdn controller\n\nChange sdn authentication from basic(username, password) to token\nauthentication, deprecate username and password and add a new\nparameter token\n\nChange-Id: Ie8b347a4705045dfa0363284ec41e9e519ec19f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6d8b50161227325b540016cf47eeb074a47d3589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6d8b50161227325b540016cf47eeb074a47d3589"}]},"branch":"refs/heads/master"},"3194f7d056fcf074a19045ef26447cf79f560510":{"kind":"REWORK","_number":6,"created":"2021-10-28 11:24:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/814554/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/54/814554/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/6"}}},"commit":{"parents":[{"commit":"dc2bdfd72c29f828c08e276acf8b82b05873284c","subject":"Merge \"ovn: Do not expect the separate plugin package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dc2bdfd72c29f828c08e276acf8b82b05873284c"}]}],"author":{"name":"waleedm","email":"waleedm@nvidia.com","date":"2021-10-19 11:02:14.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-28 11:23:54.000000000","tz":540},"subject":"Change authentication method for Mellanox sdn controller","message":"Change authentication method for Mellanox sdn controller\n\nChange sdn authentication from basic(username, password) to token\nauthentication, deprecate username and password and add a new\nparameter token\n\nChange-Id: Ie8b347a4705045dfa0363284ec41e9e519ec19f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3194f7d056fcf074a19045ef26447cf79f560510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3194f7d056fcf074a19045ef26447cf79f560510"}]},"branch":"refs/heads/master"},"8ce1ea3d936856354beff4a8df18d05fd5e2c6c7":{"kind":"REWORK","_number":7,"created":"2021-11-04 12:06:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/814554/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/54/814554/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/54/814554/7"}}},"commit":{"parents":[{"commit":"dc2bdfd72c29f828c08e276acf8b82b05873284c","subject":"Merge \"ovn: Do not expect the separate plugin package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dc2bdfd72c29f828c08e276acf8b82b05873284c"}]}],"author":{"name":"waleedm","email":"waleedm@nvidia.com","date":"2021-10-19 11:02:14.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-04 12:06:14.000000000","tz":540},"subject":"Change authentication method for Mellanox sdn controller","message":"Change authentication method for Mellanox sdn controller\n\nChange sdn authentication from basic(username, password) to token\nauthentication, deprecate username and password and add a new\nparameter token\n\nDepends-on: https://review.opendev.org/814557\nChange-Id: Ie8b347a4705045dfa0363284ec41e9e519ec19f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/8ce1ea3d936856354beff4a8df18d05fd5e2c6c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/8ce1ea3d936856354beff4a8df18d05fd5e2c6c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
