)]}'
{"id":"openstack%2Fpuppet-neutron~815796","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fstein~I87dba40544a76991f5fa16705ab1af00f75e5dc9","project":"openstack/puppet-neutron","branch":"stable/stein","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2021-10-29 12:05:14.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-10-29 12:05:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I87dba40544a76991f5fa16705ab1af00f75e5dc9","subject":"Support arbitrary configurations for ovs agent and sriov agent","status":"ABANDONED","created":"2021-10-28 08:13:22.000000000","updated":"2021-10-29 12:05:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":815791,"cherry_pick_of_patch_set":1,"meta_rev_id":"07d45f7dbe1a6bbd5e02f8965e74343de9ffbec0","_number":815796,"virtual_id_number":815796,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-10-28 09:59:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-28 08:13:22.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-10-28 09:59:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ebc0a438a91acfcbfe262f4d10024cbd56aa9f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-10-28 08:13:22.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"75a6edf51eee2fe8c8890238a3a5399aee875f4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-28 08:35:51.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3f6cf507f4f7b90e32f0f4bacb9eb46d05159dd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 09:59:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba5c968a442b4dad845ac90284a0c654 : FAILURE in 5m 39s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2c5d9b930b74313874554e2334dbb79 : FAILURE in 5m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83819869f28949dc97093913664163e5 : FAILURE in 7m 37s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9d226fdc8094672980a4bf59916b757 : FAILURE in 7m 13s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/b4125f4e0c8c48f0955d8d24ff3dc1dc : SUCCESS in 29m 43s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/1a3de88eb0bb4c90b2fd03b8fc69441c : SUCCESS in 38m 35s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d4aca1db5684a50942b2b73bc176dff : FAILURE in 10m 50s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03aec34a83564890bd9d5376120e7e4f : FAILURE in 11m 57s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/e7fd4be098d442ee8d11d159b57f5fc2 : SUCCESS in 1h 00m 05s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/778f6e3ff7fa41e4a737247d00063d3d : SUCCESS in 1h 01m 13s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b0d7d92fbb624da08365f4592d1b153b : SUCCESS in 1h 10m 49s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/a1fbb05257324dcdba8dd00fda3dfd39 : SUCCESS in 48m 49s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/023cd5b75a224371b8ff68b46a99416c : FAILURE in 29m 14s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98bda6f6c272495a82849d2d243f394a : FAILURE in 25m 17s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd468fa7176d404aa81668181b1135f0 : FAILURE in 28m 54s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d6bf9aa80fe947348b23e5cda75aafac : FAILURE in 25m 03s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4e0cf99ef76c449a948d10570a451ebf : SUCCESS in 20m 15s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6cb9e7d80294f80923f29c8c4720f23 : FAILURE in 9m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/762062f243904ee8a47b10161da423fd : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":2},{"id":"07d45f7dbe1a6bbd5e02f8965e74343de9ffbec0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-29 12:05:14.000000000","message":"Abandoned\n\nWe are retiring stein and older","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9b9945773150520f1370c6dfbd7cd8b67a9f35ac","revisions":{"67ed6018e11e57f48e7eaabbb7ef6fa21b634ea5":{"kind":"REWORK","_number":1,"created":"2021-10-28 08:13:22.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/96/815796/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/96/815796/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/96/815796/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/96/815796/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/96/815796/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/96/815796/1"}}},"commit":{"parents":[{"commit":"471738d1c6230c67726311f33505dd8c072888f7","subject":"Stein only: Remove tripleo job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/471738d1c6230c67726311f33505dd8c072888f7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-07 13:19:16.000000000","tz":540},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2021-10-28 08:13:22.000000000","tz":0},"subject":"Support arbitrary configurations for ovs agent and sriov agent","message":"Support arbitrary configurations for ovs agent and sriov agent\n\nThis patch introduces new parameters to the neutron::config class,\nto allow us to add arbitary configurations into openvswitch_agent.ini\nand sriov_agent.ini .\n\nConflicts:\n    manifests/config.pp\n    spec/classes/neutron_config_spec.rb\n\nChange-Id: I87dba40544a76991f5fa16705ab1af00f75e5dc9\n(cherry picked from commit 34eef40377f32d44838ea1a3d7471f07c0fe6528)\n(cherry picked from commit 98e2fc0b911908b1ba9339104d844175e8201769)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/67ed6018e11e57f48e7eaabbb7ef6fa21b634ea5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/67ed6018e11e57f48e7eaabbb7ef6fa21b634ea5"}]},"branch":"refs/heads/stable/stein"},"9b9945773150520f1370c6dfbd7cd8b67a9f35ac":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-10-28 08:35:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/815796/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/96/815796/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/96/815796/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/96/815796/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/96/815796/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/96/815796/2"}}},"commit":{"parents":[{"commit":"471738d1c6230c67726311f33505dd8c072888f7","subject":"Stein only: Remove tripleo job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/471738d1c6230c67726311f33505dd8c072888f7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-07 13:19:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-28 08:35:41.000000000","tz":540},"subject":"Support arbitrary configurations for ovs agent and sriov agent","message":"Support arbitrary configurations for ovs agent and sriov agent\n\nThis patch introduces new parameters to the neutron::config class,\nto allow us to add arbitary configurations into openvswitch_agent.ini\nand sriov_agent.ini .\n\nConflicts:\n    manifests/config.pp\n    spec/classes/neutron_config_spec.rb\n\nChange-Id: I87dba40544a76991f5fa16705ab1af00f75e5dc9\n(cherry picked from commit 34eef40377f32d44838ea1a3d7471f07c0fe6528)\n(cherry picked from commit 98e2fc0b911908b1ba9339104d844175e8201769)\n(cherry picked from commit dc15af35cfadbaa025346e250f8975e6e0be491f)\n(cherry picked from commit 817d45342ad21148efdc365fd3e998e1e0bbab7c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9b9945773150520f1370c6dfbd7cd8b67a9f35ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9b9945773150520f1370c6dfbd7cd8b67a9f35ac"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
