)]}'
{"id":"openstack%2Fpuppet-neutron~817099","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fwallaby~Ie8b347a4705045dfa0363284ec41e9e519ec19f9","project":"openstack/puppet-neutron","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-20 18:18:52.000000000","reason":"Change was submitted"},"25241":{"account":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"last_update":"2021-12-20 18:18:52.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2021-11-09 06:56:40.000000000","reason":"Tobias Urdin replied on the change"},"12171":{"account":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"last_update":"2021-12-20 18:18:52.000000000","reason":"Change was submitted"},"18575":{"account":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"last_update":"2021-12-20 18:18:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie8b347a4705045dfa0363284ec41e9e519ec19f9","subject":"Change authentication method for Mellanox sdn controller","status":"MERGED","created":"2021-11-09 06:22:03.000000000","updated":"2021-12-20 18:20:46.000000000","submitted":"2021-12-20 18:18:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"817099","cherry_pick_of_change":818319,"cherry_pick_of_patch_set":2,"meta_rev_id":"aa0426ac0caa08a6a2bb12be6e26dd4c225d7819","_number":817099,"virtual_id_number":817099,"owner":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-20 18:18:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-12-19 14:16:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-12-20 15:30:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-12-20 15:30:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-09 06:35:11.000000000","updated_by":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-11-09 06:35:11.000000000","updated_by":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-11-09 06:35:11.000000000","updated_by":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"reviewer":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"state":"REVIEWER"},{"updated":"2021-11-09 06:35:11.000000000","updated_by":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"reviewer":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"state":"REVIEWER"},{"updated":"2021-11-09 09:09:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-20 15:30:09.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"243c94248a35c577db14b733a58c94b06fe7bdbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"date":"2021-11-09 06:22:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e6f806c71223f1384cb517481fd6f6be24849db2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-09 06:56:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"28ac1463668e7514a21f31a299c54e65119489fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 09:09:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/843e08fc2cc748779c3920ca665e78d6 : SUCCESS in 4m 49s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0433576ba7864c9da0e2078657a59d4d : SUCCESS in 4m 31s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed5927e52cc14392aa7fda83d3f30319 : SUCCESS in 17m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b227ed8363f4cbea90d9c4ad381e542 : SUCCESS in 13m 13s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/396621a0ce9341378754f4ae4c8c56b5 : SUCCESS in 14m 49s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8778d58e44c7415b8ca9ad37128fd6d2 : FAILURE in 32m 30s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6389ab96418e41f6832b50ce43dc410f : FAILURE in 40m 44s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/536a205f3e954cc48595da01ca25db55 : FAILURE in 33m 54s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d7c2fb0b2a14ca2a9a49aa06e996551 : FAILURE in 26m 30s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/08337e50d5bb46dcbec6751c38645d5f : SUCCESS in 1h 12m 50s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/069ee0245146433e84206c95a637bad6 : TIMED_OUT in 2h 03m 36s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ce22f93e00f4385a7518e80ecf016c0 : SUCCESS in 1h 11m 34s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/83e232d079914a48800b38bc84d10be2 : SUCCESS in 57m 29s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/d48d43595be548c5b4c011f3f86eb453 : SUCCESS in 22m 55s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5a3b1198871482594efffb9ba5720b3 : FAILURE in 15m 54s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/cdaef6235c3e4339b6e7707d05bcabcd : SUCCESS in 2h 43m 14s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ab788670f3da47a88d1e63b43ec637d9 : SUCCESS in 1h 37m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea59b80abf584c27a734b733403ddb07 : SUCCESS in 4m 18s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a1509870776c47daa6f933f5d902c62f : SUCCESS in 55m 56s","accounts_in_message":[],"_revision_number":1},{"id":"63ab47a4aa6675d39f463aa07f0f6578e9976b18","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-09 09:17:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b6e87326498f8c175ce3c8bf2eac5af24a7afe60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"date":"2021-11-18 09:26:22.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"846c14e70441cb5659c2718727b9858a80d4bdac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 10:36:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8d5ae567b484c90a9f9d15d50888e1f : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/314f863baeba4692b4ca46a1e9aaea46 : SUCCESS in 3m 03s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/06170c3440424d65beb1dabe5d283650 : RETRY_LIMIT in 5m 45s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6793628b24b4a5b81ea3b8ecf3ce6e0 : SUCCESS in 14m 10s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21a0972f79ca4b1e9e099af14bfe6a74 : SUCCESS in 15m 23s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6593978ba732404cbc42c3f4eb8e0b99 : FAILURE in 36m 21s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7b111ca0e444834b8977d3370d92513 : FAILURE in 47m 17s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a0f71052241410b95cf8147e5665851 : FAILURE in 1h 01m 49s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bf1d16762c34190a1799daf5416770e : FAILURE in 27m 09s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2487d7197be94b13817c82cd1e20a7be : RETRY_LIMIT in 5m 40s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/209457edf920456a91ae94648b5ffadc : RETRY_LIMIT in 6m 09s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/90485b6cd08e410686f081d263786319 : RETRY_LIMIT in 5m 40s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/26db629fb0914c0d84fb019ca8ef0154 : RETRY_LIMIT in 7m 19s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/8211a527649044049a3a8e9049c2e02b : RETRY_LIMIT in 5m 29s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47de7654412147f7b42bc30d07472b50 : FAILURE in 15m 22s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c50a6433b7c94087bc67cd9d534e35f2 : RETRY_LIMIT in 6m 28s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1497e8844ee141108152a0850d6bc707 : SKIPPED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51312c61e9454da99db66e2b74a8120e : SUCCESS in 4m 15s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/507ae190d46a46c5a44f918bd8b6cdbe : SKIPPED","accounts_in_message":[],"_revision_number":2},{"id":"c6892ba8817947865e9c416e2dd385740b197b33","author":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"date":"2021-11-21 06:17:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2894247001ee68838e289ebc8ef96aecc1384e78","author":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"date":"2021-11-21 12:29:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"814d243ae1f69bba0c0d72e247bbe3bcedb5d224","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-21 15:00:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d15844a348c64e37bd0981c7791b69d1 : SUCCESS in 3m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bdaba2d858141d08352625007448bac : SUCCESS in 3m 32s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3dba2983932c49e6a064a1046b2e882c : SUCCESS in 16m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f21122c8f04c4e49afa6c5e18745e2aa : SUCCESS in 15m 48s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b4ba24416ed4f7cba674c03b8e3cbc4 : SUCCESS in 15m 42s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5739493757a420c8ebf6c3dc02be444 : FAILURE in 37m 32s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e062a0683e3c466e9f82b2febe637531 : FAILURE in 32m 41s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8d7646df9a448e59b1806571a47de85 : FAILURE in 32m 50s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca2ffba2aa284eb587b368e74ad5b107 : FAILURE in 12m 50s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc8831aecab74bf6acb6b06b27c0b8ce : SUCCESS in 1h 22m 00s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d81cf9b3af8f484c8e408c0ef3db6f0f : SUCCESS in 1h 14m 27s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e0a1a72231044917929078be6d44c614 : SUCCESS in 1h 23m 11s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2c5b3caba6148ddb8a84ef3e2b34b2a : SUCCESS in 47m 43s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/77d466e827634547ae7e34db028ccca1 : SUCCESS in 22m 02s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c08a893d540b480292752cecd8897193 : FAILURE in 19m 18s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/cf79dd386a9e4d3bb6a3d698f3731bb0 : SUCCESS in 2h 23m 08s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/312517ec345c4a62b7261068ca60c5ea : SUCCESS in 1h 02m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd2d289926b043d1851bd8ac2fd5aa03 : SUCCESS in 4m 03s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ac6f6127b84c4ee0b0358f6c3a091c9c : SUCCESS in 1h 06m 39s","accounts_in_message":[],"_revision_number":2},{"id":"2f67c1bfe6e5b89e4a663963529ffedc9f942df7","author":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"date":"2021-12-12 07:19:09.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"95bb582e2464db55a98637cc17c3bc84d6367786","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"date":"2021-12-15 06:03:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c0e20a7d32df1ba432ccea5f7c7ffb16be4ba1e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 09:07:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41731b1d5c16409ea4bbcddee08f5195 : SUCCESS in 4m 06s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0c5f316f2c7497393ab9b5444000461 : SUCCESS in 3m 17s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/15eeab0026614f5080ac657089ae3d50 : SUCCESS in 14m 29s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75f9f10eac7840bf8963648881fe72aa : SUCCESS in 12m 19s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4527182924ee40e2a5f28aa7c6f3d32f : SUCCESS in 11m 19s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c72b86eaebc046fcaff51f2f4cfaff35 : FAILURE in 37m 56s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf7cd8fc3dc048f49437125d2ce0ca6c : FAILURE in 36m 54s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb878df9d82c4a1bb931e96cfdc6f8f5 : FAILURE in 33m 24s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff5a63734e1b490b85ef012212714f32 : FAILURE in 25m 51s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f4400ea963894598816ccd72787b3fa8 : SUCCESS in 1h 05m 02s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2540d00ade9644fa99715d0831882a6d : SUCCESS in 1h 19m 30s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dfe16e1bd1e140b3957cfeffe66b8ab8 : SUCCESS in 1h 12m 26s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/950e0f9d10864254b839d070c0e61405 : SUCCESS in 50m 36s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/a3f360b5f4034ff886fb0414c2f3fe91 : SUCCESS in 21m 17s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/284712d495c04383af7b4d8ed7042e52 : SUCCESS in 15m 37s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/4d48c47f30394d579766d07b692c762b : SUCCESS in 3h 01m 40s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e05c8c8b46734c73a136472a0143f6d4 : SUCCESS in 57m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d83e7eed651e438cb797304b3c5929f7 : SUCCESS in 3m 18s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a968cd027057451596681c8ce71be64c : FAILURE in 1h 28m 07s","accounts_in_message":[],"_revision_number":3},{"id":"fd3c7c0c90ee815d93818b88fa1f7138a2393b62","author":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"date":"2021-12-15 11:08:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"154e8be66b4e58beb6cec2a96e28a174ca862472","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 13:49:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edb93ce3523f4398939c6e3e32576209 : SUCCESS in 3m 46s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36c9dc3e05d44011920732ee8994b2cf : SUCCESS in 3m 32s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/32f09d6f60bd4759b9184adf7e5bbfc0 : SUCCESS in 16m 56s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/792210bb8c374165b6696c38795e6078 : SUCCESS in 14m 02s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff6fda47dd20433f8b9ac7c821c168cc : SUCCESS in 14m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed3c21b255f447ba9a96b068bee49ff2 : FAILURE in 38m 39s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc66c5fd4a1f4dc3abcd1566f7090312 : FAILURE in 38m 51s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26ed78420ae54420985e673368856fdd : FAILURE in 32m 26s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2eeb8d5480e846d5b8550fa22ed1808a : FAILURE in 24m 33s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b229c955d1bf47a3ab4bca5164519923 : SUCCESS in 1h 11m 26s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9b0cd653837d41639a97925df3eb7c74 : SUCCESS in 1h 27m 01s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3657acabdcc14842bed80667d8737b85 : SUCCESS in 1h 26m 31s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/35f8753927ed4583a0ccb461a96b0665 : SUCCESS in 1h 07m 13s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/9598842c5d564d6992f498fa3b5c6edf : SUCCESS in 25m 50s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e2c385d02ab42af848029d2cbc4b730 : FAILURE in 17m 14s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/341b1ec44c104ff5b65852463ef46239 : SUCCESS in 2h 38m 01s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c4ec7f1603a348d390a9145e29f1b7af : SUCCESS in 1h 02m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/900405957a35440494967905e3588c0b : SUCCESS in 4m 32s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f9eee799ebf04d378f3cdf1804f2907c : SUCCESS in 1h 10m 20s","accounts_in_message":[],"_revision_number":3},{"id":"0d518cb2780be79343e056f5bc101673185d1be6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-19 14:15:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a0b4fd04cf368a8ed647dbf0a81acffaf73397ec","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-19 14:16:28.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"25642cc881cd32fcbf818518f629dc51184392a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-19 16:42:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5a8d56298af4610ae7930a5bfbe4d7b : SUCCESS in 3m 10s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1c16f16b310433ea984b9852f51bba6 : SUCCESS in 4m 03s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/88d403cc00c04397b41087651f7276e1 : SUCCESS in 16m 39s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6c572c51ab643eebcbecad92f47ae1f : SUCCESS in 15m 28s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f441dc9155334dc989a24685debfb0cd : SUCCESS in 15m 07s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bcb6ef339654e028df79e5ba721bf6c : FAILURE in 50m 52s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/575dddfe034c40c3b8ef02ee3087cc60 : FAILURE in 1h 07m 29s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65e87cfe2e2e40b78df88928a175ba64 : FAILURE in 27m 46s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/572d93bb95794760921010b0971145f6 : FAILURE in 27m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8a2c42cf6d26425abe419a7a7a563f13 : SUCCESS in 1h 04m 47s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d13b18fd4d3047b69dc99b23fad57396 : SUCCESS in 1h 29m 12s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd48256598154888beab98aba8beb59a : SUCCESS in 1h 00m 25s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e94ac70d54b545fa89f78246689881b5 : SUCCESS in 1h 07m 04s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a63d1990e1a438fab0f2d1268c80072 : SUCCESS in 17m 02s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cca8d3f74054c0090b3d8e7e2e47e49 : FAILURE in 16m 46s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f406aca5ad04478b89eff5118c6f00e2 : SUCCESS in 2h 21m 43s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/6f8eb89adb7c4e37a6d37c36090c7d5d : SUCCESS in 46m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34ecf95d844c44df9c770a9e84dae093 : SUCCESS in 4m 19s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4d6a051917f443ff89386555ed3bd682 : SUCCESS in 51m 53s","accounts_in_message":[],"_revision_number":4},{"id":"221a2700df1fe2a6834c54ed976bf781033d6810","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-20 15:30:09.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3372788bf5147c2fd8cde2f61749fdb6f729fd8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 15:32:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3efd359b82d6cd0cf4080088acd03c26b49ffc1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 18:18:52.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3046bbe11254fa6a36cd25632cc3df8 : SUCCESS in 3m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc661e7fb9dc4e8395c5008cf6816a10 : SUCCESS in 3m 12s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8ca6e088ad884f00962b4665a241359c : SUCCESS in 15m 25s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc957f736133460a9962211a3155cb57 : SUCCESS in 13m 06s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7aeae865d84d4959923bfa4f1ea49fec : SUCCESS in 1h 23m 15s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/88a4dbf88279401eb93bd60c4bffbffc : SUCCESS in 1h 21m 58s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4db746c7667148559a8c0658b7e3a001 : SUCCESS in 1h 20m 53s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4d634f6e1be443279b5a70c294ebe0c0 : SUCCESS in 1h 06m 25s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/90f64f9554ff4bfe8e8fb5725cd0b3ed : SUCCESS in 24m 40s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7080dcf89c594e6bbf215d791d7eac41 : FAILURE in 17m 36s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7e13d003257d4e4aa805a5dadedd15ed : SUCCESS in 2h 43m 05s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/23964767f1d041a488a66b4797fc242c : SUCCESS in 1h 02m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f77f85df11d43009d38a7725f03df1d : SUCCESS in 3m 46s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4b252ea08e1c4d389e94a3f73756455a : SUCCESS in 1h 17m 08s","accounts_in_message":[],"_revision_number":4},{"id":"d606d72b7800d762cf21512453d08cc90fb31a9e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 18:18:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"aa0426ac0caa08a6a2bb12be6e26dd4c225d7819","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 18:20:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/640ca6074820438db3764f702b307f36 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a58536a9571a282da81bd184a3a87fe3091e1ee5","revisions":{"bfced03d4d5945b2f5e0ef23be42192785e4e166":{"kind":"REWORK","_number":1,"created":"2021-11-09 06:22:03.000000000","uploader":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"ref":"refs/changes/99/817099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/99/817099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/99/817099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/99/817099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/99/817099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/99/817099/1"}}},"commit":{"parents":[{"commit":"bd97629f73bca7e8b70cf0f3cc15201902f2a49e","subject":"Add support for logging service plugin configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bd97629f73bca7e8b70cf0f3cc15201902f2a49e"}]}],"author":{"name":"waleedm","email":"waleedm@nvidia.com","date":"2021-10-19 11:02:14.000000000","tz":0},"committer":{"name":"waleed mousa","email":"waleedm@nvidia.com","date":"2021-11-09 06:22:03.000000000","tz":0},"subject":"Change authentication method for Mellanox sdn controller","message":"Change authentication method for Mellanox sdn controller\n\nChange sdn authentication from basic(username, password) to token\nauthentication, deprecate username and password and add a new\nparameter token\n\nDepends-on: https://review.opendev.org/814557\nChange-Id: Ie8b347a4705045dfa0363284ec41e9e519ec19f9\n(cherry picked from commit 8ce1ea3d936856354beff4a8df18d05fd5e2c6c7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bfced03d4d5945b2f5e0ef23be42192785e4e166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bfced03d4d5945b2f5e0ef23be42192785e4e166"}]},"branch":"refs/heads/stable/wallaby"},"a346a0ee07991ff02db430cbe976f91b49d3f4f3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-11-18 09:26:22.000000000","uploader":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"ref":"refs/changes/99/817099/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/99/817099/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/99/817099/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/99/817099/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/99/817099/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/99/817099/2"}}},"commit":{"parents":[{"commit":"bd97629f73bca7e8b70cf0f3cc15201902f2a49e","subject":"Add support for logging service plugin configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bd97629f73bca7e8b70cf0f3cc15201902f2a49e"}]}],"author":{"name":"waleedm","email":"waleedm@nvidia.com","date":"2021-10-19 11:02:14.000000000","tz":0},"committer":{"name":"waleed mousa","email":"waleedm@nvidia.com","date":"2021-11-18 09:26:22.000000000","tz":0},"subject":"Change authentication method for Mellanox sdn controller","message":"Change authentication method for Mellanox sdn controller\n\nChange sdn authentication from basic(username, password) to token\nauthentication, deprecate username and password and add a new\nparameter token\n\nDepends-on: https://review.opendev.org/817258\nChange-Id: Ie8b347a4705045dfa0363284ec41e9e519ec19f9\n(cherry picked from commit 8ce1ea3d936856354beff4a8df18d05fd5e2c6c7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a346a0ee07991ff02db430cbe976f91b49d3f4f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a346a0ee07991ff02db430cbe976f91b49d3f4f3"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"},"7764b80391d09275caef958e8f1a627c878b0f5a":{"kind":"REWORK","_number":3,"created":"2021-12-15 06:03:10.000000000","uploader":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"ref":"refs/changes/99/817099/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/99/817099/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/99/817099/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/99/817099/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/99/817099/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/99/817099/3"}}},"commit":{"parents":[{"commit":"bd97629f73bca7e8b70cf0f3cc15201902f2a49e","subject":"Add support for logging service plugin configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bd97629f73bca7e8b70cf0f3cc15201902f2a49e"}]}],"author":{"name":"waleedm","email":"waleedm@nvidia.com","date":"2021-10-19 11:02:14.000000000","tz":0},"committer":{"name":"waleedm","email":"waleedm@nvidia.com","date":"2021-12-15 06:03:08.000000000","tz":0},"subject":"Change authentication method for Mellanox sdn controller","message":"Change authentication method for Mellanox sdn controller\n\nChange sdn authentication from basic(username, password) to token\nauthentication, and as we are in stable branch, we are going to\nbackport only the new parameter \u0027token\u0027\nand avoid backporting deprecation part.\n\nDepends-on: https://review.opendev.org/817258\nChange-Id: Ie8b347a4705045dfa0363284ec41e9e519ec19f9\n(cherry picked from commit 8ce1ea3d936856354beff4a8df18d05fd5e2c6c7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7764b80391d09275caef958e8f1a627c878b0f5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7764b80391d09275caef958e8f1a627c878b0f5a"}]},"branch":"refs/heads/stable/wallaby"},"a58536a9571a282da81bd184a3a87fe3091e1ee5":{"kind":"REWORK","_number":4,"created":"2021-12-19 14:15:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/817099/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/99/817099/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/99/817099/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/99/817099/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/99/817099/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/99/817099/4"}}},"commit":{"parents":[{"commit":"768c15c4ec676a661ac214adacd8ab2defbcc38e","subject":"Add support for the resource_provider_default_hypervisor parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/768c15c4ec676a661ac214adacd8ab2defbcc38e"}]}],"author":{"name":"waleedm","email":"waleedm@nvidia.com","date":"2021-10-19 11:02:14.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-19 14:15:38.000000000","tz":0},"subject":"Change authentication method for Mellanox sdn controller","message":"Change authentication method for Mellanox sdn controller\n\nChange sdn authentication from basic(username, password) to token\nauthentication, and as we are in stable branch, we are going to\nbackport only the new parameter \u0027token\u0027\nand avoid backporting deprecation part.\n\nDepends-on: https://review.opendev.org/817257\nChange-Id: Ie8b347a4705045dfa0363284ec41e9e519ec19f9\n(cherry picked from commit 8ce1ea3d936856354beff4a8df18d05fd5e2c6c7)\n(cherry picked from commit 4f2eda540e02ddeaf2f7b5d8b00c21a356009810)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a58536a9571a282da81bd184a3a87fe3091e1ee5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a58536a9571a282da81bd184a3a87fe3091e1ee5"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
