)]}'
{"id":"openstack%2Fpuppet-neutron~817413","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fussuri~I35c0a76cade37ccbaf7920b30ac162fe0f49e173","project":"openstack/puppet-neutron","branch":"stable/ussuri","topic":"bug/1926693-stable/ussuri","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-11-11 07:08:16.000000000","reason":"Rodolfo Alonso replied on the change"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-11-11 18:50:07.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-11 18:50:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I35c0a76cade37ccbaf7920b30ac162fe0f49e173","subject":"Add support for the resource_provider_default_hypervisor parameter","status":"MERGED","created":"2021-11-10 13:40:57.000000000","updated":"2021-11-11 18:52:16.000000000","submitted":"2021-11-11 18:50:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817413-bug/1926693-stable/ussuri","cherry_pick_of_change":817412,"cherry_pick_of_patch_set":2,"meta_rev_id":"b5b478b8725c80ed3fbdd002c02a4479cae82509","_number":817413,"virtual_id_number":817413,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-11 18:50:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-11-11 07:08:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-11-10 14:16:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-11-11 16:07:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-11-11 16:07:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-10 13:40:57.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-11-10 13:51:32.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-11-10 18:02:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-11 16:07:13.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"3ee5d69718463bcb1970f525c953493f228eea66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-10 13:40:57.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"25333b6fa95385f8dc321cd00c38943c4cd8f847","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-10 14:16:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1be886cc614440724042946189ab10e651ce8040","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-10 14:16:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a3031d940c2e037958504f8b6794bba35e0bd66a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-10 14:18:24.000000000","message":"Topic bug/1926693-stable/wallaby-stable/victoria-stable/ussuri removed","accounts_in_message":[],"_revision_number":2},{"id":"0abf2406b8d37e5784c9cb2d77ed5d5ff3343e6c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-10 14:18:28.000000000","message":"Topic set to bug/1926693-stable/ussuri","accounts_in_message":[],"_revision_number":2},{"id":"c66db8b07eca058a574ef24f1cf6fa93ed4276eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 18:02:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4055506d9b694daaa6cb1e093fb9f837 : SUCCESS in 3m 57s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c553cf1d328b44ed93e8f254fd2755eb : SUCCESS in 3m 43s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b44157999a14772bd868d37b87c6377 : SUCCESS in 3m 54s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/bdc0ed2e150d4e4b9587e13c79b04971 : SUCCESS in 17m 13s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/f28f5080368e42e884db738738ba256e : SUCCESS in 17m 28s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95a615d00d834ef39f44a82ffc6a35c1 : SUCCESS in 16m 37s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfd2655b39194d4f90efb4858986045a : FAILURE in 15m 28s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ec3012603b914d5da55e19c6af7ab3ce : SUCCESS in 1h 11m 51s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/126ed0acb37345028b77e3f42844454c : SUCCESS in 1h 03m 09s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a1dba7b8ff340fb930432f085c8d4e9 : SUCCESS in 54m 58s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a98ebb5ab4874148ba954212a98386b3 : SUCCESS in 37m 43s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/df97ad343f47499c887d56c837f49a90 : SUCCESS in 58m 45s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fcf486db1e29405f879c80e449e871ed : SUCCESS in 1h 47m 54s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/778e61a1a875446485e695f98b4dbc61 : SUCCESS in 1h 04m 34s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/73257cfa596b451c81b2981ab34962d8 : SUCCESS in 50m 26s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/404e7b2312f843afa7141dc1f5504add : SUCCESS in 1h 01m 59s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/05f68856b51e4ce9b058db56cf8164a6 : SUCCESS in 1h 15m 52s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c7321356bee40b1bdc5d93ca7301882 : SUCCESS in 1h 07m 39s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5722ca748e454291bd986d0c531dc87d : SUCCESS in 47m 00s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/bef233ea715c409caf066a02a38ee5be : FAILURE in 26m 37s (non-voting)\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/b93e7178b00d4538a804815f934d63f5 : FAILURE in 26m 52s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb344c974e094a89b61af00f30c7bf16 : FAILURE in 6m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39cda47d0325430f952550181905a108 : SUCCESS in 5m 30s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4d9bc499ca07460f9b4bc5fd500cc845 : SUCCESS in 1h 42m 10s","accounts_in_message":[],"_revision_number":2},{"id":"12fbb855b68e37ed2e840990dd95fd630277b84d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-11 07:08:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d6229f23d9195d5b12ba634919aa50d14c907ba3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-11-11 16:07:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"77c2d5f2984a1125b9a5ba2f4eac062cf4a093ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 16:12:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"21c8758651000cacdf1eab1ec4bdf28650e6593d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 18:50:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d633f4b360f4bc9882e9eba76a78b06 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84672e9f795e486c94877030ca4786a9 : SUCCESS in 5m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3915dd6ed3c4be3b79e54f3dc566b00 : SUCCESS in 4m 22s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/5cd9d29dcbb84fd59680931947597f00 : SUCCESS in 19m 55s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/813ce5fcad6b4ee8a30d6823552ab110 : SUCCESS in 18m 52s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77cf7f91f43c4c838dd486b5eba51b46 : SUCCESS in 14m 10s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d6c1de8ecb2a4fce952dafb01166c9a8 : SUCCESS in 1h 08m 37s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cfb9ad89baaa4f26af880f774b2434ff : SUCCESS in 1h 18m 46s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3a8ab8778a234b0e859412ecd8bbba02 : SUCCESS in 1h 03m 54s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/413732266a7b4ccba84c2c16a89a930b : SUCCESS in 45m 13s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/ed911b11f32140199998e21040f7f7f6 : FAILURE in 23m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98c594a78a3541b8a01dbf34e5ff4d1e : SUCCESS in 6m 12s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/28028a3701ae4fcf858ad83fe249fbaa : SUCCESS in 2h 13m 22s","accounts_in_message":[],"_revision_number":2},{"id":"caa34b0df921b4ea82460179f5b39877f656d479","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 18:50:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b5b478b8725c80ed3fbdd002c02a4479cae82509","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 18:52:16.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9978a8fc0dd340b9b8f5ba291d7e58bc : FAILURE in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"863a7d2253f514e74fe1089fe23a085759d2dc5b","revisions":{"6a1157dcf4fbed82b2f56a8c2a047a570cea98a7":{"kind":"REWORK","_number":1,"created":"2021-11-10 13:40:57.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/13/817413/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/13/817413/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/13/817413/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/13/817413/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/13/817413/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/13/817413/1"}}},"commit":{"parents":[{"commit":"dc15af35cfadbaa025346e250f8975e6e0be491f","subject":"Support arbitrary configurations for ovs agent and sriov agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dc15af35cfadbaa025346e250f8975e6e0be491f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-15 00:49:31.000000000","tz":540},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-11-10 13:40:57.000000000","tz":0},"subject":"Add support for the resource_provider_default_hypervisor parameter","message":"Add support for the resource_provider_default_hypervisor parameter\n\nConflicts:\n    manifests/agents/ml2/ovs.pp\n\nRelated-Bug: #1926693\nDepends-on: https://review.opendev.org/763563\nChange-Id: I35c0a76cade37ccbaf7920b30ac162fe0f49e173\n(cherry picked from commit 38d85c0e7822fca68662371fe11018d7a457a19b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6a1157dcf4fbed82b2f56a8c2a047a570cea98a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6a1157dcf4fbed82b2f56a8c2a047a570cea98a7"}]},"branch":"refs/heads/stable/ussuri"},"863a7d2253f514e74fe1089fe23a085759d2dc5b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-11-10 14:16:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/13/817413/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/13/817413/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/13/817413/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/13/817413/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/13/817413/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/13/817413/2"}}},"commit":{"parents":[{"commit":"dc15af35cfadbaa025346e250f8975e6e0be491f","subject":"Support arbitrary configurations for ovs agent and sriov agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dc15af35cfadbaa025346e250f8975e6e0be491f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-15 00:49:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-10 14:16:01.000000000","tz":0},"subject":"Add support for the resource_provider_default_hypervisor parameter","message":"Add support for the resource_provider_default_hypervisor parameter\n\nConflicts:\n    manifests/agents/ml2/ovs.pp\n\nRelated-Bug: #1926693\nDepends-on: https://review.opendev.org/796853\nChange-Id: I35c0a76cade37ccbaf7920b30ac162fe0f49e173\n(cherry picked from commit 38d85c0e7822fca68662371fe11018d7a457a19b)\n(cherry picked from commit 768c15c4ec676a661ac214adacd8ab2defbcc38e)\n(cherry picked from commit 15a3ccaefcb32e7de53e32d4de9db15fca98d022)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/863a7d2253f514e74fe1089fe23a085759d2dc5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/863a7d2253f514e74fe1089fe23a085759d2dc5b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
