)]}'
{"id":"openstack%2Fpuppet-neutron~817414","triplet_id":"openstack%2Fpuppet-neutron~stable%2Ftrain~I35c0a76cade37ccbaf7920b30ac162fe0f49e173","project":"openstack/puppet-neutron","branch":"stable/train","topic":"bug/1926693-stable/train","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-11-11 07:08:20.000000000","reason":"Rodolfo Alonso replied on the change"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-11-11 18:50:10.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-11 18:50:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I35c0a76cade37ccbaf7920b30ac162fe0f49e173","subject":"Add support for the resource_provider_default_hypervisor parameter","status":"MERGED","created":"2021-11-10 13:41:06.000000000","updated":"2021-11-11 18:52:49.000000000","submitted":"2021-11-11 18:50:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817414-bug/1926693-stable/train","cherry_pick_of_change":817413,"cherry_pick_of_patch_set":2,"meta_rev_id":"b71a7f9206fd270dbfb3594ab4bc8f83b63768f0","_number":817414,"virtual_id_number":817414,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-11 18:50:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-11-11 07:08:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-11-10 14:18:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-11-11 16:07:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-11-11 16:07:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-10 13:41:06.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-11-10 13:51:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-11-10 17:58:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-11 16:07:15.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"6fee0e7418a3c33f38e218c65a8470bb3684014e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-10 13:41:06.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"18b2b08556e09021e8f60310fce9b2e96546ddad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-10 14:17:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e6271b5683630ecb915ab2c88870cc8c7b92ec6a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-10 14:18:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"43893f5add20a5035bae944565587e97914a20ba","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-10 14:18:17.000000000","message":"Topic bug/1926693-stable/wallaby-stable/victoria-stable/ussuri-stable/train removed","accounts_in_message":[],"_revision_number":2},{"id":"e1a923153bed618d1a470550f69c68348e86d149","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-10 14:18:20.000000000","message":"Topic set to bug/1926693-stable/train","accounts_in_message":[],"_revision_number":2},{"id":"00606cbe39530b9b57d07bc19f921b92e1d8c064","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 17:58:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e631cf248844b50bb9b69229c30c19c : SUCCESS in 7m 09s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/001329672130461ea63d5e1a038e01ef : SUCCESS in 5m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d26190b010624a93a81b8c0aa6fead12 : SUCCESS in 6m 13s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3875de0dccbc48aba923a692deb35ff4 : SUCCESS in 26m 20s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/9c6c587541634fcc957113dd6eccdcfa : SUCCESS in 18m 00s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/59d218791b3346a480dec8e6d9d012bf : SUCCESS in 18m 42s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/706ebbd0980b4ff5bd5170deae3c59bf : SUCCESS in 22m 47s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cac12c5661e9489ba063d8a6d9f41836 : FAILURE in 17m 42s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/69b5208a0ec44270bedef87a72aa01c0 : SUCCESS in 53m 39s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4a8bdd4641184a1e9d5232dbbcfff723 : SUCCESS in 1h 03m 05s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/475a5d2e1dba448fbd638f7316c61712 : SUCCESS in 1h 02m 57s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/b6f645fbdd1647cf8d20775fa7f16bdb : SUCCESS in 56m 06s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/aad7cdc3acc94ea49f63675d7cf1592e : SUCCESS in 45m 52s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f971ef17103466fa27f5d05513f9f43 : SUCCESS in 47m 12s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5640bcc655e4346b7bc52174eaf2783 : SUCCESS in 37m 03s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/289dded52f7e4630aadb3cba71f049d5 : SUCCESS in 37m 09s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e66a60c81a1443e4b62312d14b7080d2 : SUCCESS in 56m 48s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/eddfe76464a742f69637ca7fdffe38ec : SUCCESS in 1h 06m 07s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dba0575ea6814aa9bfa887e668b55cd2 : FAILURE in 32m 05s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/954c1eb34b1d439e83a01b58a383bead : SUCCESS in 44m 19s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4717c5e5c93845b8802cee5277edfb60 : SUCCESS in 22m 44s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea6a0c84988a4fc5a70ae43d56cdce7f : FAILURE in 9m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/909cb6c5aeca49e3a6cdb7e1a744af79 : SUCCESS in 5m 21s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3f302d5141444a9d9cf16822a71db0e1 : SUCCESS in 1h 50m 53s","accounts_in_message":[],"_revision_number":2},{"id":"b2067fa60a6031e0ae51c2d3dff6f68f5956e0d4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-11 07:08:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"17b1ced740f9c0f66fda8f2cd802442adb176e13","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-11-11 16:07:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5d939eea34348fa422eb913ac09a5d2bc6e9e6c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 16:12:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5574c0aa7817730e432dc33649d411b20459803a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 18:50:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5404ff77340545c2b8822e30b56d7102 : SUCCESS in 5m 49s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67acdd433db7490b91dea8b2c2b0f716 : SUCCESS in 6m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89b84ec3c9a0472f8fa681b7aa1dfd39 : SUCCESS in 6m 34s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/71dcb094f110497d81969562e5d18ecc : SUCCESS in 35m 17s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15e88edb98da4660a49d9c7825f2593f : SUCCESS in 17m 41s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/bfa45e969f2641778bb69db669c563c2 : SUCCESS in 57m 32s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/45d09331498647ae9434e12bcabade51 : SUCCESS in 1h 06m 04s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/263e49c62a2343cd89492f4d6db9d057 : SUCCESS in 1h 20m 13s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/6e2cd08caebf4d058c1e42da61c69892 : SUCCESS in 49m 58s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4cccf6e9dd664b9fa859621be7af8184 : SUCCESS in 18m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be582892ce2b4884a3c86c769bba0eb0 : SUCCESS in 4m 39s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fe1ea0a46509418c92575fd14490eb18 : SUCCESS in 2h 17m 10s","accounts_in_message":[],"_revision_number":2},{"id":"9bdef10ff0a7209df9d26bda410bc783a8ff4b40","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 18:50:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b71a7f9206fd270dbfb3594ab4bc8f83b63768f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 18:52:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8dfad2ff502f4b449d931e23bf735dc8 : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"db82fc60d8bb6f28c00207b64db3da80f375d58c","revisions":{"8ab1ccc54857fb8497d9be8bcdf7ca3b2af092cf":{"kind":"REWORK","_number":1,"created":"2021-11-10 13:41:06.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/14/817414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/14/817414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/14/817414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/14/817414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/14/817414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/14/817414/1"}}},"commit":{"parents":[{"commit":"eb509c49e33873eebc6c24410af2c203d9255f87","subject":"Merge \"Support arbitrary configurations for ovs agent and sriov agent\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/eb509c49e33873eebc6c24410af2c203d9255f87"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-15 00:49:31.000000000","tz":540},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-11-10 13:41:06.000000000","tz":0},"subject":"Add support for the resource_provider_default_hypervisor parameter","message":"Add support for the resource_provider_default_hypervisor parameter\n\nConflicts:\n    manifests/agents/ml2/ovs.pp\n\nRelated-Bug: #1926693\nDepends-on: https://review.opendev.org/763563\nChange-Id: I35c0a76cade37ccbaf7920b30ac162fe0f49e173\n(cherry picked from commit 38d85c0e7822fca68662371fe11018d7a457a19b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/8ab1ccc54857fb8497d9be8bcdf7ca3b2af092cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/8ab1ccc54857fb8497d9be8bcdf7ca3b2af092cf"}]},"branch":"refs/heads/stable/train"},"db82fc60d8bb6f28c00207b64db3da80f375d58c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-11-10 14:17:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/817414/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/14/817414/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/14/817414/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/14/817414/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/14/817414/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/14/817414/2"}}},"commit":{"parents":[{"commit":"eb509c49e33873eebc6c24410af2c203d9255f87","subject":"Merge \"Support arbitrary configurations for ovs agent and sriov agent\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/eb509c49e33873eebc6c24410af2c203d9255f87"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-15 00:49:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-10 14:17:06.000000000","tz":0},"subject":"Add support for the resource_provider_default_hypervisor parameter","message":"Add support for the resource_provider_default_hypervisor parameter\n\nConflicts:\n    manifests/agents/ml2/ovs.pp\n\nRelated-Bug: #1926693\nDepends-on: https://review.opendev.org/796836\nChange-Id: I35c0a76cade37ccbaf7920b30ac162fe0f49e173\n(cherry picked from commit 38d85c0e7822fca68662371fe11018d7a457a19b)\n(cherry picked from commit 768c15c4ec676a661ac214adacd8ab2defbcc38e)\n(cherry picked from commit 15a3ccaefcb32e7de53e32d4de9db15fca98d022)\n(cherry picked from commit 863a7d2253f514e74fe1089fe23a085759d2dc5b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/db82fc60d8bb6f28c00207b64db3da80f375d58c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/db82fc60d8bb6f28c00207b64db3da80f375d58c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
