)]}'
{"id":"openstack%2Fpuppet-neutron~819228","triplet_id":"openstack%2Fpuppet-neutron~master~I01404d4341c19294a3de230323ca3ddfd7a72ec2","project":"openstack/puppet-neutron","branch":"master","topic":"coordination","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-09 00:29:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I01404d4341c19294a3de230323ca3ddfd7a72ec2","subject":"Use oslo::coordination to manage tooz backend package","status":"MERGED","created":"2021-11-25 01:54:41.000000000","updated":"2021-12-09 00:29:44.000000000","submitted":"2021-12-09 00:29:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819228-coordination","meta_rev_id":"d41b6fe4b459d64fbc42f8bf934f782e6853dd84","_number":819228,"virtual_id_number":819228,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-09 00:29:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-11-29 08:17:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-12-08 21:16:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-12-08 21:16:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-25 07:11:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-26 07:35:04.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-12-08 21:16:45.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"3e382f6a6e7418863942378bd692f9c06b8e1b28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-25 01:54:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a21afcf34a1e000501a2ea31f4a7298d488a04c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-25 04:30:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"be6b25aeb30761e278c3e0702d92c9c94fd10337","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 07:11:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e0ab59e49f94461936bfc39d1db5f0f : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e93f3ceb01e4a90936afa07410c57e2 : SUCCESS in 3m 09s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2fc7b2393a74754878b6ee63ceaa1a4 : SUCCESS in 3m 11s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e664e8609b9416b99d295de38817098 : SUCCESS in 15m 13s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/e118620add4d43b9bd9432efb562234d : SUCCESS in 20m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b404ceec73d4436d976f297cda5b8a96 : SUCCESS in 13m 16s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c602d38b8444e9aa09295a5f8ddb81e : SUCCESS in 14m 09s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aae0b722e24944b2a1bdcfd99ba91e6c : SUCCESS in 13m 27s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23b3b9149f88471e865116a32a2a73d1 : FAILURE in 34m 56s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95ea848e710943aeb144075a90db8944 : FAILURE in 1h 06m 18s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52d2e87b664d42e198aaadbcd83f0493 : FAILURE in 1h 06m 30s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1830e63213f943edb0a55b42e071c4a7 : FAILURE in 22m 54s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/197dde43439e43798de641044ac66c8a : SUCCESS in 27m 47s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/da29a9fcbaa842f380aa43067aa1a88a : SUCCESS in 1h 13m 53s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b741433e5e9949fdb0f7f23bfcfbce7d : SUCCESS in 1h 12m 11s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fd34937f1bf64f288ed2b521f1bbd3ab : SUCCESS in 1h 32m 50s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc1e2892a1e247dfb021ed071458bb18 : SUCCESS in 54m 36s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6e2a6a1a1704e3e854cd2c01ffc9a42 : SUCCESS in 14m 54s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3d77a4c84e254706bab2b25d590bb9fa : SUCCESS in 21m 16s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ec9db5b10824f73b38cc161b591e9d5 : FAILURE in 15m 14s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e491757609d41508de1e81c12ee7011 : FAILURE in 12m 32s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5514add0f5664c47abaaa1cd482f496c : SUCCESS in 2h 38m 18s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/141ab99b95cb4e6d8730ec8c6fdf2af5 : SUCCESS in 43m 36s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9f175cc5fc6443f8ac3e5203f2149088 : SUCCESS in 1h 12m 43s","accounts_in_message":[],"_revision_number":2},{"id":"82b14a889cb7007ce365286823433cb5217e4320","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-26 07:35:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a18f7888e727a4a4f87c57a51f9dd7a874b9cf25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 01:34:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"56bb3073cb0debaaf3346303759ec2a9404174ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 04:19:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dff7c525c5ba42acae980335e111870c : SUCCESS in 3m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf0cdcbefdbb4edba5e50e5dcb6ae61e : SUCCESS in 3m 20s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc6814e556f84533817f1c55833fd88e : SUCCESS in 3m 58s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0be4dddf3d4449f2a6798c6616863a42 : FAILURE in 18m 43s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/eb5e812be47343d9adbfd429d0ba439b : FAILURE in 14m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5484ba328b3041e6b5c4b142d2733ed7 : FAILURE in 16m 21s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4baf24dd0bf45d0adbbf7395e5016af : FAILURE in 13m 09s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a13058a3e3a4d14bafb927f0d8b4f6c : FAILURE in 11m 59s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e41c1d516b049bfb2212a3c3cfc85a5 : FAILURE in 1h 05m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2b545dc85e046749e20500b905a1344 : FAILURE in 1h 21m 40s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a9f9bbbfae9428b9c8b5cca94bf6084 : FAILURE in 7m 41s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/158a4872666e41b9be270731b0a6afc3 : FAILURE in 25m 32s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/797d98608aca41f19794606ac35f5b5f : SUCCESS in 35m 24s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/95250578078543929a5415d1ce90c05d : SUCCESS in 1h 10m 40s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1125e8abaa34405f9ab788757fd60bad : SUCCESS in 1h 26m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dcd2e246cd1e4bb39f010cb336dd81d1 : FAILURE in 12m 03s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ded84c173cba4ec8bdaa5fc35a08745c : SUCCESS in 55m 43s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b05703095834a9aa908149f93585fbc : SUCCESS in 19m 33s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3819ed1760bd4c0c9e7fcc9edadd7ce4 : SUCCESS in 55m 03s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f02bdf3997904b88b1a6a7c651716c19 : SUCCESS in 56m 51s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e0015fa149e46a5864232287201d7fe : FAILURE in 10m 46s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f094870735e40ef8ea2aed367c7bc27 : SUCCESS in 44m 28s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5c678710d514dcd855a716158479613 : FAILURE in 18m 29s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/6006efca0ee649f0a3a80738022e0332 : SUCCESS in 21m 44s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b66305d7ce694cf6a957a8edf8b1ce5c : SUCCESS in 15m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88aeb25d9aec434d87ae493286f5fd27 : FAILURE in 13m 52s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2f6c1243bb68476a8a8a44e553c08791 : SUCCESS in 2h 41m 59s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/27241d43a0b543d7ac3e794d56c28913 : SUCCESS in 52m 17s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/51abdf39337d4ea983d980832dd53e70 : SUCCESS in 1h 17m 08s","accounts_in_message":[],"_revision_number":3},{"id":"384e308652ba84b681efc60c9e40d6a1c6c60ca6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 04:21:28.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"d03b72d66f18423e1ecc1e018ef50a983b762e78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 06:44:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5856531c77784f32914727a1c993b6ad : SUCCESS in 3m 51s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/088d65a11c6f4af488ad9879f885053b : SUCCESS in 3m 58s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9da3d0c990ca47f7a82041bd9505a399 : SUCCESS in 3m 59s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/68b28e366b2a4e7ea0e8e276902a0dbe : SUCCESS in 17m 17s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/4dc29c6f2b704ebdae80af112b0ee1dd : SUCCESS in 17m 17s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8652078ba7440309bef223b49d10f82 : SUCCESS in 20m 24s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9bd3337da004300b73ba24305c873e2 : SUCCESS in 13m 44s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05d5598d12e24f99bc5a2f6d514b2a5c : SUCCESS in 12m 22s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dad27d1f2a2c4bac829536f15f05b04b : FAILURE in 1h 00m 09s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/090b4cd36f274809a35a9d26a9261705 : FAILURE in 39m 39s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de07005333a04cb2980990d04b5955cd : FAILURE in 27m 17s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7fe39cbec7346f49d106c7cd7072e20 : FAILURE in 29m 27s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b6f0eb779fe8403b8e12e1ef4015a7b4 : SUCCESS in 19m 26s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/28cc9a19ec104c5f989def1f257f6afb : SUCCESS in 1h 18m 24s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c9d43f4edca943da8cbda0d767b4dbce : SUCCESS in 1h 57m 40s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/931f0396c6b34df88cc70e92874e92e1 : SUCCESS in 1h 43m 58s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6f34e563a4b643cfbf0640719e6a36ac : SUCCESS in 58m 15s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c3373f54ad04cb881e6ce6fb0c1e797 : SUCCESS in 18m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4fe8360c19a4456a89f612ee53aa2fdf : SUCCESS in 1h 00m 21s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/041814c22c33424a8d17e0132842ca55 : SUCCESS in 50m 50s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e93093b17aa9459a8502ec876844fe4f : SUCCESS in 1h 12m 01s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab7d6c0887884532b9fad4b3d6efeadf : SUCCESS in 52m 44s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71cee54988484d73807e37909ca7ac0d : SUCCESS in 26m 26s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/4b09996720484c0f84dd722c214193f2 : SUCCESS in 34m 35s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/096de3c11c2e4bc59f3e0aebcf455a79 : FAILURE in 16m 32s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e976960d246b4374a04747d55f5824d5 : FAILURE in 16m 39s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8471401c95ae4a1c8c01085fa8fe2c2d : SUCCESS in 2h 19m 53s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/fd5238185f414c878f87530c4609c9a2 : SUCCESS in 38m 14s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0bfd9210fd8945b4bbef852b65e6fa9e : SUCCESS in 58m 26s","accounts_in_message":[],"_revision_number":4},{"id":"812f1a452597a9964a8b0099b57102aeca9b33cc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-29 08:17:30.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5f33231ebf5010f46880ac832459e496384ec8e0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-08 21:16:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8a7ca3b1f925d5620504fd46a66d7e62b02984a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 21:21:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5379a1657e96ffb21a925af402392f0f063a44e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 00:29:44.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69ed2e9c119f42d98ccbc482192f3d24 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7703ea06652443689d2d622c2c60e77d : SUCCESS in 3m 43s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7849dbb784343d7b5348807b29934dd : SUCCESS in 3m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/058d6f950fbc49469afd95b6f063429f : SUCCESS in 15m 31s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/cf8b07fe72ba4c5fa460f78b52589bf4 : SUCCESS in 15m 06s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c85f4b9b1d14a418fdf5cf487ffa891 : SUCCESS in 13m 29s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d1dc6d76fb84c96a19e4868460bd785 : SUCCESS in 15m 40s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d5afa3e7da834188bee4708bfbb7cb62 : SUCCESS in 57m 15s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/259fe85284c249b181113f04fae3d8fc : SUCCESS in 1h 09m 19s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4ffc3836f1440beb614de81257a2802 : SUCCESS in 1h 25m 39s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7061d9717e8d49feb0d821d5e2b6bb5e : SUCCESS in 1h 13m 37s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/367a5d1d4ae2449abe1fa6faa2aac5e1 : SUCCESS in 23m 16s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4baf5029b1344539f37e15b74d08ccd : SUCCESS in 18m 36s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8701a3421e72420f89012d0ee1784858 : SUCCESS in 2h 52m 14s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4543ac45fa3b42eb891f5fd0ca94a9b1 : SUCCESS in 50m 04s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2b5154fb11344e31a079e48662cc2e3f : SUCCESS in 1h 15m 34s","accounts_in_message":[],"_revision_number":4},{"id":"d41b6fe4b459d64fbc42f8bf934f782e6853dd84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 00:29:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"52935b3e69670edf0a704f83789cc38327e54a25","revisions":{"0a0031d6a351536d2207f04c525e4530cd19054a":{"kind":"REWORK","_number":1,"created":"2021-11-25 01:54:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/819228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/28/819228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/28/819228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/28/819228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/28/819228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/28/819228/1"}}},"commit":{"parents":[{"commit":"47ee09958717f262638a8c0ea533f2b5e766144d","subject":"Remove unused neutron::params::haproxy_package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/47ee09958717f262638a8c0ea533f2b5e766144d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-25 01:54:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-25 01:54:15.000000000","tz":540},"subject":"Use oslo::coordination to manage tooz backend package","message":"Use oslo::coordination to manage tooz backend package\n\nChange-Id: I01404d4341c19294a3de230323ca3ddfd7a72ec2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0a0031d6a351536d2207f04c525e4530cd19054a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0a0031d6a351536d2207f04c525e4530cd19054a"}]},"branch":"refs/heads/master"},"3f08f8cf3986708535e207b3976eda8b65064522":{"kind":"REWORK","_number":2,"created":"2021-11-25 04:30:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/819228/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/28/819228/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/28/819228/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/28/819228/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/28/819228/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/28/819228/2"}}},"commit":{"parents":[{"commit":"47ee09958717f262638a8c0ea533f2b5e766144d","subject":"Remove unused neutron::params::haproxy_package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/47ee09958717f262638a8c0ea533f2b5e766144d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-25 01:54:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-25 04:30:07.000000000","tz":540},"subject":"Use oslo::coordination to manage tooz backend package","message":"Use oslo::coordination to manage tooz backend package\n\nChange-Id: I01404d4341c19294a3de230323ca3ddfd7a72ec2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3f08f8cf3986708535e207b3976eda8b65064522"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3f08f8cf3986708535e207b3976eda8b65064522"}]},"branch":"refs/heads/master"},"14751dface4dce2f2abab7d089c225586d17c854":{"kind":"REWORK","_number":3,"created":"2021-11-29 01:34:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/819228/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/28/819228/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/28/819228/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/28/819228/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/28/819228/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/28/819228/3"}}},"commit":{"parents":[{"commit":"60ba5c80298c578493d0d17815d3560b172ad2ef","subject":"Install cache backend packages before starting services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/60ba5c80298c578493d0d17815d3560b172ad2ef"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-25 01:54:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 01:34:26.000000000","tz":540},"subject":"Use oslo::coordination to manage tooz backend package","message":"Use oslo::coordination to manage tooz backend package\n\nChange-Id: I01404d4341c19294a3de230323ca3ddfd7a72ec2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/14751dface4dce2f2abab7d089c225586d17c854"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/14751dface4dce2f2abab7d089c225586d17c854"}]},"branch":"refs/heads/master"},"52935b3e69670edf0a704f83789cc38327e54a25":{"kind":"REWORK","_number":4,"created":"2021-11-29 04:21:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/819228/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/28/819228/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/28/819228/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/28/819228/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/28/819228/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/28/819228/4"}}},"commit":{"parents":[{"commit":"60ba5c80298c578493d0d17815d3560b172ad2ef","subject":"Install cache backend packages before starting services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/60ba5c80298c578493d0d17815d3560b172ad2ef"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-25 01:54:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 04:21:28.000000000","tz":0},"subject":"Use oslo::coordination to manage tooz backend package","message":"Use oslo::coordination to manage tooz backend package\n\nChange-Id: I01404d4341c19294a3de230323ca3ddfd7a72ec2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/52935b3e69670edf0a704f83789cc38327e54a25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/52935b3e69670edf0a704f83789cc38327e54a25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
