)]}'
{"id":"openstack%2Fpuppet-neutron~819552","triplet_id":"openstack%2Fpuppet-neutron~master~Ib38db6f75006adb86937db20a990eb430ff36391","project":"openstack/puppet-neutron","branch":"master","topic":"cache-deps","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-08 01:39:15.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"Ib38db6f75006adb86937db20a990eb430ff36391","subject":"Install cache backend packages before starting services","status":"MERGED","created":"2021-11-29 01:32:44.000000000","updated":"2021-12-08 06:45:03.000000000","submitted":"2021-12-08 06:45:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819552-cache-deps","meta_rev_id":"8aef02b92b49a9a65097c19275886186cd71b532","_number":819552,"virtual_id_number":819552,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-08 06:45:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-11-29 08:17:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-12-07 15:57:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-12-07 15:57:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-29 04:09:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-29 08:17:15.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-12-07 15:57:14.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"ebc36d8f719fcd66e234c4c7ca077cdd6151fa03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 01:32:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f261d277fea6a7dced0c0ae1eea386f99d691a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 04:09:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd554d5e997940b1879a678656372e65 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e14f4e29eaaa49d69d1d6f973b8539be : SUCCESS in 5m 47s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b22763b3abb4424f8ba2022f02c9e06b : SUCCESS in 3m 15s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/02742dfb368c4953beb846eab9ef12d1 : SUCCESS in 15m 33s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/6f896aecbbfb420c9b3e4ae00e6f51a2 : SUCCESS in 15m 17s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c8b95883e5545a78bbadee12c5dbd93 : SUCCESS in 12m 21s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c6722930e3f4ee2b66e308d236c48d4 : SUCCESS in 12m 14s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/096932c673954f7a912c205455002389 : SUCCESS in 13m 03s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/557865f7b0894283a57c0819e64c8a79 : FAILURE in 37m 41s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e04f88b26ef346a8b8c2d9370a9d7a27 : FAILURE in 1h 07m 04s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24a971b50d844a55b5f810d12baed18a : FAILURE in 1h 21m 07s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b08b6c936d74687b8bc027e4cd3480c : FAILURE in 24m 13s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8edd1d5ec1844aadabd83523c6edc1ef : SUCCESS in 19m 51s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/17966aa0953b412f92424b065f2d4204 : SUCCESS in 1h 00m 39s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec26a3f9735b4072b30048ef5a6c7144 : SUCCESS in 1h 19m 57s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/368d979ded64486a8a22d928ee6a31c9 : SUCCESS in 1h 32m 05s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2e9474f1b7e4f13bac215babb087e5b : SUCCESS in 1h 27m 34s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/076b107b16754a0da229a219b4f5a7c4 : SUCCESS in 20m 11s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7b555ebaef649aea9b6e8ea3baa9797 : FAILURE in 1h 24m 22s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/373796acf1fe4f43b6272d16f38c3f0a : SUCCESS in 54m 50s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cfa43723a42a4525920ccad070fcc95b : SUCCESS in 1h 13m 08s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c7df12e07a54b028dbc4e12e988cf43 : SUCCESS in 51m 21s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0adc29ca1dd04cebb016a9c71427c067 : FAILURE in 18m 02s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/6d053625bf8d4dd18c4c9c9ed3a2403f : SUCCESS in 21m 02s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1257001553254f3a83a2e1dca3618573 : FAILURE in 17m 31s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4502fe3e6ed4b59a4b70b7d69757475 : FAILURE in 17m 56s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8e00272ac9e747a9afcfae16d6e2bc0c : SUCCESS in 2h 34m 43s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1e70ca0b42d64e369bbacb5c2c8ceb28 : SUCCESS in 47m 35s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ccaa5f6e19644b3e862d4308e98a6c10 : SUCCESS in 1h 10m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7ba44854d9a9b182ca7e6064c2e3d32ceaa4dd55","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-29 08:17:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cd15376908334036c92bdee9e605f89ecd0bf482","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-07 15:57:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"34264f21cdd7b916eb6c6ff20209c86a04053428","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 23:08:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"adc3a05e3e2f7e1f2f0632e878814b380f67288e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 01:37:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b1c9e2ee25543f3941357c4e27f9fd4 : SUCCESS in 3m 32s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd4ad9dc18a6457ea3b572456ee5f07a : SUCCESS in 3m 56s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a07cf847cc34478d95a1150461e652e9 : SUCCESS in 4m 11s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d7489cf835c64965a3a5e0387befb3f4 : SUCCESS in 15m 13s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f5521b74e9444e31b04cb61499f40c1c : SUCCESS in 16m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8039c4f0e9b426d9ab73208f7cc7855 : SUCCESS in 15m 04s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36d3793104fb4a9ba75695df7ed7d7f3 : SUCCESS in 14m 55s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50b9697dc68f474cb74105ac9bbfdcef : SUCCESS in 14m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8501c77793564452b647e4cf07c9f2b5 : FAILURE in 33m 28s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/795fe127bac0480e9857ea909dec4f28 : FAILURE in 1h 23m 40s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/714e20c60e104f26ba38ac6a95bad44e : FAILURE in 1h 03m 35s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86bbaad558c7479faf5148f06651a048 : FAILURE in 55m 09s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9ed81dd71a74dcda585502b0880411a : SUCCESS in 22m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7c8d64cdcb4470ca9dce76ee8c61c8f : SUCCESS in 1h 20m 47s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/71d28209527c4a9eba2f0e3a1b9c41a6 : SUCCESS in 1h 31m 32s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/604a971b1e88446a8487f5a289fca9a1 : FAILURE in 57m 53s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fcda7fcc95a44ff2933bd8ea74c4be5b : SUCCESS in 1h 00m 27s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ede78702def4afb9ed485e8a4b2f6df : SUCCESS in 19m 49s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf0d96c579a04826b96833a405b52154 : SUCCESS in 1h 26m 19s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/36f52eb9884e4764a04768efcdc82d89 : SUCCESS in 1h 07m 07s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c32c21e77d794275b29365e9061ea5ca : SUCCESS in 57m 40s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aaa2b2b6e00a4f6b915177316e49a13d : SUCCESS in 51m 42s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd98d8b4ba6243e28070678cb9ed4d64 : SUCCESS in 18m 54s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f4514ae745fd413aa1b139ff5dda0732 : SUCCESS in 20m 11s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74965cb2136748eeae2e95d4d34cb3a4 : FAILURE in 16m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfd2eaed1c7d45439a7fc4c2d8b5001a : SUCCESS in 17m 21s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1be2e4bea1dd455eb2bd8a1cd8110397 : SUCCESS in 2h 17m 16s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/79fb9e931634463b83923b94e3bf87de : SUCCESS in 36m 38s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9820f857588446a59fd9bb1ae63c9ea9 : SUCCESS in 58m 39s","accounts_in_message":[],"_revision_number":1},{"id":"00269e87be508e1aa1f038f8b9504fdfd0bdb63b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-08 01:39:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f5131751f88445c7f5224a532e4af41708abc9f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-08 03:39:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ceae5ff1c0c08d018a02e425462e9d1136461ca1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 04:29:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d487c57e16d8427fa351dc62290a9b47 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/362e59f1358c4a969afdf201002acd9e : SUCCESS in 3m 47s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f216842b5a4e457c834f01a9b892e1e9 : SUCCESS in 3m 17s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b07c1e8a4f0f48f09eb31c46965a7bdc : SUCCESS in 19m 56s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca2bfb5e3df443a0958a8a8a1868b987 : SUCCESS in 16m 38s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/605498b4f8d24bf2a023107f54cf109a : SUCCESS in 13m 57s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddf7d2bd7c1c462bbd5ad3b452b89e4b : SUCCESS in 13m 28s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e23f1bbc4c294741a197130172ea1f62 : SUCCESS in 14m 13s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ad2add8175f48cdac8e6ac1334bd1b8 : FAILURE in 32m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d315fd6826f74b52aa851758ca71ebe6 : FAILURE in 37m 21s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2e7adad2921487cafe87203b5aa3243 : FAILURE in 59m 11s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4187c1a1dc1b43659d61e33bd62cc860 : FAILURE in 21m 45s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d8dccb53d314ecda7ef621751ef7c2c : SUCCESS in 20m 51s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/aed958b5cf1e422cbcb675ddd2024fb9 : SUCCESS in 1h 17m 58s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/00f502eb88f54c8bb961ca7f30499a78 : SUCCESS in 1h 20m 14s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/687a7a8d98fa4363b6db86b9b73b9cb4 : SUCCESS in 1h 20m 55s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc09c7312a964195abeac6a2c3fe5524 : SUCCESS in 57m 04s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d7c03cb9f7e48ac9f093bb05311962f : SUCCESS in 16m 38s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7cc3848c6cea485ea2b2c83784e3cb5e : SUCCESS in 47m 39s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0eb0faa45b1449eba303496c61345908 : SUCCESS in 56m 58s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e32782e669814990bd7849c75c723af3 : SUCCESS in 52m 54s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/707fa787a2cf4d2c9534affa90cc1d29 : SUCCESS in 39m 16s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ea35091538743a2b1cd673af534aa9c : SUCCESS in 15m 12s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/36a0753958ac4259a6727a56a5d90aef : SUCCESS in 20m 22s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de7455f03a73477d8e0e71901d1c2b4d : FAILURE in 18m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0086638dcba4f9b9513294e5893e380 : FAILURE in 17m 18s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9ab69a9c697b4c2f88f7c3d9b6a80766 : SUCCESS in 2h 46m 13s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c5c3736933624626975e36dfe879e6c9 : SUCCESS in 1h 00m 41s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4282add3639c434ab9da1de40903baca : SUCCESS in 1h 22m 14s","accounts_in_message":[],"_revision_number":1},{"id":"681f13343e98bc831296a7a9390946b8811224ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 04:30:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1248cc8ce399edea0d99f5851df778b733e467c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 06:45:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e213e3f18a44e3f8a7e4ee532e20fcf : SUCCESS in 4m 16s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4256513b86484834b4b07bd5dc66197a : SUCCESS in 4m 21s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a911ee3a8c84a7896c384147a2a7eb6 : SUCCESS in 3m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/94d130c3ccb64e2cb9f8f5d2ca33260a : SUCCESS in 16m 55s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/6e80b70b5a4f465987c1247a02c5bd03 : SUCCESS in 18m 18s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd419e8aeb1e4a74bb46f552db2626d2 : SUCCESS in 14m 23s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b674db6dff944f5bade0f57f8d76a8f : SUCCESS in 14m 20s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c477699b74c14d14871f5d5ed353f74c : SUCCESS in 1h 05m 00s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/65e31e5e11314b5ca9b3a29ba9e65881 : SUCCESS in 1h 33m 46s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa51747fab444f778706abb4142534dd : SUCCESS in 1h 13m 50s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ea2993e7556c47f59eff0bfe8377fa52 : SUCCESS in 1h 05m 16s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3f1d51bd35264f6ea89bc634fbd54c2e : SUCCESS in 26m 21s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/388690b62dca47d5b2a68be6587d8be8 : FAILURE in 16m 56s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3bf1e5e7a9fc414391e7c773c4623796 : SUCCESS in 2h 13m 05s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/702adacf7eb449b4a3720390b10d22c0 : SUCCESS in 36m 24s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9a49a9a5898a475888a4b7de228610b9 : SUCCESS in 57m 08s","accounts_in_message":[],"_revision_number":1},{"id":"8aef02b92b49a9a65097c19275886186cd71b532","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 06:45:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"60ba5c80298c578493d0d17815d3560b172ad2ef","revisions":{"60ba5c80298c578493d0d17815d3560b172ad2ef":{"kind":"REWORK","_number":1,"created":"2021-11-29 01:32:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/819552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/52/819552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/819552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/819552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/819552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/52/819552/1"}}},"commit":{"parents":[{"commit":"47ee09958717f262638a8c0ea533f2b5e766144d","subject":"Remove unused neutron::params::haproxy_package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/47ee09958717f262638a8c0ea533f2b5e766144d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 01:32:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 01:32:26.000000000","tz":540},"subject":"Install cache backend packages before starting services","message":"Install cache backend packages before starting services\n\nChange-Id: Ib38db6f75006adb86937db20a990eb430ff36391\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/60ba5c80298c578493d0d17815d3560b172ad2ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/60ba5c80298c578493d0d17815d3560b172ad2ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
