)]}'
{"id":"openstack%2Fpuppet-neutron~820650","triplet_id":"openstack%2Fpuppet-neutron~master~I0c3f4d3a80853f1f07325ee31317e415e922f128","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-13 14:31:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0c3f4d3a80853f1f07325ee31317e415e922f128","subject":"Add support for the resource_provider_inventory_defaults parameter","status":"ABANDONED","created":"2021-12-07 04:58:53.000000000","updated":"2022-03-13 14:31:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d31afc0e30e5666d62f8810944603f9e201de62e","_number":820650,"virtual_id_number":820650,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-07 19:45:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-07 07:28:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd515b24533bc5bb8da243fb5651a4d328486918","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 04:58:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d40f9f480ab32bae17ebca86dee62b6274be207","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 05:00:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e000a4fe6ce63fbb3ead1bb2a4da7a83c4f014f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 07:28:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c19d0ae365d340b59afb371623ec8680 : SUCCESS in 3m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9614e751d73047a1a6dc724f3e2b4d4b : FAILURE in 2m 55s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bd7def79b5944b9b6a8b89b94ccb9ee : FAILURE in 2m 48s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ed5fec5cc0047d6a74592e3c02665c2 : FAILURE in 13m 55s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c1f122acbc2d4cafa90b302419027d81 : FAILURE in 13m 37s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f674fe4789754f7eb21a219e6d92e20b : FAILURE in 11m 09s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9303685c1c7a4b6fb911ee781f30a032 : FAILURE in 11m 26s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4aedf7a7fa7d446b93de222dbe269ba8 : FAILURE in 11m 38s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94b6c3eca8504862b4efb84d01c08999 : FAILURE in 13m 32s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4169fe3f26994eaebafc8191ff9f213e : FAILURE in 10m 12s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07e9222425c94b6ab410041a8c54d5fb : FAILURE in 29m 38s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3f14fc2df76463bae3f931b18c75e6a : FAILURE in 8m 17s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/de93eda5f27d46e0907bf4eb044a5d85 : SUCCESS in 20m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/630466fe0f7248c69a6242aec31d01dd : FAILURE in 11m 50s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e096c559d4634b78882c108c8e14fe80 : FAILURE in 11m 04s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b7642fb4311495f868ef3ec109127a7 : FAILURE in 12m 08s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/86a338559a6a4e6ea2a24da2a7b5734b : FAILURE in 10m 44s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5df7cb960954071bf77aa649e7b7c0e : FAILURE in 26m 14s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe91c9aba855429ab7695b7784541d93 : FAILURE in 9m 26s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/06a2b2f3cdae4a3793c1ba075d37e538 : FAILURE in 9m 56s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a96a6d0857a54f1cb6c4190559a4fc7b : FAILURE in 10m 06s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/628cdcfe002747129c51da8715d8bf61 : FAILURE in 10m 09s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20a5d1c8c4f54961abdfa4134074965d : FAILURE in 8m 13s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2b3a275d9c0c4a15bc7ec3a7c3826fbd : FAILURE in 11m 04s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/525daaa1bc1c45da8a82c5a0ad74f03a : FAILURE in 9m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b81cbf595c834ab3a27463c36c9ec819 : FAILURE in 7m 41s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/23102a729d5840c0a56d3baded1a8851 : SUCCESS in 2h 23m 59s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/05ecc95845444edbaade4e67959a6c88 : FAILURE in 37m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/800bd926c1714055a99136192c617fc0 : SUCCESS in 3m 15s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d431755d6a9849c38877e95e8c2ac85f : SUCCESS in 1h 07m 29s","accounts_in_message":[],"_revision_number":2},{"id":"bc41799535cf95bb497b7f44ab5e86f2c7f7c823","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 09:48:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1a659d0c742ee027e4665916d421258fefe0c492","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 11:03:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5aded8f56de4cc390c3bfd4105f0ded9d503ac1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 12:46:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d3fc6e9097086518cf0fb1b4e69644e6f255d196","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 19:45:27.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05db164e8b8848f19ce70fd8ce719605 : SUCCESS in 4m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31dad7f4ce514ec489726840434ec5c7 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ae02ec43a5543068e41dbd42deee622 : SUCCESS in 5m 05s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c0c2c26d46c744cebb466f049e6c1a57 : FAILURE in 18m 57s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/89c6c64fc2ac4290b5002f871c6c5e04 : FAILURE in 17m 22s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e93312061f5447f9c93ff9354e0a4a4 : FAILURE in 18m 23s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3fd6244885c450491337dcd79660e57 : FAILURE in 18m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae61544ea1b74ca394c2ee4b40e2843b : FAILURE in 17m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/138bb309c5884e57b8c4b8eb6a406c18 : FAILURE in 40m 40s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c2a3ee2a5b54be7b0d875a2fbd74510 : FAILURE in 53m 37s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb8a9a3dcab34b328ecc5dc2e2c5ad8e : FAILURE in 58m 11s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8110209242454743aecde3a328218f80 : FAILURE in 37m 22s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5d812bb88e6f4407b8162a3a333ab448 : SUCCESS in 23m 29s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0835cbdb8ff049db9c8b247fd0178097 : TIMED_OUT in 2h 05m 12s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9253a744af2d43d48542841f42edfe20 : FAILURE in 1h 26m 11s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3050c49d499b48afa16323bfd4a0c761 : FAILURE in 1h 36m 00s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d14aed0aaa8f4c619496579a89cb1e60 : FAILURE in 1h 04m 12s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2fcac01b66a5474fbc9e407d92224b6e : SUCCESS in 18m 45s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c4298d48b104d73af1866d2cd666f04 : FAILURE in 56m 54s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ef36ecd94d547bdb9f2b6741edb3240 : FAILURE in 1h 02m 49s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c0322793f434f10b39db93601084adc : FAILURE in 55m 22s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/10c388750ce1462b8ad351582a65954c : FAILURE in 36m 31s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9e2fc3f2aa54692b305925ffb5eca60 : SUCCESS in 19m 40s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/fcb4d274752c422eadb4621e941c9344 : SUCCESS in 27m 17s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af5a2e03b0ca45f8b142bd8279d906d3 : SUCCESS in 18m 48s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edf828720ae44e41bfe8a6b6b7b2c145 : SUCCESS in 17m 48s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/32bff861b38047319c324592bbda1e7e : SUCCESS in 3h 02m 27s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/7ca5e0bee8164dd4a02eb69bd366d6d0 : SUCCESS in 55m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cc9ac39d2b749329f82eccb8a85114e : SUCCESS in 6m 02s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1a4677f7c5c945e2ab906f654334f17b : SUCCESS in 1h 14m 00s","accounts_in_message":[],"_revision_number":5},{"id":"d31afc0e30e5666d62f8810944603f9e201de62e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-13 14:31:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"680c36b5fe60a0b2b78d8705e568241009f9e0e2","revisions":{"700538820f3966d4679ac846fe3bca17b2e94864":{"kind":"REWORK","_number":1,"created":"2021-12-07 04:58:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/820650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/50/820650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/50/820650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/50/820650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/50/820650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/50/820650/1"}}},"commit":{"parents":[{"commit":"fb68425dcefddf0d90be8f0fd77708c1646b4a37","subject":"Merge \"Use openstack cli to manage neutron routers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fb68425dcefddf0d90be8f0fd77708c1646b4a37"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-07 04:58:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-07 04:58:42.000000000","tz":540},"subject":"Add support for the resource_provider_inventory_defaults parameter","message":"Add support for the resource_provider_inventory_defaults parameter\n\nChange-Id: I0c3f4d3a80853f1f07325ee31317e415e922f128\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/700538820f3966d4679ac846fe3bca17b2e94864"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/700538820f3966d4679ac846fe3bca17b2e94864"}]},"branch":"refs/heads/master"},"b45eedcc31f1e4f71310ed8ff165def3b9d70ec6":{"kind":"REWORK","_number":2,"created":"2021-12-07 05:00:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/820650/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/50/820650/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/50/820650/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/50/820650/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/50/820650/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/50/820650/2"}}},"commit":{"parents":[{"commit":"fb68425dcefddf0d90be8f0fd77708c1646b4a37","subject":"Merge \"Use openstack cli to manage neutron routers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fb68425dcefddf0d90be8f0fd77708c1646b4a37"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-07 04:58:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-07 05:00:39.000000000","tz":540},"subject":"Add support for the resource_provider_inventory_defaults parameter","message":"Add support for the resource_provider_inventory_defaults parameter\n\nChange-Id: I0c3f4d3a80853f1f07325ee31317e415e922f128\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b45eedcc31f1e4f71310ed8ff165def3b9d70ec6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b45eedcc31f1e4f71310ed8ff165def3b9d70ec6"}]},"branch":"refs/heads/master"},"c8569e307cbf2520a5f5a8d1b515e3cac7acd7fa":{"kind":"REWORK","_number":3,"created":"2021-12-07 09:48:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/820650/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/50/820650/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/50/820650/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/50/820650/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/50/820650/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/50/820650/3"}}},"commit":{"parents":[{"commit":"fb68425dcefddf0d90be8f0fd77708c1646b4a37","subject":"Merge \"Use openstack cli to manage neutron routers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fb68425dcefddf0d90be8f0fd77708c1646b4a37"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-07 04:58:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-07 09:48:09.000000000","tz":540},"subject":"Add support for the resource_provider_inventory_defaults parameter","message":"Add support for the resource_provider_inventory_defaults parameter\n\nChange-Id: I0c3f4d3a80853f1f07325ee31317e415e922f128\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c8569e307cbf2520a5f5a8d1b515e3cac7acd7fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c8569e307cbf2520a5f5a8d1b515e3cac7acd7fa"}]},"branch":"refs/heads/master"},"31bb8ee1c28cd4bde6c5de5859b18ce4649f0977":{"kind":"REWORK","_number":4,"created":"2021-12-07 11:03:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/820650/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/50/820650/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/50/820650/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/50/820650/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/50/820650/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/50/820650/4"}}},"commit":{"parents":[{"commit":"fb68425dcefddf0d90be8f0fd77708c1646b4a37","subject":"Merge \"Use openstack cli to manage neutron routers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fb68425dcefddf0d90be8f0fd77708c1646b4a37"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-07 04:58:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-07 11:01:13.000000000","tz":540},"subject":"Add support for the resource_provider_inventory_defaults parameter","message":"Add support for the resource_provider_inventory_defaults parameter\n\nThis change adds support for the resource_provider_inventory_defaults\nparameter, which is used by openvswitch-agent and sriov-agent.\n\nChange-Id: I0c3f4d3a80853f1f07325ee31317e415e922f128\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/31bb8ee1c28cd4bde6c5de5859b18ce4649f0977"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/31bb8ee1c28cd4bde6c5de5859b18ce4649f0977"}]},"branch":"refs/heads/master"},"680c36b5fe60a0b2b78d8705e568241009f9e0e2":{"kind":"REWORK","_number":5,"created":"2021-12-07 12:46:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/820650/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/50/820650/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/50/820650/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/50/820650/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/50/820650/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/50/820650/5"}}},"commit":{"parents":[{"commit":"fb68425dcefddf0d90be8f0fd77708c1646b4a37","subject":"Merge \"Use openstack cli to manage neutron routers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fb68425dcefddf0d90be8f0fd77708c1646b4a37"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-07 04:58:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-07 12:46:22.000000000","tz":540},"subject":"Add support for the resource_provider_inventory_defaults parameter","message":"Add support for the resource_provider_inventory_defaults parameter\n\nThis change adds support for the resource_provider_inventory_defaults\nparameter, which is used by openvswitch-agent and sriov-agent.\n\nChange-Id: I0c3f4d3a80853f1f07325ee31317e415e922f128\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/680c36b5fe60a0b2b78d8705e568241009f9e0e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/680c36b5fe60a0b2b78d8705e568241009f9e0e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
