)]}'
{"id":"openstack%2Fpuppet-neutron~821582","triplet_id":"openstack%2Fpuppet-neutron~master~I173e0d516aeccc61b836573b81141a5dcab6d825","project":"openstack/puppet-neutron","branch":"master","topic":"deprecate-service-validation","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-16 08:29:37.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I173e0d516aeccc61b836573b81141a5dcab6d825","subject":"Deprecate service validation for neutron-server","status":"MERGED","created":"2021-12-13 14:34:13.000000000","updated":"2021-12-16 13:32:25.000000000","submitted":"2021-12-16 13:30:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821582-deprecate-service-validation","meta_rev_id":"e6524372c7199c874c18271d27943f6424a03b1f","_number":821582,"virtual_id_number":821582,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-16 13:30:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-12-14 08:27:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-12-14 17:17:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-12-14 17:17:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-13 17:29:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-14 08:27:23.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-12-14 17:17:38.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"edd67705b4ccb2fa704c89fc3529347d90a9bcc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-13 14:34:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20d8f46b44bea1e35c34a90626c8dfd3928976f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-13 14:46:45.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"41b1f32c495d4357e4c03fae83a7617fe9b29b89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 17:29:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce4695665d2a4d08a2d807877e855e83 : SUCCESS in 4m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4eced1eaa75b4f2db9fb27cee7e46f4f : SUCCESS in 4m 02s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b1984042ed0449a8020114d70ca8a42 : SUCCESS in 4m 25s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/258c8ce38e4c4aab89c812ea24d14ac0 : SUCCESS in 15m 53s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/33672a1dde04447bb649778f72aad01f : SUCCESS in 15m 29s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25a962f82449471bb666723be123a434 : SUCCESS in 14m 34s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a9a89e6a6704ff6a114577bb90a3748 : SUCCESS in 17m 20s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/070ce1b948d04ecd84ed36a85a361e8c : SUCCESS in 13m 49s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d9c75b72d0e4382a1b20f8af4ae619d : FAILURE in 33m 16s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c53b139abad49f3bd9e494575e6e2f8 : FAILURE in 42m 03s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2401565af47a4a2f94de6ebdd95aeeb4 : FAILURE in 32m 18s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af9e1b7626aa49bdbf3ff440e7379ccc : FAILURE in 27m 19s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/682e03f9c0f844d0b3e337a73903586c : SUCCESS in 22m 13s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/159b6def308b4170931308b4521a9cfb : SUCCESS in 1h 16m 25s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/91085bc558d544bf8ac4a48ac83a2878 : SUCCESS in 1h 14m 30s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b609759341fb4de19abbb36bbb98d288 : SUCCESS in 1h 22m 55s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3771dc4add7a480c84b838b3b9d72a70 : SUCCESS in 57m 47s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f89fe248cf824de2ad6d3b38e6ea2589 : SUCCESS in 18m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0acf7adbfec043bc8dcd4b0463c59f18 : SUCCESS in 47m 00s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e1a745b6af840e28faa88eac08cb3ba : SUCCESS in 1h 00m 44s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3009ef40cb64031ae2cd96269404c47 : SUCCESS in 58m 41s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/def42a1c9e384d868f7a343d2118347f : SUCCESS in 45m 57s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bb63cd990594b0ab5305b085f8f6a0f : FAILURE in 16m 12s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3bda9147d744d2d9f0be625a0151b5e : SUCCESS in 25m 01s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/699baf79d496478e9132cac62fd2252e : FAILURE in 13m 44s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed9e259f27fa4aeab41bb3783fb48f00 : FAILURE in 16m 47s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/67d675ed1cfd4d779dcf5e155f686e23 : SUCCESS in 2h 38m 00s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/638c2e0475f845d9a50b9c0a04648fe7 : SUCCESS in 52m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea4e33ab9a2c408fa1c5fdbbcd04ed0e : SUCCESS in 4m 19s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/534c0d6a398848bf846d8a6845f494a2 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":2},{"id":"86c66d7435636e613fab785a8d9550d5edd61eaa","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-12-14 08:27:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ec234d2dcd85d40f5d21b6e96d02ba87fcacf0ff","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-14 17:17:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2c1dc1ad0d146ee5331162d8c41d15cf97b18c60","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-12-15 19:30:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"60dc57a81a1473c96c9ee1d568561faf98c4e8f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 21:30:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b34363ecd5074698b4567b409a0d02eb : SUCCESS in 3m 37s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b57ca16f71454355b33ca6f6023557bc : SUCCESS in 4m 18s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fa172e7b9324e2f890b7c3a1e2eebb9 : SUCCESS in 3m 51s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b419b80929534c5dac407905ad0b1f74 : SUCCESS in 18m 54s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d2dc348c8c848b09d72913009158b44 : SUCCESS in 18m 43s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5f693a964e442cea62d3fa108342b1b : SUCCESS in 15m 21s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa492c70bf5b49719fda050eebc9ab6d : SUCCESS in 13m 58s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b93d044d50fe40b1aaa8c447b693b905 : SUCCESS in 13m 41s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e3e46b5b571491f92d466d23087d449 : FAILURE in 33m 38s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d1e3c1ed114476fbf99fb693f943586 : FAILURE in 37m 29s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b798513e7195458db2d9fea72b9505ff : FAILURE in 35m 20s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab26a14471a8490aab6bbba4236eeacb : FAILURE in 24m 37s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6dc67efadfa54718b577ded552601469 : SUCCESS in 25m 28s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4d603de8ffca44dab055633664e3b954 : SUCCESS in 1h 17m 05s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e8f28e7d791c454496d45ef882b1ab1e : SUCCESS in 1h 31m 01s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c9f9883a79d458eae12138f4ed4221c : SUCCESS in 1h 21m 09s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/58043c6c558f4547bc0c9fce840b70cc : SUCCESS in 1h 13m 53s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c3ec73b71ad430d8531ecb8dd8a256d : SUCCESS in 20m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ee71a73e5194ebe8f0bc521dda179b2 : SUCCESS in 50m 59s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/25f19a1c84904fe2ac39d0a736aff3aa : SUCCESS in 55m 47s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1727548251c46bd9e5bf590265eb64d : SUCCESS in 55m 45s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c524fe9f1f5b4ad983090221830202c9 : FAILURE in 1h 52m 44s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1449557c2204cefa4c32a5dddce29a8 : FAILURE in 13m 58s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/37faf699bc064737a54261e045263143 : SUCCESS in 21m 59s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4868e413327c4edea79600e8032197ab : FAILURE in 15m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d8e113febae4eab9c8fcbe5c4684d27 : FAILURE in 18m 59s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a576c6d4864a4667a3a7c623861f4b19 : FAILURE in 1h 02m 46s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/79b28c4602c24b6da77c686a14f2859c : SKIPPED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54eeedc4310a4335b95d677a047055fd : SUCCESS in 3m 34s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fcbd5fe644324cb69f1dd6c7d5d7ec61 : SKIPPED","accounts_in_message":[],"_revision_number":2},{"id":"0b81776fd634388b918d9a4bb073ae1092e550b8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-16 02:10:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"79128e480217b2ae4a311da8e24f27c40ac5f131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 05:13:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d06f4685bce54332a25c7f385626f55d : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f38ae4516b85417181da6b41753ed9ed : SUCCESS in 3m 58s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/066c48b0bdb44bd29f8039fd0ed18191 : SUCCESS in 3m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f01517857fa842ca838865fcd9afb853 : SUCCESS in 18m 59s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2aeab8eac45a446b829fff4cc887cc77 : SUCCESS in 21m 16s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/136ad47640ba450a901c89e247336689 : SUCCESS in 15m 15s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08d513e600ee4f2bad0a22c5ba93b55e : SUCCESS in 16m 01s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb7f60c516104ad6a01e5e28ffc1c319 : SUCCESS in 14m 19s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38f2e726d9f74b06a64054d241c9329f : FAILURE in 43m 34s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/920ece1c9e5143889ac5ab072c300094 : FAILURE in 38m 33s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac4c91eae9f04e0b9816a8c4711757d0 : FAILURE in 33m 12s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acc0903cc3c649059eda1231e42e997e : FAILURE in 24m 20s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/822d46f3d6d24a999b6f74fad70ba3df : SUCCESS in 24m 12s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0da2622821654b00ae221cae0ecfaa76 : SUCCESS in 1h 27m 44s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6fcf8a2c775448acb8d22407a4fe02e0 : SUCCESS in 1h 31m 25s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/150aadc5969949689ba75e999a1045f1 : SUCCESS in 1h 23m 04s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2fd1107175b043ac8856425173fe4eee : SUCCESS in 1h 05m 26s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bcf3a3ca322e449ea3638cd996e041c2 : SUCCESS in 24m 57s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a35c1f2aaac48bfbff231bba32b9032 : SUCCESS in 58m 08s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/30c862e777184172bceb2264cb232962 : SUCCESS in 55m 30s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4b9b420054949158b466872697fd612 : SUCCESS in 1h 09m 45s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2555072eb5e940cba1117e9175de8802 : SUCCESS in 50m 49s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65f1801115f143c5967662297c98c944 : FAILURE in 17m 30s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/27e677a3f89b463cbdff074a201bbc6d : SUCCESS in 25m 33s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f538d977dea74d5d8be2c0963747918c : FAILURE in 21m 47s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12c40ca5e3ae490eb884762171d990cc : FAILURE in 21m 05s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/99d99b148baa4b4fa43246f3b0c5e217 : SUCCESS in 2h 38m 08s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/75b1bcd3682947d592af3d23695699c8 : SUCCESS in 43m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3183021106be4b97b847925c57258306 : SUCCESS in 4m 14s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9e26598af9074f8489f6121f8955aecb : SUCCESS in 1h 07m 36s","accounts_in_message":[],"_revision_number":2},{"id":"b46098f163a802c5eb7e5e34e7d9569fe3d47b91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 05:15:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4832137843fd1db518e48e81fb7ba11a748ec806","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 07:22:26.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7064ec0918294cc9a5158a57b9135202 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a747538457a24bf1a8560512d9ac4e05 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74eda462d3fe49bd9082f5bdd1c4f70f : SUCCESS in 4m 09s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/38ae343c37ba4295a43473d4c1eb745f : SUCCESS in 15m 16s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d0aee5b564e437c9211125bb1e8e277 : SUCCESS in 14m 52s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3d6082bb6494fc482f3742eabd96ee3 : SUCCESS in 13m 38s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec6a98a0f98b4d36b4f11d28f77d9657 : SUCCESS in 12m 55s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9136944432f74ec6818632a70d6b8d29 : FAILURE in 1h 05m 48s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6f80ae7a9cea4e7cb243c2ddc7e6c989 : SUCCESS in 1h 16m 10s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e602d61ea694c65a50ae562c5f42e16 : SUCCESS in 1h 12m 49s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0b2dc4881c9420684a18b032cd87bcc : SUCCESS in 54m 26s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/5d5ffd2758f84aa28cab5ef1889d7b1a : SUCCESS in 21m 19s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b99accb16ee246b68d60f269f8327d67 : FAILURE in 17m 55s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0d68aec6950b465db4bc0a779fa9067d : SUCCESS in 2h 05m 55s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/bbd930d768b5434f8c979c1c47d07306 : SUCCESS in 1h 03m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ef6bf6fcf4048899652d45bd86c6aef : SUCCESS in 3m 53s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6c57098f58a4461cb911f7db757df227 : SUCCESS in 57m 02s","accounts_in_message":[],"_revision_number":2},{"id":"03e6745ae449edd21f22303ce88b5244648187ff","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-16 08:29:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c700308a9c98d4cc57f0335f4dbc8bb74d4c68bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 10:56:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/828a73cd9abb469498c2d40a134fda80 : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9079c6ae9f64cddbad40bc6fc739034 : SUCCESS in 4m 43s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/724e9adad96f4596add5efe123219ab2 : SUCCESS in 4m 44s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7809d0b0a86d4687ae9f24c004cafe77 : SUCCESS in 18m 04s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/0642f29e4f8340b68b27910954896101 : SUCCESS in 18m 06s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb8123f498bf4ac981834810c16fa122 : SUCCESS in 16m 28s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6297a4d2d29b4893968f330fa4b534f7 : SUCCESS in 17m 52s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a82c99f0c3d24cec942957b7afa8e96b : SUCCESS in 14m 53s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02fcb66d7df2466b8fad991c0f6a7562 : FAILURE in 47m 45s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a11c472ecae45baa79629e6ceb20db4 : FAILURE in 40m 50s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/114f1357d66046e78657593ec02159c8 : FAILURE in 49m 22s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24f296353b894277b01deb1333095a41 : FAILURE in 28m 14s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/53144857adf845409290b20f76b0f438 : SUCCESS in 22m 17s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/36fba76058b2492cbb72bb375d74b600 : SUCCESS in 1h 17m 53s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/40aa12cf64bf4c7489ebec1d6a215be0 : SUCCESS in 1h 40m 38s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d9262c4a5b754cccbbebba9dfbc8979e : SUCCESS in 1h 25m 07s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/13f97b00c12f4e85a04b955a36500f5d : SUCCESS in 57m 57s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2285c8ddfa944dee9d23382ef67434cc : SUCCESS in 20m 56s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dbc45eeb13e447dc978239e571220d86 : SUCCESS in 50m 26s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3cc1ccb7ec84a768ac210f3427f371c : SUCCESS in 56m 50s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/df2a998dfeb647b1939594aa89d5b84e : SUCCESS in 58m 17s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/01384ab681e24055a63c9b1702bd7a85 : SUCCESS in 48m 37s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42087446e4194d6d8274e1e6c36310cf : FAILURE in 18m 16s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a3807363ad34f91b21cd7720874ae6e : SUCCESS in 23m 00s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70d6ddba3e4c47f68e629055c1f24f7f : FAILURE in 18m 24s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8e7607a1c1b4dbeba4b835694906583 : FAILURE in 20m 07s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5d0972113b7549f3ab2f2309c3e75376 : SUCCESS in 2h 16m 11s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/233d263741d2461693bf548ca3663011 : SUCCESS in 40m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cfdcc25ca38452fab9e5c751d1251a6 : SUCCESS in 4m 29s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e35ff5b957b24fd3851f4c28e844cde3 : SUCCESS in 1h 01m 01s","accounts_in_message":[],"_revision_number":2},{"id":"f81def33215b4c33463e9ea769f7617f7c60fccd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 10:57:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9d9c5a4c883423421afe2d0815feb15c22e42cec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 13:30:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8e14c3c51ea46548572814310b528e0 : SUCCESS in 3m 06s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bd17a3f037340018266e046353c045c : SUCCESS in 2m 59s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a08a60faabdf455e90157f4cd07146fd : SUCCESS in 3m 26s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4f7323f871f6499b931ce0f39e3bacc4 : SUCCESS in 15m 33s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/5afbeae153b440ba9ce5bb93b7ef9289 : SUCCESS in 15m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8de23bafb21547f48f656ebd455e848f : SUCCESS in 14m 26s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/966c21284f7243ffbe9daa814bdc3b52 : SUCCESS in 11m 07s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fd1b35910a54432a990a3ae072f445ab : SUCCESS in 1h 21m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/967d1896d07e4cfdbb3cadb73e85cbdf : SUCCESS in 1h 32m 07s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6586ef060f4f4eb1bd5d61b621b2537f : SUCCESS in 1h 30m 54s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/27a7514e168d4163b2713b61174eeb45 : SUCCESS in 1h 07m 48s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6a8f75d5067413db0381934ddf47cbb : SUCCESS in 20m 55s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0e31af41ad048d787e055c901685f92 : FAILURE in 18m 43s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ae8ec666e26344ee98c7fed3f7bce60b : SUCCESS in 2h 32m 41s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/9cc29a0185f44167a381d0ac6a8a7113 : SUCCESS in 47m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56ade0550ecd46d0b6a803394d372739 : SUCCESS in 3m 49s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/42ed0706032645c2836b50c8148e2b74 : SUCCESS in 1h 11m 39s","accounts_in_message":[],"_revision_number":2},{"id":"68f52ad0842b2fdce6bdeda360bd2b84d947ab6d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 13:30:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e6524372c7199c874c18271d27943f6424a03b1f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 13:32:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/709166f749ce44f9b46efd624ca2e948 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"35659b8d918e8e10e8dbbadf29339d04069bceb3","revisions":{"ed990fb237dc57787ce5f242f6fe15335ad9a84c":{"kind":"REWORK","_number":1,"created":"2021-12-13 14:34:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/82/821582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/82/821582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/82/821582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/82/821582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/82/821582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/82/821582/1"}}},"commit":{"parents":[{"commit":"e4e61a9e4ce53cfb7afe2a027d384339573b83f9","subject":"Merge \"Clean up direct dependencies on puppetlabs-apache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e4e61a9e4ce53cfb7afe2a027d384339573b83f9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-13 14:33:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-13 14:33:49.000000000","tz":540},"subject":"Deprecate service validation for neutron-server","message":"Deprecate service validation for neutron-server\n\nFollowing the past deprecation in puppet-keystone[1], this change\ndeprecates service validation for neutron-server.\n\n[1] 3c95205e0253895e477b9135c10f5175d4166cfc\n\nChange-Id: I173e0d516aeccc61b836573b81141a5dcab6d825\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ed990fb237dc57787ce5f242f6fe15335ad9a84c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ed990fb237dc57787ce5f242f6fe15335ad9a84c"}]},"branch":"refs/heads/master"},"35659b8d918e8e10e8dbbadf29339d04069bceb3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-12-13 14:46:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/82/821582/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/82/821582/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/82/821582/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/82/821582/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/82/821582/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/82/821582/2"}}},"commit":{"parents":[{"commit":"e4e61a9e4ce53cfb7afe2a027d384339573b83f9","subject":"Merge \"Clean up direct dependencies on puppetlabs-apache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e4e61a9e4ce53cfb7afe2a027d384339573b83f9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-13 14:33:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-13 14:46:15.000000000","tz":540},"subject":"Deprecate service validation for neutron-server","message":"Deprecate service validation for neutron-server\n\nFollowing the past deprecation in puppet-keystone[1], this change\ndeprecates service validation for neutron-server.\n\n[1] 3c95205e0253895e477b9135c10f5175d4166cfc\n\nDepends-on: https://review.opendev.org/821585\nChange-Id: I173e0d516aeccc61b836573b81141a5dcab6d825\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/35659b8d918e8e10e8dbbadf29339d04069bceb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/35659b8d918e8e10e8dbbadf29339d04069bceb3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
