)]}'
{"id":"openstack%2Fpuppet-neutron~822993","triplet_id":"openstack%2Fpuppet-neutron~master~I91abf7a820685c06d195541c173042469e7b583f","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-07 22:26:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I91abf7a820685c06d195541c173042469e7b583f","subject":"Remove deprecated nova::config::api_config","status":"MERGED","created":"2021-12-27 07:16:39.000000000","updated":"2022-01-07 22:30:56.000000000","submitted":"2022-01-07 22:26:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822993","meta_rev_id":"ecf608046a6cdc4be6e9ab4a2ddbda55a01a41c9","_number":822993,"virtual_id_number":822993,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-07 22:26:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-01-07 16:35:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-01-07 18:49:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-01-07 18:49:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-27 09:40:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-07 16:35:50.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-01-07 18:49:57.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"14e7fdea58cd5b524d11bb833fafd4ecba7c6e38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-27 07:16:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ca260cf7cf0c188a1d73802d04837df85738c53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 09:40:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42a68d238b0c4e18b8f9b65475032d0a : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0b7a67689c049fb8b208e2e179b75fd : SUCCESS in 3m 19s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f489c6f8e5046d1ab0f03e95fd4aa42 : SUCCESS in 3m 09s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4734ad2d858f46c4b6f135519f76e277 : SUCCESS in 16m 15s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa655c99d61d4c5899deca1937f15540 : SUCCESS in 14m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9302069664dc4ecca49bd82ffcbe0496 : SUCCESS in 11m 44s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b762b4973d4473e91d057243b683cfb : SUCCESS in 11m 54s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da9e2975c5d949989e1b33f0ce5035e3 : SUCCESS in 14m 14s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1e8679d14aa42dfa624790ddd8a5eda : FAILURE in 27m 27s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ec17b64ca0e466e86f8a8e55513a749 : FAILURE in 43m 24s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e570ba2406e47cb853208d5f9abcecf : FAILURE in 32m 53s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f87ce5504024f529242e4b0e7895775 : FAILURE in 24m 08s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e3f04bc7814a4797a9ff3dff3bf1990b : SUCCESS in 23m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/26ac6f05dbc2472d9a4debebdae40895 : SUCCESS in 1h 07m 06s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f4329ec73aa4adeac27ca2dcf39f514 : SUCCESS in 1h 07m 07s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ed978c7674f4c8f916c8edbabf96b76 : SUCCESS in 1h 12m 28s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3edd4840b5aa4ba28d5a656c5616cbaf : SUCCESS in 1h 06m 08s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a373b5a6def74e3180b0a582b1bf2084 : SUCCESS in 20m 35s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/915258b610444ed5a1f0fbd7e711957e : SUCCESS in 48m 46s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8aacc629f0fb44d5b2f80537383459ea : SUCCESS in 42m 54s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/54fe7e05b2b04adbaeb6c7b3eca5c8a4 : FAILURE in 23m 36s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c9d1148f3974c42a10e09a934d45c1f : SUCCESS in 46m 55s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d04024210c824876919e19f4a8042644 : FAILURE in 13m 27s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/6707591de98945a99f89b8619c94da28 : FAILURE in 24m 22s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a439e64fb78e4f85987a70e13d720b28 : FAILURE in 18m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b7795f7ff6f44db964ee2186d86bbbf : FAILURE in 16m 53s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a5bea38c65a0403b9f76dd32c3beee3a : SUCCESS in 2h 21m 34s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a6b9c5f05c4a4946afc21b0264213080 : SUCCESS in 44m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57d27d9dd97540c1b80035b7bf38959a : SUCCESS in 3m 53s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ab54ce452234434788bbc5cc1ded0a55 : SUCCESS in 1h 12m 35s","accounts_in_message":[],"_revision_number":1},{"id":"f79ac0bd5d92df0c5cef0b8de9970a2fc5b3643c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-27 13:47:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1d14330adedbb452503cbb428ad09d7c62a34e10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 16:24:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4dfc36884cd3401da900f7b3be271600 : SUCCESS in 3m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b17837c4b1574767aa030db343e3aaf4 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7948038cf22343eab403bc91b5c69424 : SUCCESS in 3m 15s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/757b54dd618e480b8e13d57a0e3bb5c1 : SUCCESS in 16m 28s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/68ca0c7bbe944d1986cff13ea5f6fa73 : FAILURE in 18m 59s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d21f2c84adda4de1b2a0965dce38271f : SUCCESS in 10m 31s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c8d2e7a02d64a00abe7aaf3fd1655f6 : SUCCESS in 14m 44s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c6938a1e9254f9484f9ad157f80b682 : SUCCESS in 13m 25s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b6a2a73137043b98ca61ed477018233 : FAILURE in 40m 24s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7863acc45364c21b71f3130bced2984 : FAILURE in 47m 21s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba1b22751804431a92e478848fa940f4 : FAILURE in 34m 52s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b120a9d4fbc94516b07add934958ca9a : FAILURE in 33m 56s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/41f685a4fdd14742ace8f0c694265cfb : SUCCESS in 21m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/296a15ba51244b52af867bc9003b136e : SUCCESS in 1h 21m 56s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9b817476ab43430ebfc3d4d1b96f81c1 : SUCCESS in 1h 43m 10s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/866a2a89bac44239b666a9a435343378 : SUCCESS in 1h 29m 11s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4d49333dc6e04c82915857387694c49a : SUCCESS in 59m 57s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/607eedd5207c40f98ff4f8cd8fe0ba83 : SUCCESS in 17m 19s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae9c6890951f4c279b6ad85e5de6216f : SUCCESS in 42m 54s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/75e2639a4e184e90835ff69446bc9f79 : SUCCESS in 49m 25s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/712b0bda2310434a9c2a090be8d51554 : SUCCESS in 49m 16s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee09e4e847f74d15ad441bb0ac15aef8 : SUCCESS in 38m 19s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7abb833ee4e44d66acb4b8c945be4dc4 : FAILURE in 14m 36s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a49e2172ec5437d86c0480a89b9f1b1 : SUCCESS in 23m 40s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/483d3ec8a8544580b6a4dfd7ee69e64d : FAILURE in 16m 54s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/384801eb649c4328969dc6a4ed56dcb2 : FAILURE in 12m 33s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3ab33f292fef4968b54e487a251b5d34 : SUCCESS in 2h 27m 26s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/7d1da8285a6143cc805c7171463d96ad : SUCCESS in 43m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7368a46399c487fab109bf95f63e8b5 : SUCCESS in 3m 56s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6d1792fe48eb491fb44dac2b6f0d4a23 : SUCCESS in 1h 09m 55s","accounts_in_message":[],"_revision_number":2},{"id":"58a5490595e43a47055beb0790262f7f67ede768","tag":"autogenerated:gerrit:setPsDescription","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-28 01:56:54.000000000","message":"Description \"Rebase\" removed from patch set 2","accounts_in_message":[],"_revision_number":2},{"id":"bc5ed306c78bb399085abfc1ebfb445c9ae59565","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-28 01:56:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a9cf0aa089dcf8d8c397f4c794967087e6560d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 04:33:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c66709a27b1c42a9814bb7bab41ab389 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ec1e372c4d3479cae3ee47f58633ec6 : SUCCESS in 3m 44s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59fd6c9852094858afd8bbca0169b821 : SUCCESS in 3m 56s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/effa90b3ab624ca5b7cd6f5b200f9636 : SUCCESS in 17m 49s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/79be71fd6975408e9c1977ea8b42dccb : SUCCESS in 34m 06s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ab6b2dc8bfb4c1c919f3ddef298e9cd : SUCCESS in 14m 35s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ab759c26a684f0bb0d8f721cc0ff63f : SUCCESS in 14m 58s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f21769b8aeb428cbe2ab0e44bdedbf6 : SUCCESS in 14m 43s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6eea3676efad42a195b883b796b9b86b : FAILURE in 35m 25s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca44cbb580b346ac8e487ba5a7980bcd : FAILURE in 38m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e52212d118c4fee80e07fa82312d36d : FAILURE in 34m 04s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/988ab0ffdb564cdfa44c262b9df83104 : FAILURE in 28m 21s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ef8faa09f234feeb3c8b08c1d2fdea9 : SUCCESS in 25m 43s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/15e4a62fc5e04efaaf0fbd0de1511a67 : FAILURE in 1h 18m 38s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e3ad325f6c6141b2acedc7ce3aa79fe1 : SUCCESS in 1h 19m 49s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/10d5ef0f864e4534adf80c5b016639be : SUCCESS in 1h 25m 23s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5034f4297024479993584aab19b180e5 : SUCCESS in 1h 03m 20s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d88937f1e074ae391d6f0f8ff7b0a65 : SUCCESS in 24m 42s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/76067171ce064949af9629f0a7a73e93 : SUCCESS in 56m 43s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/360cd50484274a849f56d057a2229eb3 : SUCCESS in 1h 11m 55s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a403379aa1f4fa9a3525db4b223dadd : SUCCESS in 1h 03m 29s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc6075ba6cab4cdb955910285cef8cd5 : SUCCESS in 53m 44s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1ac1ebf3f114d07950ab97f0a04aacb : FAILURE in 17m 57s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/4d2712846ec340ce99f18fd4fcc866a7 : SUCCESS in 23m 32s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/883d02457972422d9b33db9286772aa1 : FAILURE in 18m 02s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a5d0dd6d0614a989d60ce143236a6db : FAILURE in 17m 53s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/04e71ce2ad334502932756c853f06008 : SUCCESS in 2h 33m 55s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2ed718078d494479b0d8e2392966e543 : SUCCESS in 45m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/818d41cbcb3b48289825299c45477f90 : SUCCESS in 4m 07s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7fda748a2c864cde88365a1d3adaaf16 : SUCCESS in 1h 10m 00s","accounts_in_message":[],"_revision_number":2},{"id":"5d5dc71089a4f46d43c64d84a2011dfb2ac54917","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-28 04:37:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5dbc4a0448f2be86af222b7a76fb7c08f44fd678","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 07:07:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/187c6a3050804acc984b382f864a218f : SUCCESS in 3m 37s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f86dcbb075a54db6ad398a4dfb5e7f69 : SUCCESS in 3m 29s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9e66fad6e3d4765b88ab1a36a76b754 : SUCCESS in 3m 36s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6d5cc4bea3024f4a962abdfeb9c8e5b2 : SUCCESS in 16m 04s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb499f6f7956471583e7ed515942ea02 : SUCCESS in 13m 26s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a40b4275875642e59153c43ea340e498 : SUCCESS in 14m 30s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/917b0a6f0a7344959d89fa8179ef7711 : SUCCESS in 15m 41s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39a693be6c3c48c89938223997d87efd : SUCCESS in 10m 50s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd542d08a0dc4ac3804d9f8d45334040 : FAILURE in 32m 06s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a42dd636d42a49eca517b0f3e2a1e847 : FAILURE in 28m 29s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc1326c5f3604caaa9cc70cf8898e234 : FAILURE in 25m 16s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/825bdbec75ff4d6d993ffcb5bc9d0d4a : FAILURE in 26m 29s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a70cdc39af264f1480057a6c4b16140a : SUCCESS in 27m 28s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/287f78bc53a747a090ee76ed481aeec8 : SUCCESS in 56m 13s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8273264f98c44e25b78adbaa05a6fdd3 : FAILURE in 1h 41m 47s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/87647b310b094b6d8edc7ccb649bfc42 : SUCCESS in 1h 07m 59s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/97ab3ce6b4b14eedaa2b2d23d7aaa829 : SUCCESS in 56m 54s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d34f20b867246c6aa38a130ab7b12e7 : SUCCESS in 18m 29s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b505463ca244ccd901e16dea5cd66ac : SUCCESS in 48m 08s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f17bfd7b89394b8fb9ae9f5933207a14 : SUCCESS in 50m 45s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dbe0e607b8dd49c18d2c5899a691acec : SUCCESS in 54m 19s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/116b70e640c94324b34763948adf9705 : SUCCESS in 41m 37s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d83e76eea2f4c9c9c95f3313687ad58 : FAILURE in 17m 53s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/216e35fbfe9645e3bab8e996020368b2 : SUCCESS in 22m 47s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83e2a332b1a946f1ad28c2cad312ea67 : FAILURE in 18m 24s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f970d26cf6e457bb6fd9aab81a026b6 : FAILURE in 16m 59s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/de05959f29314e51a1af49c2c84198bd : SUCCESS in 2h 26m 58s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/3e31d894ecf542358e90726477273900 : SUCCESS in 44m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed43383f7abb4715be800825e51ee711 : SUCCESS in 6m 01s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0cff2fe4b06b4a32975e599209cd44a4 : SUCCESS in 1h 08m 16s","accounts_in_message":[],"_revision_number":2},{"id":"4ca3de06bd68b459df3182e9da0d8f52f38bbabf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-06 02:07:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6b827eab8314bc6050e1b82d9cf58dfc8c40f7d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 05:27:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fdd06dbdb8404ab9bbb1b130b1b294cb : SUCCESS in 4m 07s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27dc5598fc2b4efea65b08e86fac5a1b : SUCCESS in 4m 45s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b8c7acfc695483a9a5b9abaa34d7686 : SUCCESS in 4m 46s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/209cf3ec2b624bb6bf1c8874452ae8ec : SUCCESS in 19m 40s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/17870deb6a51457295fd63dd09f5a7f5 : SUCCESS in 20m 02s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2ab3d66897e4ff9936bfe086a851378 : SUCCESS in 15m 41s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10f446c8e8124c1b9bdc63dfdec2f357 : SUCCESS in 15m 32s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f738607d03b34faaaaee83f6b89072c6 : SUCCESS in 17m 48s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acc82e96f94a4606842bf0eb37b6a653 : FAILURE in 33m 44s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4214cd0846f4d3fb3577bb3f1f2604d : FAILURE in 31m 53s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3106f4fcdec406b9b788bb341cb79b2 : FAILURE in 28m 39s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/220c6611a0964268bf77675b9937d65e : FAILURE in 23m 17s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/56588c32ebf646d98109caddea6d1106 : SUCCESS in 27m 10s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/56fee32fa4c14b5e8140f787417331bf : SUCCESS in 1h 25m 48s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/27a4aa1081fa4aceb75fdf922bcdaf8c : SUCCESS in 1h 27m 56s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6739aafc1ee54c759653312498c02851 : SUCCESS in 1h 23m 44s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb5d7684fdbe41519e7017effe3ec0d2 : SUCCESS in 1h 09m 18s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea33efdf51ea45f6bf2226927460fcc6 : SUCCESS in 23m 07s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/272325abf2724cb38e6d94509b9ea3c4 : FAILURE in 53m 40s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/15e4a320f7f44da1b2d7958183db355c : SUCCESS in 1h 07m 09s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/797b1d0f00834426b9e8da5718defd03 : SUCCESS in 1h 04m 51s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/99ec925a4e4540efb2cc3cc8a2ad1be3 : SUCCESS in 48m 21s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5083591f37af420694640d38f7c35590 : FAILURE in 18m 32s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ebce1baa5112482fbaa03348425bf77d : SUCCESS in 25m 43s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8a91501d6204295a802effeb2bfcaa4 : FAILURE in 18m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adf5913edf4d40b7ac982cf8e95104c3 : FAILURE in 18m 44s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/52f902debe6143b4ad2fbea17add83f4 : SUCCESS in 2h 47m 30s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4293a62702da453eaef0273cd811058b : SUCCESS in 51m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8262fda1e67c48a79d9c9b03bb6352a1 : SUCCESS in 4m 08s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fc1d166b8f564f57b53a1f3c19331f91 : SUCCESS in 1h 15m 58s","accounts_in_message":[],"_revision_number":2},{"id":"b5682ceea4ab9fa267e46f5801c42b32a8898e00","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-07 16:35:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c1f0d01046e59f4736d0c560248f47d844a855b5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-01-07 18:49:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fef119d3894d552125d74a1a559af4343b0600ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 18:51:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"00042e6b9ffa1b558db74739090131e6dd71f6f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 22:26:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f15c3a62a7ea4bc69b6d8fa913499428 : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb40b8108d8c4ffc957f542a0f64830e : SUCCESS in 3m 39s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f7e1e9ff8644dc3b904e62eca06a46f : SUCCESS in 3m 30s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/33601382dc644b429f95a56d6ae8d7d8 : SUCCESS in 16m 41s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c27d776cd46f4359b5b7cc289042a20e : SUCCESS in 16m 50s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24d2b2eb0bd14ee1a48cedfe185655ae : SUCCESS in 14m 49s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d23d8f25d3084311b357ad4c75e4fbaa : SUCCESS in 15m 32s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c584722be16d4d8386acff6fba91c436 : SUCCESS in 1h 14m 05s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b7f72b20636f48cb8f5e31a6b4257402 : SUCCESS in 1h 30m 08s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9147db025e514feba5f1e36908ad83ba : SUCCESS in 1h 20m 47s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/499c2e26e23f4343bca102b5ba8718f1 : SUCCESS in 1h 15m 58s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/09078a9d8b134db4b7ffc28dcae63e11 : SUCCESS in 26m 49s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eaf6d54254a74b5cb7c7a8f47fb10021 : FAILURE in 17m 08s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/de77e8eac6f945cea9024191930ba238 : SUCCESS in 3h 27m 15s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a20e6431ff6e4f3d897e32c7812cb466 : SUCCESS in 58m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/866854dfb83349d692b5e022c976070b : SUCCESS in 3m 57s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/323c7e621f304313913ef702659d3811 : SUCCESS in 1h 19m 49s","accounts_in_message":[],"_revision_number":2},{"id":"6cb5c3c71f332e0f168d17bfba12045a2f2ec2f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 22:26:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ecf608046a6cdc4be6e9ab4a2ddbda55a01a41c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 22:30:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/604537abf0a346b1982e2f7461703bd5 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5e5f2eb614eb37a7f222cc62622656ded056efb5","revisions":{"6a99d83c379ecf81e87cdf0cfa496e01287642ad":{"kind":"REWORK","_number":1,"created":"2021-12-27 07:16:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/822993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/93/822993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/822993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/822993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/822993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/93/822993/1"}}},"commit":{"parents":[{"commit":"e4e61a9e4ce53cfb7afe2a027d384339573b83f9","subject":"Merge \"Clean up direct dependencies on puppetlabs-apache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e4e61a9e4ce53cfb7afe2a027d384339573b83f9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-27 07:15:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-27 07:16:25.000000000","tz":540},"subject":"Remove deprecated nova::config::api_config","message":"Remove deprecated nova::config::api_config\n\n... because it was deprecated during Victoria cycle[1].\n\n[1] 2daa0b16444f2a2a865b618cbca76fd18c6cd7c6\n\nChange-Id: I91abf7a820685c06d195541c173042469e7b583f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6a99d83c379ecf81e87cdf0cfa496e01287642ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6a99d83c379ecf81e87cdf0cfa496e01287642ad"}]},"branch":"refs/heads/master"},"5e5f2eb614eb37a7f222cc62622656ded056efb5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-12-27 13:47:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/822993/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/93/822993/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/822993/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/822993/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/822993/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/93/822993/2"}}},"commit":{"parents":[{"commit":"c67a8190d59d838d80de2fea73842c1008f74741","subject":"Merge \"Remove deprecated parameters of neutron::server::notifications\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c67a8190d59d838d80de2fea73842c1008f74741"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-27 07:15:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-27 13:47:10.000000000","tz":0},"subject":"Remove deprecated nova::config::api_config","message":"Remove deprecated nova::config::api_config\n\n... because it was deprecated during Victoria cycle[1].\n\n[1] 2daa0b16444f2a2a865b618cbca76fd18c6cd7c6\n\nChange-Id: I91abf7a820685c06d195541c173042469e7b583f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5e5f2eb614eb37a7f222cc62622656ded056efb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5e5f2eb614eb37a7f222cc62622656ded056efb5"}]},"branch":"refs/heads/master","description":""}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
