)]}'
{"id":"openstack%2Fpuppet-neutron~828409","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fwallaby~I6fc0eb085c09fb1e24ff09d4f8199d2cf57dd209","project":"openstack/puppet-neutron","branch":"stable/wallaby","topic":"bug/1940838-stable/wallaby","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2022-02-10 12:19:45.000000000","reason":"Change was submitted"},"28223":{"account":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"last_update":"2022-02-10 06:50:29.000000000","reason":"cjeanner (Tengu) replied on the change"}},"hashtags":[],"change_id":"I6fc0eb085c09fb1e24ff09d4f8199d2cf57dd209","subject":"Add report_inteval parameter in ml2 baremetal","status":"MERGED","created":"2022-02-09 14:23:20.000000000","updated":"2022-02-10 12:21:32.000000000","submitted":"2022-02-10 12:19:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828409-bug/1940838-stable/wallaby","cherry_pick_of_change":806192,"cherry_pick_of_patch_set":2,"meta_rev_id":"0f27db50eb21d549ffd41fad001938fe6614bd1b","_number":828409,"virtual_id_number":828409,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22954,"name":"Juan Badia Payno","email":"jbadiapa@redhat.com","username":"jbadiapa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-10 12:19:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-02-09 14:54:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-02-09 20:13:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-02-09 17:23:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22954,"name":"Juan Badia Payno","email":"jbadiapa@redhat.com","username":"jbadiapa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-09 14:45:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-02-09 20:13:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22954,"name":"Juan Badia Payno","email":"jbadiapa@redhat.com","username":"jbadiapa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22954,"name":"Juan Badia Payno","email":"jbadiapa@redhat.com","username":"jbadiapa"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-09 14:45:46.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2022-02-09 14:54:43.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-02-09 17:23:22.000000000","updated_by":{"_account_id":22954,"name":"Juan Badia Payno","email":"jbadiapa@redhat.com","username":"jbadiapa"},"reviewer":{"_account_id":22954,"name":"Juan Badia Payno","email":"jbadiapa@redhat.com","username":"jbadiapa"},"state":"REVIEWER"},{"updated":"2022-02-09 17:32:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-09 20:13:31.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"1bc804d30f695be8be65a0025dbb095ed40e721e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-02-09 14:23:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* manifests/agents/ml2/networking_baremetal.pp\n* spec/classes/neutron_agents_ml2_networking_baremetal_spec.rb","accounts_in_message":[],"_revision_number":1},{"id":"2458939ae6a693b3026f2cba5a55e3ef462a6ca8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-02-09 14:38:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2388e996fbc7daa4c5463fb915ea05764978796e","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-02-09 14:45:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"38c3f21c15740762be382d56e6bc569d9868db63","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-02-09 14:47:09.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"84cfe180d9b760d9d174b1c23efcd5d0ef8eda07","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-09 14:54:43.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"65294a13bf8e9839c50dae1a46cf6ea6e53ff8c7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-09 14:55:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e9867f828f238e07dc1399d66203f6f99c9ba4ee","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-02-09 16:56:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"749b53c2c1e56bb3c8fa0bf3b1d9b1db36e9ce76","author":{"_account_id":22954,"name":"Juan Badia Payno","email":"jbadiapa@redhat.com","username":"jbadiapa"},"date":"2022-02-09 17:23:22.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ed0ca580dabacd5f8b28ce26aadd0c24ee6027ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 17:32:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa67ece570b4477bb80a3ac29399c7c3 : SUCCESS in 3m 49s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a229775ee1f45b493d7b5ca0242c69d : SUCCESS in 3m 40s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ae223b7a4fbb4cfda19d12d951cbff71 : SUCCESS in 18m 22s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3b7816ac0ef47c6a0c2c075b53010c9 : SUCCESS in 17m 08s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21473f96ebbd4c28b7030d2aefca68ed : SUCCESS in 14m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4fd5fe5ccbd4adfbde4240237a2310e : FAILURE in 40m 42s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e57a383ba9f4d6796e2b49fd478fcc3 : FAILURE in 1h 05m 37s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e3339b3be4a499984e5f78ac16ae643 : FAILURE in 33m 33s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dbe43f4228b425fac4c1465d8f25b4d : FAILURE in 23m 51s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/667767f3c7c3439f9dcc70098e2e8792 : SUCCESS in 1h 23m 59s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/10c1bbf608c1402db55c66d4db780b38 : SUCCESS in 1h 22m 45s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/11847b5d72b9492b8482ff8c4c8b8926 : SUCCESS in 1h 30m 17s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4c4d82af66b45da9116a3968f5c1b4e : SUCCESS in 1h 02m 38s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/c3396405c85a4adf9a4b15568b5e24a4 : SUCCESS in 30m 42s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e03589f8e65846b0a9625838e87a75ef : FAILURE in 16m 08s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c4bde0808ef045928eb5dfd0e28d127d : SUCCESS in 2h 19m 27s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/62b169b1c0be46baaeeb5c4890507ba6 : SUCCESS in 57m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a46ceff3f9746fd923b599807619402 : SUCCESS in 5m 29s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3e650dea87ec48cdbc1c4c482c4ab9d5 : SUCCESS in 1h 03m 54s","accounts_in_message":[],"_revision_number":2},{"id":"8597e91320fd2c479c993f58332cd425ad0c19df","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-02-09 20:13:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0f27d6ac46d48f9bc0b04f4da31b61da853e7288","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 20:15:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f92b8768e2040ebe1e1c7a3b4279478adfc96c9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 22:35:22.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af9029fc5fec4cb0b171c27e5c59fe2d : SUCCESS in 5m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f84f5fdc0b84b8887d68af55edd1d68 : SUCCESS in 5m 00s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e3b1f0ddc5e4c83985a0bea27d891e1 : SUCCESS in 22m 44s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fffa6451931f426b9fa9bec5d4bd7e91 : SUCCESS in 17m 35s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/36c2be9b23824218a7e1b7a3b756a5a0 : SUCCESS in 1h 16m 44s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/89c7db654a3f4c3fb5c6c204e17d6af9 : TIMED_OUT in 2h 04m 15s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/632494d59f64497594f271670c881944 : SUCCESS in 1h 18m 24s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/48918c5761964cfe8363e82131b53aa8 : SUCCESS in 1h 04m 17s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/02d37060ad8c4d68989f5f3b2bb86566 : SUCCESS in 24m 33s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acc4222086574ca586a5033f3265fe2a : FAILURE in 24m 48s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e581b8a1ce754cc4946f80e069711f84 : SUCCESS in 2h 16m 40s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/0aaef8892103424ab552775b2dd4d7e4 : SUCCESS in 58m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07f7a23ebff447f79e58c773e205a2b3 : SUCCESS in 4m 21s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/631241d6c8f049f89783e59789cad713 : SUCCESS in 59m 36s","accounts_in_message":[],"_revision_number":2},{"id":"2539802f9666b0cd27530e3241a3bc3069f28c63","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-02-09 22:43:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ea120dbaa306d3d89aca68f53a5a3625fb83c8bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 04:40:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2af4374e4b9d47f682c463ccc3690d7c : SUCCESS in 3m 51s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d38d5286319d4fbc91968067f6cc96c1 : SUCCESS in 4m 23s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/84678c8b80204b3ca037c3e4e8232521 : SUCCESS in 16m 55s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad32e4c38a824694aefae170fc87fde4 : SUCCESS in 14m 03s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ba5bbc21e2c4b398b035922cf2add33 : SUCCESS in 11m 05s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3a1e448293b49818647639b4c1c2fff : FAILURE in 38m 48s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15dd4162ef3a4cebad6faff9364ac193 : FAILURE in 34m 48s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b08435ed5d004cdfaf76a260c5e19ee4 : FAILURE in 36m 24s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16c95b8e1ec24b3081455a24525da4eb : FAILURE in 1h 05m 07s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e788877326984c8aa9961b5662540342 : SUCCESS in 1h 28m 59s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8ea836f3572849628a00d6d996f8b903 : SUCCESS in 1h 40m 05s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dc1c9958a0bd4c94876e390610510350 : SUCCESS in 1h 32m 05s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/206fce43e3234ea882f686dbe091ed8e : FAILURE in 1h 11m 06s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/10137a3054a1451fba5ed6166f45f829 : SUCCESS in 26m 57s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2936c83dc39b48219d67aae17c9ebdad : FAILURE in 15m 15s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e8d16aa10d294adc8c5f745ac7413185 : SUCCESS in 3h 20m 02s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/5dcfe7bd8b90487db984ba8591960961 : SUCCESS in 1h 04m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9842a6e640346ff8123b305a7fb1171 : SUCCESS in 6m 51s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/694932f021bb46a3abaa4fd421a2dc59 : SUCCESS in 1h 10m 10s","accounts_in_message":[],"_revision_number":2},{"id":"050ec659f4a185f873d270dcf73a6257aab72d0a","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-02-10 06:50:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9938e2b945483ce9d49a727e308109e4f1737382","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 09:21:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc8dfbca00f341148d11846ba33f5a1e : SUCCESS in 3m 02s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4067c03cd91e4766bc75068482344e8c : SUCCESS in 2m 46s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/286abbc3677c4ce4aaaf9cd7e4fa6668 : SUCCESS in 15m 03s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61d3f38ab3444983808fc06ca9c3329c : SUCCESS in 12m 03s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1894341d3f824ef5a24abeb6c568dac3 : SUCCESS in 11m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16cd46e7b46143248466ae6d7c04bf0b : FAILURE in 45m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be58634a952342dca45989dc412c2c1c : FAILURE in 39m 10s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ec18c8746c8409c8214337eedf440ea : FAILURE in 35m 08s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e67db142e90943239072fd79e8209540 : FAILURE in 32m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc3a75f48c8f4e52840217dce3f479b7 : SUCCESS in 1h 10m 33s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/441229d33bea4294a22c9ceb26c27f20 : SUCCESS in 1h 22m 31s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/684c68efa893446fa3d0ba394d6a8dfb : SUCCESS in 1h 13m 52s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6559123545ba409fb9e4e67935352782 : SUCCESS in 55m 35s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/784609fa71a94239a0a30b0d13016fa5 : SUCCESS in 22m 31s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f04799fe4eda4905ad0c7b32821f1929 : FAILURE in 18m 24s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/480daaae3f2b413c86499c732123cbd2 : SUCCESS in 2h 26m 55s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/39cbbc79e2d34d9a800977db534dea53 : SUCCESS in 1h 04m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc313a2438eb491d95bd0911a2b6ee54 : SUCCESS in 3m 24s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/58a4a76f24a543eaa5b744259d83b454 : SUCCESS in 1h 05m 13s","accounts_in_message":[],"_revision_number":2},{"id":"f3fa7c205ca3938e33a00b60b26bd9eabb432709","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 09:23:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ed74644d15b7d850bb723c53469b9ea811ae0e9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 12:19:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9b8f268d3544585a19f3fcdf0f5d53c : SUCCESS in 3m 23s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42a9712621b04767b363c5ccf7608136 : SUCCESS in 3m 35s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5f5de1491dd4ad58d7d31620b3d629b : SUCCESS in 16m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c7c6c48026f4a1a9d44980b22d5ce26 : SUCCESS in 13m 57s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2db5f2976c5c4d35b611411e040d12c4 : SUCCESS in 1h 17m 09s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8eda2851a0d844eaab8867face4856b5 : SUCCESS in 1h 33m 53s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0fc387d682af4e2ba637703b68d68bed : SUCCESS in 1h 29m 41s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a97dd68b3b3b49168cbfdf28cfc9661f : SUCCESS in 1h 13m 10s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc894cea39cf43b7ba125301040fed60 : SUCCESS in 21m 45s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69c8b1e295074bf2bee0be2d9b4e69fd : FAILURE in 17m 13s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c70f730b9be643ffab5d169dc4b2a4bb : SUCCESS in 2h 52m 53s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2c15feb469cb4e3694374a74087f921c : SUCCESS in 1h 11m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d578c27046e45a6a2a502f0a070664e : SUCCESS in 3m 38s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ccdcf29040364c55a29645b3b1229e60 : SUCCESS in 1h 30m 24s","accounts_in_message":[],"_revision_number":2},{"id":"915bfc58a9931b4df4df0f24d203387f24469a6b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 12:19:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0f27db50eb21d549ffd41fad001938fe6614bd1b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 12:21:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b5e623e804b46968ed22174144595c6 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"41bcf66eafc092890cdd6d2560d02642c5a55761","revisions":{"fc8919ba6cdeaa93e7883e228d89dc3fec797e1f":{"kind":"REWORK","_number":1,"created":"2022-02-09 14:23:20.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/09/828409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/09/828409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/09/828409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/09/828409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/09/828409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/09/828409/1"}}},"commit":{"parents":[{"commit":"c37a7335a2a023faf3f4fc9ea4cbf038a4af3258","subject":"Prepare a new Wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c37a7335a2a023faf3f4fc9ea4cbf038a4af3258"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2021-08-26 13:33:19.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-02-09 14:23:20.000000000","tz":0},"subject":"Add report_inteval parameter in ml2 baremetal","message":"Add report_inteval parameter in ml2 baremetal\n\nAdd the parameter $report_interval to the ml2\nnetworking-baremetal manifest. By default the global\nreport_interval as defined in neutron.conf is used.\n\nIn some cases it makes sense to change the report_interval\nspecifically for ironic-neutron-agent.\n\nRelated-Bug: #1940838\nChange-Id: I6fc0eb085c09fb1e24ff09d4f8199d2cf57dd209\n(cherry picked from commit dae33092a120655dfe4b9a550ec29a8c2dba73a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fc8919ba6cdeaa93e7883e228d89dc3fec797e1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fc8919ba6cdeaa93e7883e228d89dc3fec797e1f"}]},"branch":"refs/heads/stable/wallaby"},"41bcf66eafc092890cdd6d2560d02642c5a55761":{"kind":"REWORK","_number":2,"created":"2022-02-09 14:38:35.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/09/828409/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/09/828409/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/09/828409/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/09/828409/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/09/828409/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/09/828409/2"}}},"commit":{"parents":[{"commit":"c37a7335a2a023faf3f4fc9ea4cbf038a4af3258","subject":"Prepare a new Wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c37a7335a2a023faf3f4fc9ea4cbf038a4af3258"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2021-08-26 13:33:19.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-02-09 14:38:05.000000000","tz":60},"subject":"Add report_inteval parameter in ml2 baremetal","message":"Add report_inteval parameter in ml2 baremetal\n\nAdd the parameter $report_interval to the ml2\nnetworking-baremetal manifest. By default the global\nreport_interval as defined in neutron.conf is used.\n\nIn some cases it makes sense to change the report_interval\nspecifically for ironic-neutron-agent.\n\nConflicts:\n  manifests/agents/ml2/networking_baremetal.pp\n  spec/classes/neutron_agents_ml2_networking_baremetal_spec.rb\n\nRelated-Bug: #1940838\nChange-Id: I6fc0eb085c09fb1e24ff09d4f8199d2cf57dd209\n(cherry picked from commit dae33092a120655dfe4b9a550ec29a8c2dba73a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/41bcf66eafc092890cdd6d2560d02642c5a55761"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/41bcf66eafc092890cdd6d2560d02642c5a55761"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
