)]}'
{"id":"openstack%2Fpuppet-neutron~828554","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fvictoria~I6fc0eb085c09fb1e24ff09d4f8199d2cf57dd209","project":"openstack/puppet-neutron","branch":"stable/victoria","topic":"bug/1940838-stable/victoria","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2022-02-10 20:18:18.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-10 20:18:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6fc0eb085c09fb1e24ff09d4f8199d2cf57dd209","subject":"Add report_inteval parameter in ml2 baremetal","status":"MERGED","created":"2022-02-09 16:56:24.000000000","updated":"2022-02-10 20:19:48.000000000","submitted":"2022-02-10 20:18:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828554-bug/1940838-stable/victoria","cherry_pick_of_change":828409,"cherry_pick_of_patch_set":2,"meta_rev_id":"08f37a3e04059c742e67c99d8d3a85f0686c7b44","_number":828554,"virtual_id_number":828554,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-10 20:18:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-02-10 12:57:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-02-10 18:09:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-02-10 18:09:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-09 21:27:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-10 12:57:34.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-02-10 18:09:31.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"8e0dd52d173ec1f573bec80881aacf5ef7b42592","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-02-09 16:56:24.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"61f26ca0606124f5ae3be742ef7e96858a62506f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 21:27:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28471cad783c4abab6ab19e9fc2f8273 : SUCCESS in 4m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c46b17df18f4e2fa08658edb8e8e7a2 : SUCCESS in 4m 02s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/b478bc3890dd4deaa39eb642baef7e11 : SUCCESS in 18m 08s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7810364a257641e281142998e38c9aa1 : SUCCESS in 14m 39s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3aff8f5ff8e546ceb60e8f51e34ab9ab : FAILURE in 17m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0e16253259a41c3b222ac739de1ebd1 : FAILURE in 37m 20s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ae65851ec084480a6ce4fadec455d58 : SUCCESS in 1h 03m 16s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9604ec9dd6904af09302c2f6787e0cf4 : SUCCESS in 46m 23s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c341b1ddca947c7bdd0754538d571fc : SUCCESS in 43m 20s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/01402119c13d478faf6fcdc5d788307d : SUCCESS in 1h 14m 37s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7fd70472fec44fffac46ce2b6d972bfd : TIMED_OUT in 2h 02m 59s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/45f613c94c814135a81fcdae02d123ec : SUCCESS in 1h 25m 11s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/83d321e11fbd494c8fe589197e52f952 : SUCCESS in 1h 13m 45s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/d3e35433774041c292a5693b0f683eb2 : SUCCESS in 22m 37s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3ddbc768f704f81a14099576736c8e6 : SUCCESS in 20m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8a8e0942c684868914d30e2356ba902 : SUCCESS in 4m 26s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8e64f19ae90b44fc886d1d1b6c5a22c5 : SUCCESS in 2h 12m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d3d076975b26502e54df6a0ca3526af73d0f2380","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 12:56:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"367d99abd66593e5d9171d034a72250e0fb3bf2c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 12:56:55.000000000","message":"Topic bug/1940838-stable/wallaby-stable/victoria removed","accounts_in_message":[],"_revision_number":2},{"id":"16a9f44a40fe052fffb46871b4798ab4b93dd070","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 12:57:01.000000000","message":"Topic set to bug/1940838-stable/victoria","accounts_in_message":[],"_revision_number":2},{"id":"27f9a04fe219c2a19bde0ff32cae931b477ea07f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 12:57:34.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"066b09650e8561fc2b23cc1740f5f9335cae66ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 15:13:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c613812a11a3457baa726fb9cb48edc9 : SUCCESS in 4m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/764aea58110743069705939d969f5fc4 : SUCCESS in 3m 37s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/d160f9aff14a4d1e8f8078b2ef863f35 : SUCCESS in 20m 29s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a349ceaf3794fb09ba4b63f6b5ae624 : SUCCESS in 14m 39s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4c1bba0f39b4d20be44ff57aa207dc2 : FAILURE in 15m 05s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9294f8d17b8648d5ae72729fa24318c0 : FAILURE in 55m 11s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c31cc49b7d048e692fc620167e524d4 : SUCCESS in 49m 40s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/820b9c871a854e6aab537bf520687610 : SUCCESS in 45m 12s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88c22e6f8e194c03aee2172333ca1ac5 : SUCCESS in 36m 59s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2edcf8a8e5eb45ce8bbdd1efd706547b : SUCCESS in 1h 18m 58s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/059c5e9ee3594cfc967aeb140672927e : TIMED_OUT in 2h 02m 15s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/531b84c941644f09beb8811f71024acd : SUCCESS in 1h 25m 56s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e8b9184918c47ceb3f2eafe2615f1d6 : SUCCESS in 56m 27s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/59d313056abe424d8e7522a2798209d8 : SUCCESS in 22m 51s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7976e55272aa4218b4dc05850de6d232 : SUCCESS in 16m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ee6104b591448afa39a853191ceb05c : SUCCESS in 3m 10s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/964f53ce285f423c8d082b5c05ddda13 : SUCCESS in 2h 09m 40s","accounts_in_message":[],"_revision_number":2},{"id":"872abb348ce3080ded94e477757cf9c0da589989","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 15:23:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cb85f1d8c9c81d62e82def8de795c313f86fcdf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 17:48:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c899525118349d68b54a8a900bec720 : SUCCESS in 3m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b00e1ddfb1ce4e92810b7bec5b338285 : SUCCESS in 3m 33s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/375b9d4c2c00484893b332cd061e38d6 : SUCCESS in 14m 59s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2529404317524abebf637b0a6f3b0cb1 : SUCCESS in 14m 12s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4d7b7d363734651950e48246f8125d9 : FAILURE in 16m 30s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36757e6ae3ac40af8c725ba3144d4bb8 : SUCCESS in 49m 28s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/456fca019bc844df9cc33fd80db4e4c0 : SUCCESS in 53m 10s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8357ffe0e5e243cfb47e8b46d42cd47b : SUCCESS in 44m 39s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32287636e3f3496a96ee02dfd17c72ea : SUCCESS in 38m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a57b251834645ccb1966ae50f967991 : SUCCESS in 1h 19m 48s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b7f885612fb04270aec43d7dfc787eea : SUCCESS in 1h 38m 05s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/27df5e8fa72543e19c2a74b108d15f34 : SUCCESS in 1h 24m 56s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e82eba3576d3467d84c5f8954d23922c : SUCCESS in 1h 00m 35s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa07a7c4a6694775bae8a52d674ed0de : SUCCESS in 22m 33s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0d4a8e144224d149430319e9a9ee41d : SUCCESS in 17m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/040724b761dc436fa81819812fed2d3a : SUCCESS in 4m 13s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/56ff3e693d194c43b6ea89fd203bfc5e : SUCCESS in 2h 18m 05s","accounts_in_message":[],"_revision_number":2},{"id":"f53a4d2e41c945a8e603a4f0b92fdd852b29fbcb","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-10 18:09:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6db692a8c7cd872617ff9e3776046692b2c63b4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 18:11:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1b96a9e0729bcdb01a50bf831fa56208974a5d9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 20:18:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca09d585ee5546a0b572505f402bea7a : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5114371c0acf4775816a102d1fa1aea7 : SUCCESS in 3m 25s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0089e98a8d54d40b1db155b6332c802 : SUCCESS in 18m 52s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2d430847be94317a471bda0166ab14a : SUCCESS in 12m 12s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/67d41e10141b4fc286fe365f60521e81 : SUCCESS in 1h 04m 13s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4189e1953d754072b38ef32c21eba4bf : SUCCESS in 1h 22m 55s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/256087d0083c4011bce2b2a59a057f5b : SUCCESS in 1h 16m 03s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6507488e4eb248b8ba757cf56a7305ac : SUCCESS in 43m 06s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/7fd4d156c0f74d70a32c265c4fcd6eb0 : SUCCESS in 21m 03s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2bb940cdcc5469bbfccbebf665cfffd : SUCCESS in 16m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36344e69e7404718b8e7134685d68788 : SUCCESS in 4m 17s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/faf29beb364e40e984d0bb655ac09e89 : SUCCESS in 2h 06m 01s","accounts_in_message":[],"_revision_number":2},{"id":"4cd15779797fc7aae83bab7e37a4d3633d580c72","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 20:18:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"08f37a3e04059c742e67c99d8d3a85f0686c7b44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 20:19:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59f5faa996484c2fa1a71b786a028d9f : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"629dbe0f1e89bec38d02050f80d3dda30e20d170","revisions":{"59b17ccc5beada2a7a8f771614d9ce59704355f4":{"kind":"REWORK","_number":1,"created":"2022-02-09 16:56:24.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/54/828554/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/54/828554/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/828554/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/828554/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/828554/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/54/828554/1"}}},"commit":{"parents":[{"commit":"15a3ccaefcb32e7de53e32d4de9db15fca98d022","subject":"Add support for the resource_provider_default_hypervisor parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/15a3ccaefcb32e7de53e32d4de9db15fca98d022"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2021-08-26 13:33:19.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-02-09 16:56:24.000000000","tz":0},"subject":"Add report_inteval parameter in ml2 baremetal","message":"Add report_inteval parameter in ml2 baremetal\n\nAdd the parameter $report_interval to the ml2\nnetworking-baremetal manifest. By default the global\nreport_interval as defined in neutron.conf is used.\n\nIn some cases it makes sense to change the report_interval\nspecifically for ironic-neutron-agent.\n\nConflicts:\n  manifests/agents/ml2/networking_baremetal.pp\n  spec/classes/neutron_agents_ml2_networking_baremetal_spec.rb\n\nRelated-Bug: #1940838\nChange-Id: I6fc0eb085c09fb1e24ff09d4f8199d2cf57dd209\n(cherry picked from commit dae33092a120655dfe4b9a550ec29a8c2dba73a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/59b17ccc5beada2a7a8f771614d9ce59704355f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/59b17ccc5beada2a7a8f771614d9ce59704355f4"}]},"branch":"refs/heads/stable/victoria"},"629dbe0f1e89bec38d02050f80d3dda30e20d170":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-10 12:56:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/828554/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/54/828554/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/828554/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/828554/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/54/828554/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/54/828554/2"}}},"commit":{"parents":[{"commit":"15a3ccaefcb32e7de53e32d4de9db15fca98d022","subject":"Add support for the resource_provider_default_hypervisor parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/15a3ccaefcb32e7de53e32d4de9db15fca98d022"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2021-08-26 13:33:19.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-10 12:56:46.000000000","tz":0},"subject":"Add report_inteval parameter in ml2 baremetal","message":"Add report_inteval parameter in ml2 baremetal\n\nAdd the parameter $report_interval to the ml2\nnetworking-baremetal manifest. By default the global\nreport_interval as defined in neutron.conf is used.\n\nIn some cases it makes sense to change the report_interval\nspecifically for ironic-neutron-agent.\n\nConflicts:\n  manifests/agents/ml2/networking_baremetal.pp\n  spec/classes/neutron_agents_ml2_networking_baremetal_spec.rb\n\nRelated-Bug: #1940838\nChange-Id: I6fc0eb085c09fb1e24ff09d4f8199d2cf57dd209\n(cherry picked from commit dae33092a120655dfe4b9a550ec29a8c2dba73a5)\n(cherry picked from commit 41bcf66eafc092890cdd6d2560d02642c5a55761)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/629dbe0f1e89bec38d02050f80d3dda30e20d170"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/629dbe0f1e89bec38d02050f80d3dda30e20d170"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
