)]}'
{"id":"openstack%2Fpuppet-neutron~828555","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fussuri~I6fc0eb085c09fb1e24ff09d4f8199d2cf57dd209","project":"openstack/puppet-neutron","branch":"stable/ussuri","topic":"bug/1940838-stable/ussuri","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2022-02-11 16:56:55.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-11 11:52:35.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I6fc0eb085c09fb1e24ff09d4f8199d2cf57dd209","subject":"Add report_inteval parameter in ml2 baremetal","status":"MERGED","created":"2022-02-09 16:56:40.000000000","updated":"2022-02-11 16:59:20.000000000","submitted":"2022-02-11 16:56:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828555-bug/1940838-stable/ussuri","cherry_pick_of_change":828554,"cherry_pick_of_patch_set":2,"meta_rev_id":"0f61d4eb078b27c7473663079419747333baac00","_number":828555,"virtual_id_number":828555,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-11 16:56:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-02-11 05:05:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-02-10 18:10:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-02-11 05:05:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-09 21:20:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-10 12:58:49.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2022-02-10 18:10:13.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-02-11 05:05:44.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"cd6da7c21b0c58c55a01860f87cf9d7e0621437d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-02-09 16:56:40.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"b83597270041134752ed51c448eb1568c5afd767","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 21:20:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13cd7081923e4eb4bd89394a1c03ca9e : SUCCESS in 4m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6bf58158baa24e9ebf6ace837fa21d27 : SUCCESS in 5m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05c69acadb4944b8bff309790adea260 : SUCCESS in 3m 31s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/fca2a43a62ad407d82fc1b9cf078baa0 : SUCCESS in 20m 52s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/f4a17129e8c44c55a7eeacc7da456eba : SUCCESS in 20m 18s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ca817c26ee34daaaa2f0e54fc14d544 : SUCCESS in 15m 39s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/720629cfccb249dd92d31b348455f4bb : FAILURE in 13m 57s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2dc228ac967341b98f719dc1deb3f326 : FAILURE in 38m 22s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81792f21d8c04dc79f9464c7e8df2d6a : SUCCESS in 54m 13s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3cffa75194634c4f84b81c781ef5509e : SUCCESS in 47m 56s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/fa84958cb7ae466691f1a8ca9bc7daff : SUCCESS in 38m 30s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5a03a078d50b4ed3b84fc73db0706241 : SUCCESS in 1h 02m 21s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5439c55d343a4910ae868684109fb92c : SUCCESS in 1h 13m 23s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/68677f8374e84931a2c560ff10c5ca1d : SUCCESS in 1h 10m 46s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e9c74c88a36843b6ba80af176fe622b8 : SUCCESS in 49m 40s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/cfba9a830b1347dc9b14a9c8bb9063c9 : FAILURE in 30m 40s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a69ff710e5004852bae4193cca6dedf4 : FAILURE in 8m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc07cb6869844533891dc265236dfe73 : SUCCESS in 5m 27s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9f9dad502592434ebcfcbaccee070e83 : SUCCESS in 1h 53m 49s","accounts_in_message":[],"_revision_number":1},{"id":"c4c1ca3c775076d9c703f885f87b196d8183ce27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 12:57:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"58774021f427b58f6c8ea7601aab3dbb8dfc8f0c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 12:57:59.000000000","message":"Topic bug/1940838-stable/victoria-stable/ussuri removed","accounts_in_message":[],"_revision_number":2},{"id":"b736ca527b2c512f6e493625544f395797ad62c0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 12:58:03.000000000","message":"Topic set to bug/1940838-stable/ussuri","accounts_in_message":[],"_revision_number":2},{"id":"8c86eca37bfe5fbc5a8ac2b6470c587093a00424","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 12:58:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a481a8521760394a5ca7d1755b42d55b17c0714b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 15:13:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7dd5be275cd64f2cb76f0bacefe16e43 : SUCCESS in 3m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6cf548bf7dc46828e254e2074566afb : SUCCESS in 3m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14c54475b9d1442e93b9923e8e2af4a2 : SUCCESS in 3m 12s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/dc5b681a3f8b43b6af9d3e92832ff54f : SUCCESS in 18m 11s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/1aba62a9fcdd4a679b41ee6b04608ad1 : SUCCESS in 19m 12s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29f5eb156e244abebdd591ab3c580d9e : SUCCESS in 12m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf077dfc40f44d9db4e19ff92a61d241 : FAILURE in 14m 01s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/00fb5841fa8c4d59bd0783553ea5f010 : SUCCESS in 50m 11s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03f9524db533475b8543dd546035d31d : SUCCESS in 50m 36s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1579f9cce91e4e05ae461744c6b670e0 : SUCCESS in 46m 39s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/40773764cd2249698f03ab5fb37ac101 : SUCCESS in 41m 35s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/33520ab6ddfa42d3913d6d85c3179932 : SUCCESS in 57m 26s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fb55a2b8a4eb431f99d88d1789879034 : SUCCESS in 1h 05m 01s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0c89ae710fc14448b04fdca0ed589c0d : SUCCESS in 1h 07m 50s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/077f01182e544b878d0983e517163e31 : SUCCESS in 47m 22s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/eec439b57faf4bb2b48ed2e052c76454 : FAILURE in 22m 03s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4d0a4ed370948779f34dc217fab83ef : FAILURE in 6m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/472a85b646fe4d6ab64495f4a2e83028 : SUCCESS in 3m 54s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/046918dab29c4b8da6f882978bce3659 : SUCCESS in 2h 07m 59s","accounts_in_message":[],"_revision_number":2},{"id":"0ad9a0300d02d3776e287f6f8e5b7fb5777b2248","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-10 18:10:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1f45797cce4829a0ff5bcd63609222e3dacd19d4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-11 05:05:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3831fe6178fb68dd877bcdafddc0029b821aeb58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 05:07:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b07022f265f3d747f994ea449a94d511e58521e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 06:55:52.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f7908ed404b44378c976f44cb463093 : SUCCESS in 2m 58s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20a927add6984ea4bf9c05df4972aa34 : SUCCESS in 3m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e92f31a4710c457a9b0e820fdeef84b1 : SUCCESS in 3m 13s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/c147a850557c4fb3a7ccbb1844c1be42 : SUCCESS in 15m 29s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/96754838e7684f4a9ee431b805be28c8 : SUCCESS in 15m 54s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/632de9e1daec40cfba0641cce2805f06 : SUCCESS in 9m 18s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/57aae47e7c434ab4ac1efcb2c899123b : SUCCESS in 1h 01m 19s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc05934c18f44354854d8fc1b84ffdf7 : SUCCESS in 1h 14m 30s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3dda650baed419f9c947c686e0f0ba5 : SUCCESS in 1h 02m 47s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/436ebf4155ab44ad8f98379a750386c3 : FAILURE in 26m 56s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/92d8414680574d69aa0422eedd61273a : FAILURE in 26m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed7e2051996e411d89dbafa5dbcbd612 : SUCCESS in 3m 32s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2db88277664b4328b7aee1f3a0ad60e9 : SUCCESS in 1h 46m 24s","accounts_in_message":[],"_revision_number":2},{"id":"1e33ff55a8250219050a080e48ef5e893833eb30","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-11 11:52:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"317e449555cd602af2e6c72b156d6d9c376617be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 14:35:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d40b4f21660b42988137001be2759949 : SUCCESS in 3m 09s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/957cdc248d33498f9deec542295fa26d : SUCCESS in 3m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/499cc678f68d4edd9b82b2c90391a8a2 : SUCCESS in 3m 10s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/732b796dc9384389b0016525715d460a : SUCCESS in 20m 16s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/55d9af8ba1d94eb29846b4d026191be0 : SUCCESS in 20m 16s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ae96bdd43554770924f9baa5d200673 : SUCCESS in 12m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6cd73c0231c742789eeb92f398d4ce04 : FAILURE in 15m 31s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/93ce11a120f24e5a9a6aa0d6a8e24709 : SUCCESS in 54m 33s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0ed86647298481abfaaa4e3fde4bc2d : SUCCESS in 53m 26s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e434b3a945114e279989ec4762fd57d7 : SUCCESS in 49m 16s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/488f65b489db4f64b9cd93a388cf2a95 : SUCCESS in 48m 10s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a038bf79eb04363b90b069a0e0cf084 : SUCCESS in 1h 03m 21s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c5e5eac9aa374ae09771fc0c1bae816f : SUCCESS in 1h 14m 27s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ce536612e0e74d70aa2537b981564152 : SUCCESS in 1h 24m 58s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/af95fa91050e45f4adcddbf942230bb5 : SUCCESS in 54m 06s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/732f2cfc6e684eac9539ea2c45186ea5 : FAILURE in 28m 25s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/600db8cfa8c44e308d67f09739b73ba0 : FAILURE in 6m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/625199c41ac9499a83565487feeb256f : SUCCESS in 4m 01s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/037934f152a041288b382b39c58750e4 : SUCCESS in 2h 35m 10s","accounts_in_message":[],"_revision_number":2},{"id":"e2cd830c90ba66f593cde0cfd6612f9f0c1edd9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 14:37:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"479d0f2f7d6e2b5083fe787135e5d7f20cb2bc3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 16:56:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86d65bfa0d2b4d8ca8a03f733e3517c9 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76ed674044fe4754972a44c3c9916851 : SUCCESS in 3m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ebae594ddacc4a938b3169abe5f4b155 : SUCCESS in 4m 02s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/8705b6f99e794c47b8b78690a97a85e0 : SUCCESS in 19m 42s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/09dc7502b0864dbf8b7a224a70b62da6 : SUCCESS in 21m 09s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a2fdea9040942258548a94904e1be2e : SUCCESS in 14m 30s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/15c05209df874a5880f3fb43edbdd7ad : SUCCESS in 1h 02m 36s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2b6292fa6e4d4fc6a1d0289e2ee8cc42 : SUCCESS in 1h 08m 58s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f8c7019466ff44f7b93387b39a509313 : SUCCESS in 1h 14m 47s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/14499e231ea84341a9d852e66645acaf : SUCCESS in 1h 01m 27s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/958af2bbb63349ffbcf0f80a87fe39a9 : FAILURE in 32m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ed0ee1ec40648c79ef2215c1e985ba3 : SUCCESS in 3m 53s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/586c373a58ed4b2c9bd19aac7f454ef1 : SUCCESS in 2h 13m 20s","accounts_in_message":[],"_revision_number":2},{"id":"c0dceedbed099868c366c3df8af9d7fe35ad0891","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 16:56:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0f61d4eb078b27c7473663079419747333baac00","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 16:59:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8d9cf24d509449e84ed6795bc20c90e : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b9f3a4baa62d72b356d3307a380f2aa1f3f7703d","revisions":{"2698deb79159e508ef9e30e5ef0cf82f8058ec69":{"kind":"REWORK","_number":1,"created":"2022-02-09 16:56:40.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/55/828555/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/55/828555/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/55/828555/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/55/828555/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/55/828555/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/55/828555/1"}}},"commit":{"parents":[{"commit":"038626a8e03cf19c95189a4d28335baf23fbbe11","subject":"Prepare the final stable/ussuri release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/038626a8e03cf19c95189a4d28335baf23fbbe11"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2021-08-26 13:33:19.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-02-09 16:56:40.000000000","tz":0},"subject":"Add report_inteval parameter in ml2 baremetal","message":"Add report_inteval parameter in ml2 baremetal\n\nAdd the parameter $report_interval to the ml2\nnetworking-baremetal manifest. By default the global\nreport_interval as defined in neutron.conf is used.\n\nIn some cases it makes sense to change the report_interval\nspecifically for ironic-neutron-agent.\n\nConflicts:\n  manifests/agents/ml2/networking_baremetal.pp\n  spec/classes/neutron_agents_ml2_networking_baremetal_spec.rb\n\nRelated-Bug: #1940838\nChange-Id: I6fc0eb085c09fb1e24ff09d4f8199d2cf57dd209\n(cherry picked from commit dae33092a120655dfe4b9a550ec29a8c2dba73a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2698deb79159e508ef9e30e5ef0cf82f8058ec69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2698deb79159e508ef9e30e5ef0cf82f8058ec69"}]},"branch":"refs/heads/stable/ussuri"},"b9f3a4baa62d72b356d3307a380f2aa1f3f7703d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-10 12:57:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/828555/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/55/828555/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/55/828555/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/55/828555/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/55/828555/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/55/828555/2"}}},"commit":{"parents":[{"commit":"038626a8e03cf19c95189a4d28335baf23fbbe11","subject":"Prepare the final stable/ussuri release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/038626a8e03cf19c95189a4d28335baf23fbbe11"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2021-08-26 13:33:19.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-10 12:57:51.000000000","tz":0},"subject":"Add report_inteval parameter in ml2 baremetal","message":"Add report_inteval parameter in ml2 baremetal\n\nAdd the parameter $report_interval to the ml2\nnetworking-baremetal manifest. By default the global\nreport_interval as defined in neutron.conf is used.\n\nIn some cases it makes sense to change the report_interval\nspecifically for ironic-neutron-agent.\n\nConflicts:\n  manifests/agents/ml2/networking_baremetal.pp\n  spec/classes/neutron_agents_ml2_networking_baremetal_spec.rb\n\nRelated-Bug: #1940838\nChange-Id: I6fc0eb085c09fb1e24ff09d4f8199d2cf57dd209\n(cherry picked from commit dae33092a120655dfe4b9a550ec29a8c2dba73a5)\n(cherry picked from commit 41bcf66eafc092890cdd6d2560d02642c5a55761)\n(cherry picked from commit 629dbe0f1e89bec38d02050f80d3dda30e20d170)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b9f3a4baa62d72b356d3307a380f2aa1f3f7703d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b9f3a4baa62d72b356d3307a380f2aa1f3f7703d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
