)]}'
{"id":"openstack%2Fpuppet-neutron~828556","triplet_id":"openstack%2Fpuppet-neutron~stable%2Ftrain~I6fc0eb085c09fb1e24ff09d4f8199d2cf57dd209","project":"openstack/puppet-neutron","branch":"stable/train","topic":"bug/1940838-stable/train","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2022-02-14 19:43:12.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-14 19:43:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6fc0eb085c09fb1e24ff09d4f8199d2cf57dd209","subject":"Add report_inteval parameter in ml2 baremetal","status":"MERGED","created":"2022-02-09 16:56:59.000000000","updated":"2022-02-14 19:45:42.000000000","submitted":"2022-02-14 19:43:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828556-bug/1940838-stable/train","cherry_pick_of_change":828409,"cherry_pick_of_patch_set":2,"meta_rev_id":"d1ddf532d02a7f206347b901d08800c9303154e3","_number":828556,"virtual_id_number":828556,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-14 19:43:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-02-10 18:09:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-02-14 17:51:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-02-14 17:51:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-09 21:34:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-10 18:09:55.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-02-14 17:51:26.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"ec61c9dcfb358b8852b97fa96698ff84a13f7186","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-02-09 16:56:59.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.\n\nThe following files contain Git conflicts:\n* manifests/agents/ml2/networking_baremetal.pp","accounts_in_message":[],"_revision_number":1},{"id":"7b130d3e2ce6d72c8047e55035dffbc939331e01","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-02-09 17:22:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f47277222ca6436ec3a09b5c164b8dd9ad0a3f87","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-02-09 17:23:09.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"465e37421c02bf3a835f762216cf5100774f3b31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 21:34:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e39819ff20a4450a8b693ff4b02275ac : SUCCESS in 5m 03s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5f33ab409c24425a0b384ffb6891a95 : SUCCESS in 6m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74380937b69c4bcf9b1d14c0865296a9 : SUCCESS in 5m 37s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/1794614d92f3449893be0eacbf21409b : SUCCESS in 34m 41s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/8592c93ca9db4cc899ba51bad0f002f3 : SUCCESS in 19m 28s\n- puppet-openstack-unit-6.4-centos-8-stream https://zuul.opendev.org/t/openstack/build/d32844ff0a394eb6a27363c8af034298 : SUCCESS in 19m 17s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b5264c80db546b19806f891d4857f4a : SUCCESS in 16m 48s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6968039befff41bbbe5ddcd5c3900ee3 : FAILURE in 18m 45s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/3943f92b00564acba7e85305fef5ce20 : SUCCESS in 1h 07m 37s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/57577888780e4b88ada44f15f8a4b568 : SUCCESS in 1h 16m 33s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9fcda2f9cad14a188dbc05d763aaed69 : SUCCESS in 1h 12m 45s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/0fee476c5675401a8741bedcab20075e : SUCCESS in 49m 52s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/71180968cbfe49409e8b3dd832222442 : SUCCESS in 45m 52s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f81cb939557645429dc41e991a5da2c0 : SUCCESS in 46m 22s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4ef8b4cdba34895ba360c679d6405d1 : SUCCESS in 40m 38s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/aa6770904f3b40acb7bc4bed7e84c55a : SUCCESS in 36m 37s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b81fef7ad98d4f67ac9ac5e7e820a598 : SUCCESS in 58m 11s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a26f7681120b4859b67db52158731edd : SUCCESS in 1h 01m 54s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/33adcae54d344b2983e69574615d7b1b : SUCCESS in 1h 00m 18s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0ebcec0629c34ad9a074024729ad75c6 : SUCCESS in 48m 57s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6f1f48a4df334683a927502d3458ca3a : SUCCESS in 25m 53s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3aa5d079ce2a4e4ca0bc9e58d1f2f55b : FAILURE in 9m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7368cbaee3ea48e685b3b38f8405ee08 : SUCCESS in 4m 16s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/36900e6c40854b19a8600f06964df014 : SUCCESS in 1h 39m 51s","accounts_in_message":[],"_revision_number":2},{"id":"724dfc3e61c71d45efda08eb8031c42b77a03159","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 13:01:58.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1e0b5b187d5a571836855bba8798349cbeda0661","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 14:39:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/daa996458fb84eceb88a433b79e690f6 : SUCCESS in 5m 49s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23c8081c196a44d187fab321fc73588e : SUCCESS in 4m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3468fee2059466d98307577a944d9ba : SUCCESS in 4m 35s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/43580d7e7f3a44bf964b24befcc290d9 : SUCCESS in 34m 18s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/26f75ab6ec344f6f87addb39bacad4f3 : SUCCESS in 19m 29s\n- puppet-openstack-unit-6.4-centos-8-stream https://zuul.opendev.org/t/openstack/build/1250a60aecd2455184c682e25040cf11 : SUCCESS in 19m 11s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5165131c4b34626b3cc48b77ad1a528 : SUCCESS in 13m 39s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1a698bbdfc24e17973caaef492eec4c : FAILURE in 16m 28s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/2f9e25f300cf47dfae66c1f566694b0c : SUCCESS in 1h 10m 34s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/056df756f7ad46a2a84c6e0f133ad4a5 : SUCCESS in 1h 05m 18s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/5c52fc5b4cfe44c594ece62f6fa4cdea : SUCCESS in 1h 03m 47s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/0479f3fde5a04b55bebb02d96e6c5052 : SUCCESS in 50m 46s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5c84362a00524f208c9e044d54b3b9bf : SUCCESS in 52m 09s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/728188f177874a8caf6a3f1121178750 : SUCCESS in 45m 47s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1b6c28bc3b647338c944b95b66ee621 : SUCCESS in 36m 55s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d39187525cd64ffeb54e3042cad4961c : SUCCESS in 29m 46s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/660115ba6b9849d3a2ae010a620e5de5 : SUCCESS in 1h 00m 33s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/10ed0e8183394cdaa2d102c82338e4d2 : SUCCESS in 59m 14s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e89533cc7af40ebae507a6e9056a765 : SUCCESS in 59m 02s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4908b6caac994fdfa9889b71e42b0587 : SUCCESS in 45m 04s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c6778a6c87a547aab8ec446a5e6bf02a : SUCCESS in 20m 27s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/660f250ee621477ca41f8bcb575dfe72 : FAILURE in 7m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c5da8c512ff48859e51022bc30ddb4d : SUCCESS in 4m 06s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/34eb50068f6544c08321dcee10b02278 : SUCCESS in 1h 33m 14s","accounts_in_message":[],"_revision_number":3},{"id":"79580e6fadb01496e4e6dde81e526361cc4f83fb","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-10 18:09:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0a9e1754ab0f45a731dbf115e49977e63d271967","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-02-14 17:51:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cf1997e15a48482a8a5f1533535bcfc63da56fad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 17:53:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2aca3d276201d81a51739bc6d78f3b861bc6c1d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 19:43:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f68296dc5e2e46aea8f823c6cef7fd97 : SUCCESS in 5m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/792b2305ffb944cd9e6fbd37a8f6afe0 : SUCCESS in 6m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85fd192d3be640dc95af49269bb07f67 : SUCCESS in 5m 54s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/7ac740575b4a49cfb358ca9ed1adc6be : SUCCESS in 30m 56s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c62e0f329e34db2a624bdb9c442c2cd : SUCCESS in 15m 11s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/9831674840134eb9a82eedb6e148bbe2 : SUCCESS in 58m 12s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/135aa93fc11540e88d9ddb9d43290777 : SUCCESS in 1h 10m 15s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/492670c490a04ee2928b0eba71b7ded4 : SUCCESS in 1h 09m 39s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/6a4438460d6d4ea2ac887f16116e764b : SUCCESS in 51m 39s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/fa6c9f43f02e43d38ee98cb6d7597a7c : SUCCESS in 22m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c91da40c1839431390917750684603e2 : SUCCESS in 4m 27s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/32191d2b614048d1a53e5caf0375fe51 : SUCCESS in 1h 48m 05s","accounts_in_message":[],"_revision_number":3},{"id":"e798646e0312308fd394909eca8e20495895326d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 19:43:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d1ddf532d02a7f206347b901d08800c9303154e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 19:45:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd0ba6b9f14b4849adefa086cd880ee3 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b793bbfc699ac35b5c789410fe82bdfc1cd3d3d9","revisions":{"4b23a385735015364696625cabe678d05c5775b9":{"kind":"REWORK","_number":1,"created":"2022-02-09 16:56:59.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/56/828556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/56/828556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/828556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/828556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/828556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/56/828556/1"}}},"commit":{"parents":[{"commit":"914b228944a959bddc798e986759d9a57d006428","subject":"Add support for the interface parameter in authtoken middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/914b228944a959bddc798e986759d9a57d006428"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2021-08-26 13:33:19.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-02-09 16:56:59.000000000","tz":0},"subject":"Add report_inteval parameter in ml2 baremetal","message":"Add report_inteval parameter in ml2 baremetal\n\nAdd the parameter $report_interval to the ml2\nnetworking-baremetal manifest. By default the global\nreport_interval as defined in neutron.conf is used.\n\nIn some cases it makes sense to change the report_interval\nspecifically for ironic-neutron-agent.\n\nConflicts:\n  manifests/agents/ml2/networking_baremetal.pp\n  spec/classes/neutron_agents_ml2_networking_baremetal_spec.rb\n\nRelated-Bug: #1940838\nChange-Id: I6fc0eb085c09fb1e24ff09d4f8199d2cf57dd209\n(cherry picked from commit dae33092a120655dfe4b9a550ec29a8c2dba73a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4b23a385735015364696625cabe678d05c5775b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4b23a385735015364696625cabe678d05c5775b9"}]},"branch":"refs/heads/stable/train"},"302f06be61fb2f24519ac71fcc410496ce02821f":{"kind":"REWORK","_number":2,"created":"2022-02-09 17:22:48.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/56/828556/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/56/828556/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/828556/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/828556/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/828556/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/56/828556/2"}}},"commit":{"parents":[{"commit":"914b228944a959bddc798e986759d9a57d006428","subject":"Add support for the interface parameter in authtoken middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/914b228944a959bddc798e986759d9a57d006428"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2021-08-26 13:33:19.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-02-09 17:22:19.000000000","tz":60},"subject":"Add report_inteval parameter in ml2 baremetal","message":"Add report_inteval parameter in ml2 baremetal\n\nAdd the parameter $report_interval to the ml2\nnetworking-baremetal manifest. By default the global\nreport_interval as defined in neutron.conf is used.\n\nIn some cases it makes sense to change the report_interval\nspecifically for ironic-neutron-agent.\n\nConflicts:\n  manifests/agents/ml2/networking_baremetal.pp\n  spec/classes/neutron_agents_ml2_networking_baremetal_spec.rb\n\nRelated-Bug: #1940838\nChange-Id: I6fc0eb085c09fb1e24ff09d4f8199d2cf57dd209\n(cherry picked from commit dae33092a120655dfe4b9a550ec29a8c2dba73a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/302f06be61fb2f24519ac71fcc410496ce02821f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/302f06be61fb2f24519ac71fcc410496ce02821f"}]},"branch":"refs/heads/stable/train"},"b793bbfc699ac35b5c789410fe82bdfc1cd3d3d9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-02-10 13:01:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/56/828556/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/56/828556/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/828556/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/828556/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/828556/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/56/828556/3"}}},"commit":{"parents":[{"commit":"914b228944a959bddc798e986759d9a57d006428","subject":"Add support for the interface parameter in authtoken middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/914b228944a959bddc798e986759d9a57d006428"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2021-08-26 13:33:19.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-10 13:01:45.000000000","tz":540},"subject":"Add report_inteval parameter in ml2 baremetal","message":"Add report_inteval parameter in ml2 baremetal\n\nAdd the parameter $report_interval to the ml2\nnetworking-baremetal manifest. By default the global\nreport_interval as defined in neutron.conf is used.\n\nIn some cases it makes sense to change the report_interval\nspecifically for ironic-neutron-agent.\n\nConflicts:\n  manifests/agents/ml2/networking_baremetal.pp\n  spec/classes/neutron_agents_ml2_networking_baremetal_spec.rb\n\nBackport note to stable/train:\nResolved additional conflicts caused by\ndfeccd674988f34edea5ca7b95bdeea827bab000\n\nRelated-Bug: #1940838\nChange-Id: I6fc0eb085c09fb1e24ff09d4f8199d2cf57dd209\n(cherry picked from commit dae33092a120655dfe4b9a550ec29a8c2dba73a5)\n(cherry picked from commit 41bcf66eafc092890cdd6d2560d02642c5a55761)\n(cherry picked from commit 629dbe0f1e89bec38d02050f80d3dda30e20d170)\n(cherry picked from commit b9f3a4baa62d72b356d3307a380f2aa1f3f7703d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b793bbfc699ac35b5c789410fe82bdfc1cd3d3d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b793bbfc699ac35b5c789410fe82bdfc1cd3d3d9"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
