)]}'
{"id":"openstack%2Fpuppet-neutron~829465","triplet_id":"openstack%2Fpuppet-neutron~master~I3711ce3654a0253dda142def1f1a07b286d06c08","project":"openstack/puppet-neutron","branch":"master","topic":"oslo.cache.2.10.0","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-20 11:44:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3711ce3654a0253dda142def1f1a07b286d06c08","subject":"Add cache client retry options for the pymemcache backend","status":"MERGED","created":"2022-02-16 02:44:20.000000000","updated":"2022-02-20 11:46:07.000000000","submitted":"2022-02-20 11:44:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829465-oslo.cache.2.10.0","meta_rev_id":"628c2123aae89de2518bd7646a57293ead4d5200","_number":829465,"virtual_id_number":829465,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-20 11:44:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-02-16 15:13:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-02-20 09:48:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-02-16 11:18:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-02-20 09:48:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-16 05:21:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-16 11:18:09.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2022-02-16 15:13:42.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-02-20 09:48:14.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"77354652466a1faeb285dc634d99f9666b202fed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-16 02:44:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ba2530617282f082719987f42460e074530ad44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 05:21:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/773ee88cd1b041769fe61127ff4da209 : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/918356203a054c86844b1e1570fbec1b : SUCCESS in 5m 14s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9732f7340644348899c39ca422914b9 : SUCCESS in 4m 17s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/34fae6f51e2a4065a0b922b75a353bae : SUCCESS in 23m 23s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/e1562202bbf548c3b36a5441863c52bd : SUCCESS in 20m 31s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a7ca1ecdf404ec09b505bf86b5cd7c9 : SUCCESS in 16m 00s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07f85080dc3542bf912681e15d65720e : SUCCESS in 17m 31s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00fb9f7756d64a8aa585ee8f3bc3470f : SUCCESS in 16m 06s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6605fa826faf43a3b470ba17e7cf7b6e : FAILURE in 33m 14s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3446dcb7dea498495f14b4152b7f060 : FAILURE in 38m 01s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1775162819347af901653d84742f4a5 : FAILURE in 33m 17s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61d4ef0da476491f8eaded3522ac02f7 : FAILURE in 26m 56s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5c750debe3f46a6ba6682e2fc241135 : SUCCESS in 24m 18s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a3785d5ae5884564913bb116c0a909cb : SUCCESS in 1h 07m 27s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/182276ce12da4335be11a2672ec45c85 : SUCCESS in 1h 49m 40s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/89c01f89af724c15a8b148ad80e8522f : SUCCESS in 1h 39m 55s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/67117702e4ae4c91955573c768a4901b : SUCCESS in 1h 06m 44s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a37d0d6c2a04c238df5159931e409bf : SUCCESS in 22m 57s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab1e24e438544d3091a91adaeebdc227 : SUCCESS in 51m 17s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/799bca3445844b7685c542a7d18542cb : SUCCESS in 1h 11m 32s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/14822f5622ea422e9d7195d723c721aa : SUCCESS in 1h 08m 51s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3056e1844a3947bd85c841e15b86c15a : SUCCESS in 46m 47s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e47ec0d01754f8e8b1802a530963806 : SUCCESS in 20m 52s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/9b6a2bfe30b640f2b46bf8bf86c0c9de : SUCCESS in 25m 22s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ec55258a0464926912bd6488646376b : FAILURE in 12m 39s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c356f5fec1e4fc5941f1cc7c19a4ba7 : SUCCESS in 20m 30s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f384d79ab524a588181701e90411608 : SUCCESS in 20m 01s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/e282dcb0a3b3424c9529fdb2b25f7528 : SUCCESS in 2h 24m 55s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/8fff13e0113d45e08931b7d39978d6ff : SUCCESS in 50m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/842cee4289734c1a9ed40718d9fa1965 : SUCCESS in 4m 32s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e5ddb300b7564205b27269328e981b8f : SUCCESS in 1h 16m 18s","accounts_in_message":[],"_revision_number":1},{"id":"294124a765c94a60c859754d8f8cd4896e31eb6e","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-02-16 11:18:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"48e56f3e7c1ec66cb75d714c87dc2fe738f038c5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-02-16 15:13:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b65c421d6d7b50959b78f5b4d9bc917d043a2b2a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-20 09:48:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e2bde11e15ae3b20e0c0834c083010037df8a278","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 09:49:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe4bee2c507f3be68ae6a258c427c139f8977617","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 11:44:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d52e8e1d054425d80dc7bc2111703a3 : SUCCESS in 3m 52s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9b03a906c4f48599154299780664450 : SUCCESS in 3m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e7abee389c842499a5db2512bc3762a : SUCCESS in 3m 46s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9555653c6b9343528af9d729b366db7b : SUCCESS in 16m 49s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/c0919ef2a81549f39794fc45d5955447 : SUCCESS in 16m 15s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b747ccaeec0d43a1a21db0308b7e4bf2 : SUCCESS in 15m 05s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd3997db004c45948f9b5cd3cf03b085 : SUCCESS in 15m 27s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/42d528af6d39469f97e920e5d7dd89ca : SUCCESS in 1h 16m 35s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a6f500a2fae4e7886af9a9337a3171d : SUCCESS in 1h 37m 40s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/03a1973c21cb4dfe955abc1df361ffcd : SUCCESS in 1h 19m 04s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/07c437f5922a49aebce11abfbcf3863b : SUCCESS in 1h 00m 30s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/d306f98181884dc5a70ed4b6704084dc : SUCCESS in 21m 56s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/b3b2ab17edbd41da9122ac9e9d8fee09 : SUCCESS in 1h 54m 22s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/8aacff956dd84121823595bdd82d04fa : SUCCESS in 49m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1cc5fe92e8a42fd9cd40dac91c64f4a : SUCCESS in 4m 02s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b7c994893829461db293a949a06713cd : SUCCESS in 1h 05m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e0e61a8996b37bc0c1d2b2d4cf36a56fe9bc2779","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 11:44:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"628c2123aae89de2518bd7646a57293ead4d5200","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 11:46:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae9a9eeb4e484af592a8374c027bdf49 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0ce4c51623ae0fe67d86aa64deb3403abbc4e799","revisions":{"0ce4c51623ae0fe67d86aa64deb3403abbc4e799":{"kind":"REWORK","_number":1,"created":"2022-02-16 02:44:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/829465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/65/829465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/829465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/829465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/829465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/65/829465/1"}}},"commit":{"parents":[{"commit":"3d03306ae41db5fa3a38c838497455163cd44242","subject":"Merge \"Deprecate tenant_* properties\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3d03306ae41db5fa3a38c838497455163cd44242"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-16 02:43:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-16 02:44:10.000000000","tz":540},"subject":"Add cache client retry options for the pymemcache backend","message":"Add cache client retry options for the pymemcache backend\n\nThis patch specifies a set of options required to setup the retrying\nwrapper feature of pymemcache (dogpile.cache) cache backend.\n\nOriginal oslo.cache change:\nhttps://review.opendev.org/c/openstack/oslo.cache/+/803747\n\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\nDepends-On: https://review.opendev.org/826869\nChange-Id: I3711ce3654a0253dda142def1f1a07b286d06c08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0ce4c51623ae0fe67d86aa64deb3403abbc4e799"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0ce4c51623ae0fe67d86aa64deb3403abbc4e799"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
