)]}'
{"id":"openstack%2Fpuppet-neutron~829466","triplet_id":"openstack%2Fpuppet-neutron~master~If51cb8b706ffca86a9ca22c8dfa172c4f6da8ef1","project":"openstack/puppet-neutron","branch":"master","topic":"oslo.cache.2.10.0","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-20 12:17:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If51cb8b706ffca86a9ca22c8dfa172c4f6da8ef1","subject":"Add HashClient retry options for the dogpile.cache backend","status":"MERGED","created":"2022-02-16 02:46:14.000000000","updated":"2022-02-20 12:18:45.000000000","submitted":"2022-02-20 12:17:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829466-oslo.cache.2.10.0","meta_rev_id":"34aaa8b9eedb68837366c53e7952260435f4c594","_number":829466,"virtual_id_number":829466,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-20 12:17:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-02-16 15:13:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-02-20 09:48:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-02-16 11:22:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-02-20 09:48:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-16 05:09:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-16 11:22:13.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2022-02-16 15:13:43.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-02-20 09:48:22.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"aca51340151e61bcb98343ee1adccd2333d24da0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-16 02:46:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1ed78eeb935f884f33ea25c9276cd148910189a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 05:09:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70b83be6920c43469cb9c453ceb3f7f0 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89b713846272448799e171ede2d2fb74 : SUCCESS in 4m 27s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10f610e9e9f14214a24c1b875641901b : SUCCESS in 4m 33s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d83bb977390b4811a27a5cad67fcb319 : SUCCESS in 21m 10s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/3004ff71bad042c09c6e7dfa0fb48260 : SUCCESS in 21m 39s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00b1ee469d4f4b908f2de07d5fc35e79 : SUCCESS in 17m 32s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7863a1c434d4eec9c68a90d130acf72 : SUCCESS in 19m 46s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23b5b73c12b44b1c9ec13d7523ac80e9 : SUCCESS in 16m 16s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94fb98226ba74541894d2f383287107e : FAILURE in 33m 14s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8bedf858caa477080eeee66f6a25728 : FAILURE in 51m 18s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d44a532d6c9f40679f9f9628a1791374 : FAILURE in 35m 09s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de7c2ad0277b4ebf8d591020aa35b0b8 : FAILURE in 38m 31s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f7e84eb834849039202c9dcebf60251 : SUCCESS in 25m 46s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/72e642ff66114e368b9bc1a61f7dec59 : SUCCESS in 1h 20m 50s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a95b288657f5483e883fb01b71da4745 : SUCCESS in 1h 09m 10s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1951eef11e3442fab93056789df64098 : SUCCESS in 1h 16m 56s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0eb121c410c94acd9a24c0a844a0ccf1 : SUCCESS in 55m 32s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c1485606cb54dea9133e10ee0070d6a : SUCCESS in 22m 12s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ca4a2a40b474c2a92b774612abc5b6b : SUCCESS in 55m 24s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/97a1ad6a4e604a6b8c43ba6f199483ff : SUCCESS in 1h 05m 29s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/78515d758410407d968648e20b211652 : SUCCESS in 1h 11m 48s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dbd06bd18e1b4bc3990f6a6ce3547fe4 : SUCCESS in 51m 30s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4eddb94d2cd94ffe82f4eb60e1dae7a4 : SUCCESS in 23m 36s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/d1581c8f486349af88f9a22cc0db8f30 : SUCCESS in 25m 37s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d89c17cfd984f35ad37141b8df07891 : FAILURE in 11m 50s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4fc514eb7944753b877114ce9dd60b6 : SUCCESS in 19m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a495e6709d944901a74ca2105159590c : SUCCESS in 18m 25s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/1184faa92d7347e1929575b2d58d4a7a : SUCCESS in 2h 10m 15s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/bb657e2d7bb64390ab83ca48979480a9 : SUCCESS in 45m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebbce740911843bdb070f2560630865b : SUCCESS in 4m 30s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/075a1d138058493c9cfdec27f9553b9c : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":1},{"id":"19d3c2f25b8b83d64c6de9d51a16bee797603c46","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-02-16 11:22:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2e80f462839ae32e29f5514e463d9a649d2d3542","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-02-16 15:13:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cc84f2420277e93d779b83f77c8872fadebd9e48","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-20 09:48:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f3911def79fc86cc1cf7e0e96ee961fabc8d1ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 09:49:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"60ebf8afc879a5d0215e14fc927132b984c5ba06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 12:17:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c826122fb99e418b94f276d92b70a9d7 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc5e7b9216da4065b5e6ef6f0edc9d8d : SUCCESS in 3m 41s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a480ae53f83044099d9bc68e031217d6 : SUCCESS in 4m 06s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e012a360d734ee0be8b98634f7f947f : SUCCESS in 33m 28s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/19b90696551f4c469986d64be6b152b0 : SUCCESS in 20m 26s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5a0401948e747f7bd8e59d229ba0fe0 : SUCCESS in 18m 41s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d7f40d36a8b4a8485e1ba122c2f3494 : SUCCESS in 15m 56s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fdd892fd66bf4c25983b33c6e01fa15c : SUCCESS in 1h 19m 58s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d1ece0a4aac142fe9652480d9b562703 : SUCCESS in 1h 48m 36s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5143666da0346b6be0f2438bace8401 : SUCCESS in 1h 20m 22s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca37bb64144f41d0b6bfde424150d10f : SUCCESS in 1h 05m 35s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba0245505e474f958b2ed487468dd09c : SUCCESS in 22m 56s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/2a3ee240542a41e299784b31e3679610 : SUCCESS in 2h 03m 40s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/8c74be7cf5174eb59dd24f697cd63c61 : SUCCESS in 40m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41e43a3d6d884502acd8fd1b25e52f29 : SUCCESS in 3m 59s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b8e970e881f440688f012059ece9deb6 : SUCCESS in 1h 04m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e3e31feac169c348eee448633f57a4eb5342c954","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 12:17:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"34aaa8b9eedb68837366c53e7952260435f4c594","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 12:18:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0865472fdf54457c889314686b98b6cd : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb7392537114be9352aeaba2afd79ecc29e90012","revisions":{"eb7392537114be9352aeaba2afd79ecc29e90012":{"kind":"REWORK","_number":1,"created":"2022-02-16 02:46:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/829466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/66/829466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/66/829466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/66/829466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/66/829466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/66/829466/1"}}},"commit":{"parents":[{"commit":"0ce4c51623ae0fe67d86aa64deb3403abbc4e799","subject":"Add cache client retry options for the pymemcache backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0ce4c51623ae0fe67d86aa64deb3403abbc4e799"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-16 02:45:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-16 02:46:03.000000000","tz":540},"subject":"Add HashClient retry options for the dogpile.cache backend","message":"Add HashClient retry options for the dogpile.cache backend\n\nThis patch specifies a set of options required to setup the HashClient\nretry feature of dogpile.cache cache backend.\n\nOriginal oslo.cache change:\nhttps://review.opendev.org/c/openstack/oslo.cache/+/824944\n\nCo-Authored-By: Hervé Beraud \u003chberaud@redhat.com\u003e\nDepends-On: https://review.opendev.org/826870\nChange-Id: If51cb8b706ffca86a9ca22c8dfa172c4f6da8ef1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/eb7392537114be9352aeaba2afd79ecc29e90012"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/eb7392537114be9352aeaba2afd79ecc29e90012"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
