)]}'
{"id":"openstack%2Fpuppet-neutron~829545","triplet_id":"openstack%2Fpuppet-neutron~master~I88f639219aa95f566ea1b7054bf2cbc339e93b93","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-18 08:51:00.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I88f639219aa95f566ea1b7054bf2cbc339e93b93","subject":"Clean up deprecated support for networking-cisco","status":"MERGED","created":"2022-02-16 13:50:39.000000000","updated":"2022-02-18 13:55:44.000000000","submitted":"2022-02-18 13:53:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829545","meta_rev_id":"524addfde0f4719b06acfdf55adb1632b0f3c241","_number":829545,"virtual_id_number":829545,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-18 13:53:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-02-17 19:42:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-02-17 23:01:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-02-17 23:01:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-16 16:22:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-17 19:42:24.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-02-17 23:01:34.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"566b47e417ab3108ea3285c20db71498849017cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-16 13:50:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d1168296cb52bf88a69223c360a45e98ca60d27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 16:22:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e0f984637af4233a6372631468c945f : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0202a9ee3fe84f83ac499202989ea05f : SUCCESS in 3m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/130d6f83ad0140f2b6418f0b2605a918 : SUCCESS in 3m 43s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/27cc57050c744e0888776794d31bc602 : SUCCESS in 16m 45s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/9e66214e4e0c49c58bb44fce00026270 : SUCCESS in 16m 15s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71790d228cad45a1a34913640f901655 : SUCCESS in 14m 18s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/780855a9c9f44f74937b44472fff985d : SUCCESS in 14m 21s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/136df377066d479ebd762577589d0af6 : SUCCESS in 13m 53s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44662e76e6bf4d368bf55c86551077bb : FAILURE in 29m 18s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6405f1d3d8774c96bf5485dc1b80c33e : FAILURE in 35m 41s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fd8c34cf49f4788802bb966f8a7fd27 : FAILURE in 31m 04s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcff77581c5547b88827e2a7231709c9 : FAILURE in 25m 17s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/00c44903884a4866bdf9f4fb9168cee4 : SUCCESS in 23m 01s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f57b3d5aef0c485e80268076c3b3a421 : SUCCESS in 1h 19m 22s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f69bb510d83b41598b44c15696ac67f7 : SUCCESS in 2h 01m 52s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4b6dab7882f5429d9dd102fabeed1fac : SUCCESS in 1h 22m 54s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/beda52e1fc0444f9b006b4c21358dc00 : SUCCESS in 1h 08m 18s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/006b8acde9a54753891f790b414b6a02 : SUCCESS in 19m 57s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/99461138e8ae486eb9e1354a74db0c57 : SUCCESS in 56m 10s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2391a762a5074a139150d34eedefb082 : SUCCESS in 54m 23s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6ee83ccaafe40788388525e64e8177c : SUCCESS in 56m 40s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3dfd585d6ed74ce2835acab64a3884aa : SUCCESS in 52m 49s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17473ee0e8ff4ab080a6ed554ecca2d3 : SUCCESS in 15m 08s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/939d7ad3adaa491a91794e6ae9d2f365 : SUCCESS in 24m 37s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b50834c4a8c458a920bc1c3ee9d4933 : FAILURE in 7m 39s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1940ea8a5174f67ae5feaf72e8d5aa9 : SUCCESS in 17m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ced87790a8364540b8c6e61e8fdce96c : SUCCESS in 19m 08s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/38c50dc4a3974f609197ff234cb7091a : SUCCESS in 2h 14m 20s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/b13cb1c37580444eb743ca89ceeb0c2d : SUCCESS in 51m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27c92a6ed91e417cabc9dc198a14f12f : SUCCESS in 4m 34s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0ab67771ad69485aa5dfcaf13fd789d5 : SUCCESS in 1h 09m 08s","accounts_in_message":[],"_revision_number":1},{"id":"8d852337be5267c0a3fd75a42c0d885bb8562156","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-02-17 19:42:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c9a8a93e18a19345ec66e986f3e007cdea9625ca","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-17 23:01:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"55ac629b534ab25418b46fb30ccbb52815b35a58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 23:03:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7336d08668073ad45f052bbe0e2feb2c71b09983","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 01:43:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e21341181c34a80bfe423f2604b90bf : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/798485c700b445fbafef2e40710547e8 : SUCCESS in 3m 13s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/987ddad07fcf41929ae74f8bc51354aa : SUCCESS in 4m 03s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/36eff99dff384882a64add634c104c51 : SUCCESS in 19m 34s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/5d52ec8f168143fe811c4cca3ed1d374 : SUCCESS in 14m 38s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/290b2150c2244c928c1fddf2e464050d : SUCCESS in 13m 01s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/544459f7335d41bdb71184d61c423216 : SUCCESS in 14m 28s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/db9a1df27c63490ba89680ab92017775 : SUCCESS in 1h 18m 43s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a1a845895e7c43d5ac496440a209e268 : SUCCESS in 1h 28m 15s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9854c5394ff341afa9de49dbc7c71f4a : SUCCESS in 1h 27m 06s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/817a0592483b42099d3f8b36dec164ed : SUCCESS in 1h 08m 03s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/5826fbb7dd3e48ca933b2f57105191c8 : FAILURE in 23m 18s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/b3bbc245ca7f430487e77225f65c4b20 : SUCCESS in 2h 32m 35s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/587512ed34a64754b869436b574ee8d4 : SUCCESS in 45m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b22c857db993472a85f7f1fbdcd659e0 : SUCCESS in 4m 52s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/864368a81c8340b08874f6f944dbfa97 : SUCCESS in 59m 15s","accounts_in_message":[],"_revision_number":1},{"id":"f4a59827471ff26e3475d7afd874ad0efd9a67d1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-18 04:16:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83fc295f176a64257fcc83afd72e47d0c2ebff06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 06:35:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3705b57528849ffa6901fb8553f9f46 : SUCCESS in 3m 46s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c37b8f9a410442319b29b0de0def2600 : SUCCESS in 3m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42e7ebae37e9469cae89e19b37c34a5d : SUCCESS in 3m 29s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/32c4fd4133fa41d988c2df86297019b0 : SUCCESS in 18m 06s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/4626011a75fc4cb4b76b43ff6f99f2dc : SUCCESS in 16m 51s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3377874dcf2a464db74279e5d4143c9d : SUCCESS in 15m 17s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3407d9522c74d049c71982055a07285 : SUCCESS in 15m 42s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ff846d71f0d49adaccd68c1511ff447 : SUCCESS in 15m 36s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a8d5760c2fb43ffa2f3bb8b158f183a : FAILURE in 33m 04s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/feabe334107847aa87741338bd45b5cb : FAILURE in 38m 00s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a02f000f03a8421e84eb7c54d4fab01a : FAILURE in 32m 47s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce6bdb007c034f7f95c3f21ac0f21e19 : FAILURE in 25m 53s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/05a7572587cc42339804d7469182c531 : SUCCESS in 22m 49s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3489426a1fae4b8b8de7340cccb16d50 : SUCCESS in 1h 02m 55s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/533e4d408e6f49679f2e5e63c4706aee : SUCCESS in 1h 33m 32s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b6f8578777e4b3cbc20c51b7637ad8e : SUCCESS in 1h 21m 10s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c81bcb6036774d3e8a2a2991d7dc13ba : SUCCESS in 45m 48s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9728e5364af94a8b9d1a41cfbf27f163 : SUCCESS in 22m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/565274f2b4eb48f8a0d6bec55bc1cc6a : SUCCESS in 43m 11s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/71926ec8408648f6ad744c119cfc144e : SUCCESS in 1h 12m 35s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/54e7964664954092a82dc0e9d0f80041 : SUCCESS in 53m 42s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a4c73feee114eb49866c13efe1fa167 : SUCCESS in 46m 41s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7ac21928cda448eb6d299cc1e5f0ec3 : SUCCESS in 12m 49s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/f4a510578a8440a1b5de4e14bf154b22 : FAILURE in 21m 51s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/79a62a33e3434f59b01b590fcbccb3b6 : FAILURE in 10m 43s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e24daab844934217aec58e3b3774c86f : SUCCESS in 19m 34s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60a08e99b6f2429587c665905ffd3030 : SUCCESS in 15m 44s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/861257ee5c794881bbcab8087f6544f5 : SUCCESS in 2h 15m 50s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/2bdfe54bea1d4cb2be3274b727d34245 : SUCCESS in 42m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f65ab22901a14c59992d30d9d2524e26 : SUCCESS in 3m 28s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c97e4c08cc934e3d99f927ece7d7632b : SUCCESS in 1h 06m 15s","accounts_in_message":[],"_revision_number":1},{"id":"2dbe34c304e380a8e420ef723c42aae9e0920b7a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-18 08:51:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e971b31f8205c533df488e6df264d82a1db302b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 11:26:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba97ff2b69dc41b8a304c36b7911c5c1 : SUCCESS in 3m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22d704205ae1457ea07e1f46dea108ca : SUCCESS in 4m 21s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4cbd9c87bb84b26b8828c165b8f8650 : SUCCESS in 3m 08s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/bd5aa949a1744b00a743aee2c3f61b47 : SUCCESS in 17m 50s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/6203c456bf5941a0a7551af7f5bec2da : SUCCESS in 19m 56s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2a884c3632142b6b3e27265d77e22d8 : SUCCESS in 12m 39s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf27b108b64045acbe5d1c939b3cd2e1 : SUCCESS in 15m 53s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba08ee59f9c648249618324ccb0b432f : SUCCESS in 11m 24s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a65846e94ec64f809a5b094ab949716b : FAILURE in 30m 00s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf5d514b463d4f8391bae053d5678809 : FAILURE in 42m 20s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae9964027ddc4c00a22fb07e01e77cce : FAILURE in 39m 19s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7ee2c4dc2804e7a8b7106ff93060db3 : FAILURE in 27m 14s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc96fc84ce084e66ac361f2ce6c7877c : SUCCESS in 25m 12s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e526a4c92e6c4652a0173a421350e353 : SUCCESS in 1h 11m 09s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6ddc58cfbe614660ae521c8f4cac99ea : SUCCESS in 1h 50m 20s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1784517f0ce942239d0ef0447764e8f3 : SUCCESS in 1h 15m 59s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf0d92fead034662af45f7ae83d7ebe2 : SUCCESS in 54m 42s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a44bd61a1ce4537ab4290c4ed354fe4 : SUCCESS in 20m 36s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/806bca5ece96420c89bbdb40d06438af : SUCCESS in 1h 02m 31s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/64e5ddfe7c8d41e6a97bc3251f4a8529 : SUCCESS in 1h 05m 34s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffde8a15729646f287a41b2e785fc681 : SUCCESS in 1h 13m 37s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/35585ee47186424b9f596c76e3126e66 : SUCCESS in 50m 43s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ba8f5e83a8d406f84173abdbc9ac6d8 : SUCCESS in 14m 28s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/d04b68a090784de9a180fd88a98d4bb6 : SUCCESS in 31m 15s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4483efc9217f4af6939f838c1962e88a : FAILURE in 8m 32s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3d074bd48be4ac1b313e05711b86b1f : SUCCESS in 16m 14s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f662ac3aa6e7481e992647803468710a : SUCCESS in 18m 18s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/57a87eb086c745938e64de1a4534b076 : SUCCESS in 2h 26m 10s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/16a541f0d6354a5a9e82b456f0468ecb : SUCCESS in 56m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47d2ab08267d4a8c8fdd5513df049ae4 : SUCCESS in 3m 17s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/63adca3816534273a761ae0954eaba2c : SUCCESS in 1h 14m 58s","accounts_in_message":[],"_revision_number":1},{"id":"03e91953bfcdc90175682579541c47e369db829e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 11:28:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b1bbf8ddc3781ecfac594a15765793163b56d90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 13:53:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b3fcb3f43ab43d1b597f4d97e479e2c : SUCCESS in 3m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5e5e23a4a9e47ecba615d8d2c1a3463 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a898db5dc0c140a79e940ab952277ce6 : SUCCESS in 3m 22s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/22f31594b290464cb13ea63925696444 : SUCCESS in 17m 44s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/1fd19746947c484a93e21ca761245949 : SUCCESS in 16m 49s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e7473a0f30a46adb748d1047410e8c9 : SUCCESS in 14m 26s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7d6fc4ff3a14c48ac1d28c0dac6af59 : SUCCESS in 14m 13s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1aa60706cae149a3a65ad6911960b401 : SUCCESS in 1h 24m 05s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a97100c48e5c4e0eb81019efdb4e8a9e : SUCCESS in 1h 47m 07s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d364aca46b3a4bffb0557ad1bf5759a4 : SUCCESS in 1h 28m 08s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/eb0e28f96a114042bd53c57d1d220b52 : SUCCESS in 1h 20m 53s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/8b1c5d5b022d40239b09b2e3c8c5fecf : SUCCESS in 22m 21s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/9f8f9a736af14fbaa07f0411a4613690 : SUCCESS in 2h 21m 13s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/dec9b89e80734f04881d27a78ad25e63 : SUCCESS in 48m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfec17e6426e4b82a64e5c2e1e6fd1a4 : SUCCESS in 3m 56s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0b1c867e50ae4502af10481cc6ca64e8 : SUCCESS in 1h 06m 53s","accounts_in_message":[],"_revision_number":1},{"id":"492fb050108d69418c51362c59de57a74cf6f7c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 13:53:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"524addfde0f4719b06acfdf55adb1632b0f3c241","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 13:55:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ddae6c2c4574e4fb331c4b97403853d : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6c4ba2c2b1b99e885813dc201c4608152d724da","revisions":{"a6c4ba2c2b1b99e885813dc201c4608152d724da":{"kind":"REWORK","_number":1,"created":"2022-02-16 13:50:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/829545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/45/829545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/45/829545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/45/829545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/45/829545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/45/829545/1"}}},"commit":{"parents":[{"commit":"3d03306ae41db5fa3a38c838497455163cd44242","subject":"Merge \"Deprecate tenant_* properties\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3d03306ae41db5fa3a38c838497455163cd44242"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-16 13:44:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-16 13:50:28.000000000","tz":540},"subject":"Clean up deprecated support for networking-cisco","message":"Clean up deprecated support for networking-cisco\n\n... which was deprecated by [1] in a past release.\n\n[1] a8d14466a65e453e2026d753c4ef16e6bea047d0\n\nChange-Id: I88f639219aa95f566ea1b7054bf2cbc339e93b93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a6c4ba2c2b1b99e885813dc201c4608152d724da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a6c4ba2c2b1b99e885813dc201c4608152d724da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
