)]}'
{"id":"openstack%2Fpuppet-neutron~829891","triplet_id":"openstack%2Fpuppet-neutron~master~I94cb39c20b15065f1f628a0bf34192960e54d57f","project":"openstack/puppet-neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-03 13:47:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I94cb39c20b15065f1f628a0bf34192960e54d57f","subject":"Remove unused csv library","status":"MERGED","created":"2022-02-18 13:55:31.000000000","updated":"2022-03-03 13:47:13.000000000","submitted":"2022-03-03 13:47:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829891","meta_rev_id":"a410bfbba2c5f8e91fd66073d548979eb732674b","_number":829891,"virtual_id_number":829891,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-03 13:47:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-02-22 14:56:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-03-03 10:45:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-03-03 10:45:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-18 16:17:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 14:56:09.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-03-03 10:45:39.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"86ffd24ddfc28d1b786ac0ab3610933c72c3d9fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-18 13:55:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab4b59c9b1a8574f7b68cbc326a226607c536cc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 16:17:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/662f774be07d4a7ebd65fda0d3ec7519 : SUCCESS in 3m 27s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d17ef43e23b4bd89b8e0754053e8021 : SUCCESS in 3m 29s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95f168d06c3b4d38bc61fcbc8056ef62 : SUCCESS in 3m 35s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d98227b9d4d74b97b8f6055254ac4e96 : SUCCESS in 13m 26s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/b68af24809ac49a981e14a2267c11469 : SUCCESS in 13m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/faa9bbb938f64d4fadd86e5c448f392d : SUCCESS in 13m 01s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36ba96a1333f498e880a6a1279a9dd1d : SUCCESS in 14m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a767e95866f04dde8e106ce4d2423c78 : SUCCESS in 12m 57s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b990f875608147d4869d704e91b98b4d : FAILURE in 29m 21s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4375c615c2041ea9d428c8012507a6c : FAILURE in 39m 42s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1876dafe924b48f1ae33d2b4d81cc316 : FAILURE in 31m 26s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbdc020d14bf424891304881ab6696f1 : FAILURE in 27m 26s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4fa93c79ae944c3b5ae15cb304373d7 : SUCCESS in 22m 19s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7c26e0f6a919406d8801249ea161887c : SUCCESS in 1h 51m 46s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/68408aff33864c40bbc67fa081463435 : SUCCESS in 1h 29m 57s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/63ca37cb200f46058a939290717ccbd6 : SUCCESS in 1h 24m 52s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/da0ebb77560c4ba491c0aedaf04e0dcb : SUCCESS in 57m 38s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e54361f43e14e2cbb27280977934728 : SUCCESS in 27m 32s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f9483a5efa049fa9ab51b2a002ec88b : SUCCESS in 52m 39s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/333052f16f114a14b125c6c2c92800ee : SUCCESS in 1h 03m 58s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c365906cf78249809dbb5f975e90fdd8 : SUCCESS in 51m 11s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0f1f1fdb15948cf9e7e595b0f47343f : SUCCESS in 47m 32s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b471aa3125e4b3bacd9259dd2a75db1 : SUCCESS in 15m 46s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/b557e3c119704512bec955543e28765d : SUCCESS in 29m 39s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/a55aa427a1a94b748172957e643c2e86 : FAILURE in 9m 22s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a0500d3a7744d22ad769044e1197a71 : SUCCESS in 15m 36s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2df1e9b42e7c430a82a9022b4262f96c : SUCCESS in 16m 17s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/0dc67dd7498040aabbca44d53500b948 : SUCCESS in 2h 14m 24s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/d0a0727e349d4c1ba4df2cbbb9e8a648 : SUCCESS in 55m 07s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4aba1b743b674706afc6cebd518b5231 : FAILURE in 1h 18m 03s","accounts_in_message":[],"_revision_number":1},{"id":"cfe62f5858dec3e8ec694ae5342bfc0deb4b6987","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-18 21:21:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36e552b9f5d666723bfa9c6300386bb11b84082d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 23:17:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30796d89099741dd86d0f17bb0af6174 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5cd98094e144485a2f9fc61ba3934fa : SUCCESS in 4m 04s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7755ca8e1df4f2c915280792fee0f27 : SUCCESS in 3m 53s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8681c190b78b4828ac7511eef14d69f7 : SUCCESS in 15m 19s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/b1caf5f162d344b49340151d9a5b15a1 : SUCCESS in 15m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/535fc2d072904188babdb064ed638052 : SUCCESS in 13m 22s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8299bdd7018f44b588dc55ede60bd151 : SUCCESS in 17m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dacaa3d435334688b0cb1defb3e6c978 : SUCCESS in 18m 31s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3f65751616b4da993a1cc52ad747f5f : FAILURE in 31m 03s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8359c073f6e04f52b4735ca2299ed0e0 : FAILURE in 37m 26s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e11378596cc745a680c7960a78092ddc : FAILURE in 50m 50s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/feba955e350747aa9c69c6fcc6282d54 : FAILURE in 44m 33s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b59e95a15e7942fca537e9efffe91260 : SUCCESS in 24m 50s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d3030c12f974ae8aac9d71b5ea6ce3c : SUCCESS in 1h 31m 19s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c08f54359d448bab6edd147b4f9900c : SUCCESS in 1h 40m 25s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5bcc7860874d45e4abf2a7cb530d1446 : SUCCESS in 1h 25m 32s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4659f486a36a4011b8ec242e82601899 : SUCCESS in 1h 16m 42s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd38aad6dc74462992ca062738bb031c : SUCCESS in 23m 35s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7e616ad21ec409aba34834ce7f794b6 : SUCCESS in 1h 00m 08s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ab59797accd4764be03f53a8598d092 : SUCCESS in 1h 07m 19s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9bed81b4fd6a4acf963526d9ad6f6f4a : SUCCESS in 1h 17m 17s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7710496cf19435881ae68cd480f0e77 : SUCCESS in 39m 38s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d50f201504d400385b0669b2ec15603 : SUCCESS in 20m 22s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed1d0209b7ef4054b2f87b5fbfebd6ec : SUCCESS in 26m 12s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/748ef69f21c542a294cf4a029d08cd24 : FAILURE in 9m 02s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75d86a4b3a594c45b0148fabb25a3411 : SUCCESS in 21m 36s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd06c6d727a6409a84cd70a85e7de99e : SUCCESS in 21m 05s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/eb2b1ba21481442290298bad2d906e97 : RETRY_LIMIT in 36s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/fe24a68050a24d8c88e6dcd4b797d6fc : SKIPPED\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/39f11a9a6a78412a945a9b7d4ff0bc9c : SKIPPED","accounts_in_message":[],"_revision_number":1},{"id":"98bfcf7c54d238dff8e267e533cb36cc1f6f530f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-19 13:24:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"26ff96e4c125d31165fefd5c66107420dd8f1d69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-19 15:34:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ed51347b134443ba3db546f1be66ec5 : SUCCESS in 4m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/332c749a6f9449a29b7207f0240a11b9 : SUCCESS in 6m 40s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/799f776f12a24068bc11dab61f635610 : SUCCESS in 4m 05s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/92fb75a0d5f640f98ef1270a4f6d7fd6 : SUCCESS in 17m 03s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/69b331b758c64809aa467837c83789cf : SUCCESS in 18m 24s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bd4984d61dc4c488c428df2796e3678 : SUCCESS in 11m 15s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50402cbd5bc249709e2ba32ca466b74b : SUCCESS in 12m 45s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/236f1273f4154b89bac7ad0337d763f8 : SUCCESS in 15m 29s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/556dbf7751404402accd465300adcc72 : FAILURE in 32m 00s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdde1d718d7e413e98cb9551c7ef50c2 : FAILURE in 37m 10s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95e9338e3f8e40aeb294dc564552fcc3 : FAILURE in 33m 44s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49c6d0ee47fc425695bb1abfd4647bf1 : FAILURE in 28m 35s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f327f53b7c0d4bec83d5705643f14c90 : SUCCESS in 25m 04s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b8a95f8b81b04aab9f78e8ff03846b26 : SUCCESS in 1h 18m 01s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/aec7afa78206404d9b262cc8c7a821da : SUCCESS in 1h 49m 34s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5268f813e6f941efbeb0df7cceb9c591 : SUCCESS in 1h 15m 51s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c64b00d5c18c4022a67d33090454d3fb : SUCCESS in 1h 02m 32s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/95706d9c778f4586841096c044ed581f : SUCCESS in 21m 39s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0de733c0cd81434bb240554ad34a6cd1 : SUCCESS in 55m 43s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/10f67fd7f3914c90a297339dc32fcb01 : SUCCESS in 55m 05s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/91ae4111b41140d8b4cde4723904bc80 : SUCCESS in 53m 34s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3aa079f0685496eab301799e4ff1b07 : SUCCESS in 39m 39s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66ea816fac1f493aad9a51de6fea0b41 : SUCCESS in 16m 26s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/f6a3b474c45544cfa1c00b1c1a1cf29a : SUCCESS in 24m 29s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae79e86c27b942fa9eb761d631b5e7f1 : FAILURE in 8m 46s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6249d73bcc8e4f6f834822d3e3560952 : SUCCESS in 18m 12s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93e4f7232f9240f18464e8e7d4063c91 : SUCCESS in 17m 26s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/a6cd37874a854f3d9194211f6329096a : SUCCESS in 2h 02m 48s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/cadb7efaad2d4e8b84e97b0919214c2b : SUCCESS in 40m 48s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/08a269c25ae244b19c3bd375b9d48a8c : FAILURE in 1h 05m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c8834a68da9bf4368164e20a01cd6ef4e062944b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-21 06:50:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f011a3fd0e0d700b24b3a2fc3d3a4bc85dc0065","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 09:10:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2dfd7db46acc41f4b7b5ad3274fa8ebe : SUCCESS in 3m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52eb79dc8a4b4cb19665595a2ae9a198 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c49be8ac3eb94ffcb26eded9601fc4bc : SUCCESS in 3m 16s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc8957f39ede490ba911f6c58a4540ed : SUCCESS in 19m 42s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/8cb23eabc7ef419ca378f0ebc9c8b238 : SUCCESS in 18m 39s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a7e5b63437549b9a5dd574c75242303 : SUCCESS in 16m 31s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c019a61df7d246478d5be7d5e4c5aabb : SUCCESS in 17m 28s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d604e1c6ba2c4ed5b8dfb6b986278918 : SUCCESS in 15m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/326480367b02403c9bb78c3ddbc4db0d : FAILURE in 28m 57s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8df0ec32e2ad46aa9d6648b7f231410b : FAILURE in 35m 41s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8921ff67d6334e51a5ca31ee965bf4d6 : FAILURE in 30m 50s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0248fef1c6664f8b8a66f6745d1ebc07 : FAILURE in 26m 58s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e0b49652247b43899c4dca9264b4a411 : SUCCESS in 21m 31s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f886cf40d0044d4b27ec8acf185370d : SUCCESS in 1h 16m 02s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c9f34cda03d482aae06de70f0e91b63 : SUCCESS in 1h 50m 45s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc561d884481414bb3c00b2073b0f2c8 : SUCCESS in 1h 21m 21s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e4f285a0bd6485d9e508a15087c8080 : SUCCESS in 1h 15m 59s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/38ae88ec7b614acd82946063faf2b915 : SUCCESS in 16m 03s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4144882e8994c23a1505044de95a50e : SUCCESS in 42m 04s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/073f00ce97e541eaab6d63f40cb0f321 : SUCCESS in 51m 56s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/56ea246b06304ce3bfa37c5ffaae45b8 : SUCCESS in 49m 38s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3447b76c43854b7f925617df8e6db772 : SUCCESS in 53m 11s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5544595262744dbdbe21a64e316311d9 : SUCCESS in 15m 23s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/508fc79246f549d68b54105d3af90272 : SUCCESS in 21m 18s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/153f0f8ff8d040a89427d83eb4aac510 : FAILURE in 6m 57s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1846dd10dfcb46c4a3c70e03e4d3f936 : SUCCESS in 15m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0b418ba4dc742a8acd39f6a073777a7 : SUCCESS in 14m 46s (non-voting)\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/959239d5b711458186e31612ec71ac05 : SUCCESS in 1h 43m 33s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/81f783acce054c0e97953e6433d8046d : SUCCESS in 37m 02s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d7d8178a4ac74dbdb5d0c75ecfb2d14c : SUCCESS in 54m 36s","accounts_in_message":[],"_revision_number":1},{"id":"728752ebb954b4b6bf77671fb94ef13474356ea4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-02-22 14:56:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"52dc99cd751f2d817229ac9178f413c8c337df24","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-03 10:45:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"85c8fa0803f778f4a2877d2cf4f41c71aee46b66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 10:50:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3337e079373533f3e50fcc7856339334b0b254fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 13:47:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d06e697e5d164edd85821260124f1a1e : SUCCESS in 5m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90997fc65064485690aa76d66c449163 : SUCCESS in 5m 32s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cdf2b827a944d3c97995732e0e85f83 : SUCCESS in 5m 18s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/301d25fd45e84cdfbe2d362a4fb14826 : SUCCESS in 21m 27s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/5bf205f2fa0e49b3b78a19e84fdf3723 : SUCCESS in 21m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00559e76420e4bc7912d8a30180a2d17 : SUCCESS in 17m 09s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9f4c6d4f72f46d7ba6a6205b952b5de : SUCCESS in 22m 10s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ae6cb0fffdf545d89fb4e09b812b7e55 : SUCCESS in 1h 01m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e0e14ebd5f414f6caf0d6396f983ebee : SUCCESS in 1h 13m 53s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/62db2faea1f04413876e40af9ee4aecc : SUCCESS in 1h 07m 34s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/132b07a5e89f4ebd9fb411527d36cd28 : SUCCESS in 1h 02m 38s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef9b105b46754a1998bcb5a6f217aa60 : SUCCESS in 25m 24s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/e13bd9e2fc894804bcb7e57850f20e85 : SUCCESS in 2h 17m 29s\n- tripleo-puppet-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/d90e5a90573341f0985336a1ae90a88a : SUCCESS in 52m 18s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/387502b43c7f42a581ece7ef976dbb52 : SUCCESS in 1h 10m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a410bfbba2c5f8e91fd66073d548979eb732674b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 13:47:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0730989e2288524e97015dc1a3500ac317913100","revisions":{"0730989e2288524e97015dc1a3500ac317913100":{"kind":"REWORK","_number":1,"created":"2022-02-18 13:55:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/829891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/91/829891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/91/829891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/91/829891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/91/829891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/91/829891/1"}}},"commit":{"parents":[{"commit":"3d03306ae41db5fa3a38c838497455163cd44242","subject":"Merge \"Deprecate tenant_* properties\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3d03306ae41db5fa3a38c838497455163cd44242"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-18 13:54:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-18 13:55:01.000000000","tz":540},"subject":"Remove unused csv library","message":"Remove unused csv library\n\nThe library is no longer used by the neutron provider directly, thus\nthe require statement can be removed.\n\nChange-Id: I94cb39c20b15065f1f628a0bf34192960e54d57f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0730989e2288524e97015dc1a3500ac317913100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0730989e2288524e97015dc1a3500ac317913100"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
